)]}'
{"id":"openstack%2Fswift~672107","triplet_id":"openstack%2Fswift~master~Ie38357e4c278335c35d186708573bb6bdabaa012","project":"openstack/swift","branch":"master","topic":"retire-api-site","hashtags":[],"change_id":"Ie38357e4c278335c35d186708573bb6bdabaa012","subject":"Update api-ref location","status":"MERGED","created":"2019-07-22 16:54:49.000000000","updated":"2019-07-23 17:57:15.000000000","submitted":"2019-07-23 17:53:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672107-1563904416410-83992aa8","meta_rev_id":"ef54db7c726929f2fce4d9321dc2cded77a333a1","_number":672107,"virtual_id_number":672107,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-23 04:09:03.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-07-23 17:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-23 15:34:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-23 15:34:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-23 15:34:19.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-07-23 17:53:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"62d79c04f8689bbbbac01772eef0fe933d3cc1dc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 16:54:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75e2e136b1dc998a57d4d4d76f376520534821ab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-22 17:18:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f51a910963bb0e4f44e16eaaad8eefdf7a709182","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 18:53:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/07/672107/2/check/test-release-openstack/fc0707e/ : SUCCESS in 4m 21s\n- build-openstack-api-ref http://logs.openstack.org/07/672107/2/check/build-openstack-api-ref/72593e0/html/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/07/672107/2/check/openstack-tox-docs/eda9086/html/ : SUCCESS in 5m 26s\n- build-openstack-releasenotes http://logs.openstack.org/07/672107/2/check/build-openstack-releasenotes/064e36c/html/ : SUCCESS in 7m 48s\n- swift-build-image http://logs.openstack.org/07/672107/2/check/swift-build-image/a957b34/ : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/07/672107/2/check/swift-tox-py27/0594dd6/ : SUCCESS in 14m 21s\n- swift-tox-py36 http://logs.openstack.org/07/672107/2/check/swift-tox-py36/72e0fef/ : SUCCESS in 12m 07s (non-voting)\n- swift-tox-py37 http://logs.openstack.org/07/672107/2/check/swift-tox-py37/e17dc2a/ : SUCCESS in 13m 10s\n- swift-tox-func http://logs.openstack.org/07/672107/2/check/swift-tox-func/aeb3d9a/ : SUCCESS in 9m 33s\n- swift-tox-func-encryption http://logs.openstack.org/07/672107/2/check/swift-tox-func-encryption/9d6001d/ : SUCCESS in 9m 26s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/07/672107/2/check/swift-tox-func-domain-remap-staticweb/3b2ee06/ : SUCCESS in 11m 59s\n- swift-tox-func-ec http://logs.openstack.org/07/672107/2/check/swift-tox-func-ec/c336b7a/ : SUCCESS in 10m 24s\n- swift-tox-func-s3api http://logs.openstack.org/07/672107/2/check/swift-tox-func-s3api/0693637/ : SUCCESS in 12m 25s\n- swift-tox-func-py37 http://logs.openstack.org/07/672107/2/check/swift-tox-func-py37/52bdb1f/ : SUCCESS in 10m 52s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/07/672107/2/check/swift-tox-func-domain-remap-staticweb-py37/7c82ce7/ : SUCCESS in 9m 17s\n- swift-tox-func-ec-py37 http://logs.openstack.org/07/672107/2/check/swift-tox-func-ec-py37/a8302c9/ : SUCCESS in 8m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth http://logs.openstack.org/07/672107/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/dd3d7ed/ : SUCCESS in 18m 32s (non-voting)\n- swift-probetests-centos-7 http://logs.openstack.org/07/672107/2/check/swift-probetests-centos-7/fcbab0b/ : SUCCESS in 30m 54s\n- swift-dsvm-functional http://logs.openstack.org/07/672107/2/check/swift-dsvm-functional/2242f20/ : SUCCESS in 26m 27s\n- swift-dsvm-functional-py3 http://logs.openstack.org/07/672107/2/check/swift-dsvm-functional-py3/39be8d5/ : SUCCESS in 30m 17s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/07/672107/2/check/swift-dsvm-functional-ipv6/1960aef/ : SUCCESS in 33m 41s\n- swift-tox-lower-constraints http://logs.openstack.org/07/672107/2/check/swift-tox-lower-constraints/3de75a7/ : SUCCESS in 12m 32s\n- openstack-tox-pep8 http://logs.openstack.org/07/672107/2/check/openstack-tox-pep8/0859beb/ : SUCCESS in 7m 15s\n- swift-multinode-rolling-upgrade http://logs.openstack.org/07/672107/2/check/swift-multinode-rolling-upgrade/b7bc899/ : SUCCESS in 18m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a2205a955dfa5e9b05d6007013953522a30a1f81","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-22 20:42:42.000000000","message":"Patch Set 2: Code-Review+1\n\nSeems to make sense. Even picked up a few http -\u003e https fixes!\n\nWhy\u0027d we back out the release note fix? Are we not supposed to issue corrections to release notes on master?\n\n(Warning: rabbit hole ahead.)\n\nIs *that* why there\u0027s both\n\n- https://docs.openstack.org/releasenotes/swift/current.html#relnotes-2-21-0 and\n- https://docs.openstack.org/releasenotes/swift/stein.html#relnotes-2-21-0-stable-stein\n\n?? No... that file only has one commit in its history... meanwhile https://opendev.org/openstack/swift/commit/a85dc0a22562ca33968e9e214c03086ff06355aa didn\u0027t drag 2.10.0 into stein, but neither is it reflected in https://docs.openstack.org/releasenotes/swift/newton.html#relnotes-2-10-0-stable-newton ... I guess we would\u0027ve needed to backport the correction to the (now closed) stable/newton branch?","accounts_in_message":[],"_revision_number":2},{"id":"408485fc27fc8f62290c1efea8eeabef79fc680f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-23 04:09:03.000000000","message":"Patch Set 2:\n\nI removed release notes so that it does not show up again as recent one, it would need updating on the proper branch.","accounts_in_message":[],"_revision_number":2},{"id":"0cc32f493c895e55440d3602d5b379800f257b0f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-23 15:34:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"24325d8d06502d4197a739e5de9985a01f522772","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:40:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"165bf1497ca47d92873f4487f81cd6daffa0328b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:53:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/07/672107/2/gate/test-release-openstack/3d2a660/ : SUCCESS in 3m 00s\n- build-openstack-api-ref http://logs.openstack.org/07/672107/2/gate/build-openstack-api-ref/c00a05e/html/ : SUCCESS in 5m 30s\n- openstack-tox-docs http://logs.openstack.org/07/672107/2/gate/openstack-tox-docs/0d6ebcd/html/ : SUCCESS in 6m 16s\n- build-openstack-releasenotes http://logs.openstack.org/07/672107/2/gate/build-openstack-releasenotes/7da32d6/html/ : SUCCESS in 7m 41s\n- swift-upload-image http://logs.openstack.org/07/672107/2/gate/swift-upload-image/2589c54/ : SUCCESS in 10m 22s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/07/672107/2/gate/swift-tox-py27/84d7886/ : SUCCESS in 12m 02s\n- swift-tox-py37 http://logs.openstack.org/07/672107/2/gate/swift-tox-py37/fd1123b/ : SUCCESS in 10m 54s\n- swift-tox-func http://logs.openstack.org/07/672107/2/gate/swift-tox-func/6c26a4d/ : SUCCESS in 9m 56s\n- swift-tox-func-encryption http://logs.openstack.org/07/672107/2/gate/swift-tox-func-encryption/790d681/ : SUCCESS in 9m 54s\n- swift-tox-func-domain-remap-staticweb http://logs.openstack.org/07/672107/2/gate/swift-tox-func-domain-remap-staticweb/44ce03b/ : SUCCESS in 11m 18s\n- swift-tox-func-ec http://logs.openstack.org/07/672107/2/gate/swift-tox-func-ec/dae2b6b/ : SUCCESS in 10m 37s\n- swift-tox-func-s3api http://logs.openstack.org/07/672107/2/gate/swift-tox-func-s3api/2efdb2d/ : SUCCESS in 12m 35s\n- swift-tox-func-py37 http://logs.openstack.org/07/672107/2/gate/swift-tox-func-py37/d2ccfc7/ : SUCCESS in 11m 40s\n- swift-tox-func-domain-remap-staticweb-py37 http://logs.openstack.org/07/672107/2/gate/swift-tox-func-domain-remap-staticweb-py37/70c9ed8/ : SUCCESS in 9m 30s\n- swift-tox-func-ec-py37 http://logs.openstack.org/07/672107/2/gate/swift-tox-func-ec-py37/4cf080b/ : SUCCESS in 9m 11s\n- swift-probetests-centos-7 http://logs.openstack.org/07/672107/2/gate/swift-probetests-centos-7/b613821/ : SUCCESS in 35m 41s\n- swift-dsvm-functional http://logs.openstack.org/07/672107/2/gate/swift-dsvm-functional/07c804a/ : SUCCESS in 30m 48s\n- swift-dsvm-functional-py3 http://logs.openstack.org/07/672107/2/gate/swift-dsvm-functional-py3/905cc20/ : SUCCESS in 38m 24s\n- swift-dsvm-functional-ipv6 http://logs.openstack.org/07/672107/2/gate/swift-dsvm-functional-ipv6/d82cc36/ : SUCCESS in 29m 35s\n- swift-tox-lower-constraints http://logs.openstack.org/07/672107/2/gate/swift-tox-lower-constraints/9d4a79c/ : SUCCESS in 12m 34s\n- openstack-tox-pep8 http://logs.openstack.org/07/672107/2/gate/openstack-tox-pep8/5734a2e/ : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":2},{"id":"bcabb2c3872b65026afbb98c7d44546ae2eff57b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:53:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"67c10efa3e7f7c9f3989769c5bdb96c8c295927f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:57:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image http://logs.openstack.org/07/672107/2/promote/swift-promote-image/3f5e593/ : SUCCESS in 1m 28s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"73ec8f15df9a1f112f981f15a3ddc323eb1fb723","revisions":{"eb02c870d6fb509edd710a349a479fddb93581fa":{"kind":"REWORK","_number":1,"created":"2019-07-22 16:54:49.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/07/672107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/672107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/672107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/672107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/672107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/672107/1"}}},"commit":{"parents":[{"commit":"2926fc96a36a573380eca18d60440a2337442015","subject":"py3: Bring functional/test_object.py under test; add func-ec-py37 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2926fc96a36a573380eca18d60440a2337442015"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:53:40.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:54:15.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that the Swift API lives at /object-store and not /object-storage.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: Ie38357e4c278335c35d186708573bb6bdabaa012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb02c870d6fb509edd710a349a479fddb93581fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb02c870d6fb509edd710a349a479fddb93581fa"}]},"branch":"refs/heads/master"},"73ec8f15df9a1f112f981f15a3ddc323eb1fb723":{"kind":"REWORK","_number":2,"created":"2019-07-22 17:18:03.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/07/672107/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/672107/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/672107/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/672107/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/672107/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/672107/2"}}},"commit":{"parents":[{"commit":"2926fc96a36a573380eca18d60440a2337442015","subject":"py3: Bring functional/test_object.py under test; add func-ec-py37 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2926fc96a36a573380eca18d60440a2337442015"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 16:53:40.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-22 17:17:55.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that the Swift API lives at /object-store and not /object-storage.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: Ie38357e4c278335c35d186708573bb6bdabaa012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73ec8f15df9a1f112f981f15a3ddc323eb1fb723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73ec8f15df9a1f112f981f15a3ddc323eb1fb723"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
