)]}'
{"id":"openstack%2Fswift~673118","triplet_id":"openstack%2Fswift~master~Ib121d6c52601488647fd16dfd4f49bb4492c7a2f","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ib121d6c52601488647fd16dfd4f49bb4492c7a2f","subject":"Make backport releases conform to existing CHANGELOG formatting","status":"MERGED","created":"2019-07-26 22:59:49.000000000","updated":"2019-08-05 17:36:43.000000000","submitted":"2019-08-05 17:35:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673118-1565026538724-30793c9e","meta_rev_id":"b9140fc4a0f282f1415fbdc917621ceccb2e3cc8","_number":673118,"virtual_id_number":673118,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-08-05 17:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":2,"date":"2019-08-05 14:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":1,"date":"2019-08-05 14:36:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 14:36:32.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-08-05 17:35:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8f626a602b990992138dc440900cb73a80e983c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-07-26 22:59:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a0cc736342605840ff0737326cb52bc483f65ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 23:17:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/18/673118/1/check/openstack-tox-docs/65b12d4/html/ : SUCCESS in 5m 19s\n- swift-build-image http://logs.openstack.org/18/673118/1/check/swift-build-image/b6438be/ : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py27 http://logs.openstack.org/18/673118/1/check/swift-tox-py27/e1001d6/ : SUCCESS in 11m 03s\n- swift-tox-py36 http://logs.openstack.org/18/673118/1/check/swift-tox-py36/7e57c88/ : SUCCESS in 11m 50s\n- swift-tox-py37 http://logs.openstack.org/18/673118/1/check/swift-tox-py37/2d71549/ : SUCCESS in 10m 27s\n- swift-tox-lower-constraints http://logs.openstack.org/18/673118/1/check/swift-tox-lower-constraints/f560309/ : SUCCESS in 12m 26s\n- openstack-tox-pep8 http://logs.openstack.org/18/673118/1/check/openstack-tox-pep8/d84dfa5/ : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"368d7c6e100780949fe5c0b2bc085b2ca460a79e","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-05 14:36:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8e8af2996c432c8b5598fa9546973a34f8fbe34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 14:36:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8d19dd3959f1159b27928f9174657ad1b2b03b98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 16:30:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/18/673118/1/gate/openstack-tox-docs/8990348/html/ : SUCCESS in 5m 10s\n- swift-upload-image https://logs.opendev.org/18/673118/1/gate/swift-upload-image/baa1103/ : SUCCESS in 13m 12s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/18/673118/1/gate/swift-tox-py27/b765fba/ : SUCCESS in 12m 54s\n- swift-tox-py36 https://logs.opendev.org/18/673118/1/gate/swift-tox-py36/2151584/ : SUCCESS in 12m 49s\n- swift-tox-py37 https://logs.opendev.org/18/673118/1/gate/swift-tox-py37/f4e9640/ : SUCCESS in 14m 45s\n- swift-tox-func https://logs.opendev.org/18/673118/1/gate/swift-tox-func/9e3dd21/ : SUCCESS in 12m 57s\n- swift-tox-func-encryption https://logs.opendev.org/18/673118/1/gate/swift-tox-func-encryption/ab01b63/ : SUCCESS in 11m 49s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/18/673118/1/gate/swift-tox-func-domain-remap-staticweb/10d07da/ : SUCCESS in 13m 18s\n- swift-tox-func-ec https://logs.opendev.org/18/673118/1/gate/swift-tox-func-ec/d60987a/ : SUCCESS in 10m 46s\n- swift-tox-func-s3api https://logs.opendev.org/18/673118/1/gate/swift-tox-func-s3api/98f8873/ : SUCCESS in 16m 35s\n- swift-tox-func-py37 https://logs.opendev.org/18/673118/1/gate/swift-tox-func-py37/26ced58/ : SUCCESS in 11m 51s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/18/673118/1/gate/swift-tox-func-domain-remap-staticweb-py37/7ef66c5/ : SUCCESS in 17m 25s\n- swift-tox-func-ec-py37 https://logs.opendev.org/18/673118/1/gate/swift-tox-func-ec-py37/514ca95/ : SUCCESS in 9m 46s\n- swift-tox-lower-constraints https://logs.opendev.org/18/673118/1/gate/swift-tox-lower-constraints/bfbc257/ : FAILURE in 13m 19s\n- openstack-tox-pep8 https://logs.opendev.org/18/673118/1/gate/openstack-tox-pep8/e47e2d9/ : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"692076745166d956acdc75c77a38d6b65f35624c","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-05 16:34:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2cefea0e52637dcf00f54451476f980bfbb399b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:09:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/18/673118/1/check/openstack-tox-docs/cf1a6a6/html/ : SUCCESS in 5m 30s\n- swift-build-image https://logs.opendev.org/18/673118/1/check/swift-build-image/d9db1d3/ : SUCCESS in 15m 25s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/18/673118/1/check/swift-tox-py27/56ef0c8/ : SUCCESS in 12m 12s\n- swift-tox-py36 https://logs.opendev.org/18/673118/1/check/swift-tox-py36/d3e5e58/ : SUCCESS in 10m 15s\n- swift-tox-py37 https://logs.opendev.org/18/673118/1/check/swift-tox-py37/c2ce82d/ : SUCCESS in 13m 26s\n- swift-tox-lower-constraints https://logs.opendev.org/18/673118/1/check/swift-tox-lower-constraints/8c1d482/ : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://logs.opendev.org/18/673118/1/check/openstack-tox-pep8/a75b3e8/ : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"035d456e8e1d8a5a91ee89345b0aa2ca06cdde71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:09:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"83bc9589288e9d5642f949b790a2a271e755edd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:35:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d002c3906a69bddb0fc8b7b73023c5a4cf0049b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:35:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/18/673118/1/gate/openstack-tox-docs/f4801c6/html/ : SUCCESS in 4m 42s\n- swift-upload-image https://logs.opendev.org/18/673118/1/gate/swift-upload-image/4633f97/ : SUCCESS in 9m 38s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/18/673118/1/gate/swift-tox-py27/ba992b6/ : SUCCESS in 10m 20s\n- swift-tox-py36 https://logs.opendev.org/18/673118/1/gate/swift-tox-py36/6bcd482/ : SUCCESS in 10m 46s\n- swift-tox-py37 https://logs.opendev.org/18/673118/1/gate/swift-tox-py37/8a8f849/ : SUCCESS in 11m 03s\n- swift-tox-func https://logs.opendev.org/18/673118/1/gate/swift-tox-func/a59779f/ : SUCCESS in 13m 43s\n- swift-tox-func-encryption https://logs.opendev.org/18/673118/1/gate/swift-tox-func-encryption/4aa6d84/ : SUCCESS in 13m 34s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/18/673118/1/gate/swift-tox-func-domain-remap-staticweb/d012895/ : SUCCESS in 11m 40s\n- swift-tox-func-ec https://logs.opendev.org/18/673118/1/gate/swift-tox-func-ec/26a623c/ : SUCCESS in 10m 03s\n- swift-tox-func-s3api https://logs.opendev.org/18/673118/1/gate/swift-tox-func-s3api/3280477/ : SUCCESS in 14m 11s\n- swift-tox-func-py37 https://logs.opendev.org/18/673118/1/gate/swift-tox-func-py37/390e9d9/ : SUCCESS in 10m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/18/673118/1/gate/swift-tox-func-domain-remap-staticweb-py37/3cf2690/ : SUCCESS in 10m 18s\n- swift-tox-func-ec-py37 https://logs.opendev.org/18/673118/1/gate/swift-tox-func-ec-py37/92f4a34/ : SUCCESS in 10m 05s\n- swift-tox-lower-constraints https://logs.opendev.org/18/673118/1/gate/swift-tox-lower-constraints/60e5321/ : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://logs.opendev.org/18/673118/1/gate/openstack-tox-pep8/2033fe5/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b9140fc4a0f282f1415fbdc917621ceccb2e3cc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 17:36:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image https://logs.opendev.org/18/673118/1/promote/swift-promote-image/c0f7335/ : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8635e8cc867a5f44f6269b1f85b2a3d360a35561","revisions":{"8635e8cc867a5f44f6269b1f85b2a3d360a35561":{"kind":"REWORK","_number":1,"created":"2019-07-26 22:59:49.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/673118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/673118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/673118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/673118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/673118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/673118/1"}}},"commit":{"parents":[{"commit":"d6e790d1b71139623ee06f767a16f3adc576ffbb","subject":"Merge \"Fix up errno checking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6e790d1b71139623ee06f767a16f3adc576ffbb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-26 22:59:35.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-26 22:59:35.000000000","tz":-420},"subject":"Make backport releases conform to existing CHANGELOG formatting","message":"Make backport releases conform to existing CHANGELOG formatting\n\nChange-Id: Ib121d6c52601488647fd16dfd4f49bb4492c7a2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8635e8cc867a5f44f6269b1f85b2a3d360a35561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8635e8cc867a5f44f6269b1f85b2a3d360a35561"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
