)]}'
{"id":"openstack%2Fswift~674052","triplet_id":"openstack%2Fswift~master~I0c3a235b8905e02d2e202215f7f0b43db295b79b","project":"openstack/swift","branch":"master","topic":"673121","hashtags":[],"change_id":"I0c3a235b8905e02d2e202215f7f0b43db295b79b","subject":"Negative test for non-empty chunked put symlink","status":"MERGED","created":"2019-08-01 14:34:27.000000000","updated":"2019-08-02 05:49:07.000000000","submitted":"2019-08-02 05:48:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674052-1564724884903-d227193f","meta_rev_id":"e2cfd9da427454d0c43e77b0ea7a44bc44d521c7","_number":674052,"virtual_id_number":674052,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-02 05:48:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-01 20:26:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-01 20:26:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 20:26:41.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-08-02 05:48:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6846fb2aeff8f7000c2cf1e2d9d55e722f0d41c8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-08-01 14:34:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcca5e3e47ae5e2a38ead88c06a3aba4c2c29e4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 19:32:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/52/674052/1/check/openstack-tox-docs/63d443c/html/ : SUCCESS in 4m 06s\n- swift-build-image https://logs.opendev.org/52/674052/1/check/swift-build-image/b66628f/ : SUCCESS in 9m 43s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/52/674052/1/check/swift-tox-py27/14cf1b5/ : SUCCESS in 11m 15s\n- swift-tox-py36 https://logs.opendev.org/52/674052/1/check/swift-tox-py36/5d5c2d2/ : SUCCESS in 11m 11s\n- swift-tox-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-py37/1f0aef9/ : SUCCESS in 11m 02s\n- swift-tox-func https://logs.opendev.org/52/674052/1/check/swift-tox-func/da8f652/ : SUCCESS in 9m 00s\n- swift-tox-func-encryption https://logs.opendev.org/52/674052/1/check/swift-tox-func-encryption/c2a8204/ : SUCCESS in 9m 41s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/52/674052/1/check/swift-tox-func-domain-remap-staticweb/176aa34/ : SUCCESS in 9m 06s\n- swift-tox-func-ec https://logs.opendev.org/52/674052/1/check/swift-tox-func-ec/f2684e3/ : SUCCESS in 10m 49s\n- swift-tox-func-s3api https://logs.opendev.org/52/674052/1/check/swift-tox-func-s3api/8d5013e/ : SUCCESS in 13m 13s\n- swift-tox-func-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-func-py37/5f7ad08/ : SUCCESS in 10m 39s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-func-domain-remap-staticweb-py37/15fdb27/ : SUCCESS in 8m 59s\n- swift-tox-func-ec-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-func-ec-py37/c8febce/ : SUCCESS in 8m 16s\n- swift-dsvm-functional https://logs.opendev.org/52/674052/1/check/swift-dsvm-functional/c49ef6d/ : SUCCESS in 27m 38s\n- swift-dsvm-functional-py3 https://logs.opendev.org/52/674052/1/check/swift-dsvm-functional-py3/2875cf4/ : SUCCESS in 32m 18s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/52/674052/1/check/swift-dsvm-functional-ipv6/08842d6/ : SUCCESS in 30m 49s\n- swift-tox-lower-constraints https://logs.opendev.org/52/674052/1/check/swift-tox-lower-constraints/e9fcd5a/ : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://logs.opendev.org/52/674052/1/check/openstack-tox-pep8/2dd5354/ : SUCCESS in 6m 30s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/52/674052/1/check/swift-multinode-rolling-upgrade/b0b873b/ : SUCCESS in 18m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d466aa16e52de0889e8a9ef491b926bcc0adbb7e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-01 20:26:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4ad0510970c3bc6b2ecec0d56059f9f3fac9fb2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 21:32:23.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"e24629f75d7fbc6353072432a75eda1f5eacbf83","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-02 03:46:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b84118a2ea395f5aa3192a5e532783de2a9139e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 04:14:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/52/674052/1/check/openstack-tox-docs/d10c942/html/ : SUCCESS in 4m 27s\n- swift-build-image https://logs.opendev.org/52/674052/1/check/swift-build-image/70c5116/ : SUCCESS in 10m 52s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/52/674052/1/check/swift-tox-py27/a1281fb/ : SUCCESS in 15m 26s\n- swift-tox-py36 https://logs.opendev.org/52/674052/1/check/swift-tox-py36/a566d07/ : SUCCESS in 11m 34s\n- swift-tox-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-py37/5e17495/ : FAILURE in 11m 24s\n- swift-tox-func https://logs.opendev.org/52/674052/1/check/swift-tox-func/5ab44de/ : SUCCESS in 11m 33s\n- swift-tox-func-encryption https://logs.opendev.org/52/674052/1/check/swift-tox-func-encryption/b1241d0/ : SUCCESS in 9m 58s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/52/674052/1/check/swift-tox-func-domain-remap-staticweb/3d1035d/ : SUCCESS in 8m 56s\n- swift-tox-func-ec https://logs.opendev.org/52/674052/1/check/swift-tox-func-ec/baa26d0/ : SUCCESS in 9m 32s\n- swift-tox-func-s3api https://logs.opendev.org/52/674052/1/check/swift-tox-func-s3api/2d2d7c6/ : SUCCESS in 12m 36s\n- swift-tox-func-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-func-py37/491ead8/ : SUCCESS in 9m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-func-domain-remap-staticweb-py37/757d871/ : SUCCESS in 8m 16s\n- swift-tox-func-ec-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-func-ec-py37/6082927/ : SUCCESS in 9m 00s\n- swift-dsvm-functional https://logs.opendev.org/52/674052/1/check/swift-dsvm-functional/4debee8/ : SUCCESS in 25m 13s\n- swift-dsvm-functional-py3 https://logs.opendev.org/52/674052/1/check/swift-dsvm-functional-py3/1f86c26/ : SUCCESS in 25m 48s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/52/674052/1/check/swift-dsvm-functional-ipv6/8d56bd3/ : SUCCESS in 25m 07s\n- swift-tox-lower-constraints https://logs.opendev.org/52/674052/1/check/swift-tox-lower-constraints/0f123ba/ : SUCCESS in 11m 53s\n- openstack-tox-pep8 https://logs.opendev.org/52/674052/1/check/openstack-tox-pep8/fccfb50/ : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/52/674052/1/check/swift-multinode-rolling-upgrade/477a5b8/ : SUCCESS in 19m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c0203edae498127b3afda615b1d09dea0e3d4dbc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-02 04:16:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75f780f1b17ec847f112e888937b1aef882a2746","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 04:50:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/52/674052/1/check/openstack-tox-docs/c592754/html/ : SUCCESS in 4m 15s\n- swift-build-image https://logs.opendev.org/52/674052/1/check/swift-build-image/b7652c7/ : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/52/674052/1/check/swift-tox-py27/bf326a2/ : SUCCESS in 11m 10s\n- swift-tox-py36 https://logs.opendev.org/52/674052/1/check/swift-tox-py36/1346bf7/ : SUCCESS in 11m 22s\n- swift-tox-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-py37/e1eb68d/ : SUCCESS in 10m 36s\n- swift-tox-func https://logs.opendev.org/52/674052/1/check/swift-tox-func/d53624d/ : SUCCESS in 8m 24s\n- swift-tox-func-encryption https://logs.opendev.org/52/674052/1/check/swift-tox-func-encryption/187c191/ : SUCCESS in 8m 42s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/52/674052/1/check/swift-tox-func-domain-remap-staticweb/e5c1ea0/ : SUCCESS in 8m 36s\n- swift-tox-func-ec https://logs.opendev.org/52/674052/1/check/swift-tox-func-ec/9c8ff65/ : SUCCESS in 9m 17s\n- swift-tox-func-s3api https://logs.opendev.org/52/674052/1/check/swift-tox-func-s3api/d2ba7e6/ : SUCCESS in 11m 25s\n- swift-tox-func-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-func-py37/701749f/ : SUCCESS in 9m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-func-domain-remap-staticweb-py37/e9f5bd2/ : SUCCESS in 8m 24s\n- swift-tox-func-ec-py37 https://logs.opendev.org/52/674052/1/check/swift-tox-func-ec-py37/0bf75ff/ : SUCCESS in 8m 26s\n- swift-dsvm-functional https://logs.opendev.org/52/674052/1/check/swift-dsvm-functional/c06f0a1/ : SUCCESS in 28m 21s\n- swift-dsvm-functional-py3 https://logs.opendev.org/52/674052/1/check/swift-dsvm-functional-py3/d21fd5f/ : SUCCESS in 28m 54s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/52/674052/1/check/swift-dsvm-functional-ipv6/aa09030/ : SUCCESS in 27m 21s\n- swift-tox-lower-constraints https://logs.opendev.org/52/674052/1/check/swift-tox-lower-constraints/92d0371/ : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://logs.opendev.org/52/674052/1/check/openstack-tox-pep8/7937cdd/ : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/52/674052/1/check/swift-multinode-rolling-upgrade/fb74636/ : SUCCESS in 15m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3e84d532a18ac9ac135f32dc8d6c91050013c976","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 04:51:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ceccc12bffe26881f08e140aa21f2a8ac8cff29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 05:48:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3dc5ec105952458aa811310f895c7f1159557093","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 05:48:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/52/674052/1/gate/openstack-tox-docs/2496bf0/html/ : SUCCESS in 4m 03s\n- swift-upload-image https://logs.opendev.org/52/674052/1/gate/swift-upload-image/f6ba807/ : SUCCESS in 7m 51s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/52/674052/1/gate/swift-tox-py27/f665f8e/ : SUCCESS in 13m 57s\n- swift-tox-py36 https://logs.opendev.org/52/674052/1/gate/swift-tox-py36/95e5d8e/ : SUCCESS in 10m 25s\n- swift-tox-py37 https://logs.opendev.org/52/674052/1/gate/swift-tox-py37/d83e8b4/ : SUCCESS in 10m 34s\n- swift-tox-func https://logs.opendev.org/52/674052/1/gate/swift-tox-func/a413b81/ : SUCCESS in 9m 17s\n- swift-tox-func-encryption https://logs.opendev.org/52/674052/1/gate/swift-tox-func-encryption/7f7c917/ : SUCCESS in 9m 37s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/52/674052/1/gate/swift-tox-func-domain-remap-staticweb/0f79609/ : SUCCESS in 9m 52s\n- swift-tox-func-ec https://logs.opendev.org/52/674052/1/gate/swift-tox-func-ec/4c1dff3/ : SUCCESS in 12m 35s\n- swift-tox-func-s3api https://logs.opendev.org/52/674052/1/gate/swift-tox-func-s3api/ef18f09/ : SUCCESS in 11m 30s\n- swift-tox-func-py37 https://logs.opendev.org/52/674052/1/gate/swift-tox-func-py37/a7a0449/ : SUCCESS in 9m 32s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/52/674052/1/gate/swift-tox-func-domain-remap-staticweb-py37/d59d4be/ : SUCCESS in 11m 09s\n- swift-tox-func-ec-py37 https://logs.opendev.org/52/674052/1/gate/swift-tox-func-ec-py37/b585397/ : SUCCESS in 8m 12s\n- swift-dsvm-functional https://logs.opendev.org/52/674052/1/gate/swift-dsvm-functional/d47e297/ : SUCCESS in 29m 57s\n- swift-dsvm-functional-py3 https://logs.opendev.org/52/674052/1/gate/swift-dsvm-functional-py3/23cb6f5/ : SUCCESS in 28m 02s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/52/674052/1/gate/swift-dsvm-functional-ipv6/90fa54b/ : SUCCESS in 27m 40s\n- swift-tox-lower-constraints https://logs.opendev.org/52/674052/1/gate/swift-tox-lower-constraints/d848071/ : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://logs.opendev.org/52/674052/1/gate/openstack-tox-pep8/6d1f964/ : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9d189c362cef28a56ddedcb51ffd8a48c06320b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 05:49:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image https://logs.opendev.org/52/674052/1/promote/swift-promote-image/81c0e18/ : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7678e94a2c3907df13d271b0d4c82054bbb0bf8a","revisions":{"7678e94a2c3907df13d271b0d4c82054bbb0bf8a":{"kind":"REWORK","_number":1,"created":"2019-08-01 14:34:27.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/674052/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/674052/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/674052/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/674052/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/674052/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/674052/1"}}},"commit":{"parents":[{"commit":"a1455b3a3566a1a0d5680e540d19ba6dc417b1db","subject":"symlink: Allow symlinks to be created via chunk-encoded PUTs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1455b3a3566a1a0d5680e540d19ba6dc417b1db"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-08-01 14:31:56.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-08-01 14:34:03.000000000","tz":-300},"subject":"Negative test for non-empty chunked put symlink","message":"Negative test for non-empty chunked put symlink\n\nChange-Id: I0c3a235b8905e02d2e202215f7f0b43db295b79b\nRelated-Change-Id: I66f2f16d7cf0b9cb0888e3b9b1a17b9112233d21\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7678e94a2c3907df13d271b0d4c82054bbb0bf8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7678e94a2c3907df13d271b0d4c82054bbb0bf8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
