)]}'
{"id":"openstack%2Fswift~674705","triplet_id":"openstack%2Fswift~master~If62da0b4b34b9d1396b5838bf79ff494679f1ae3","project":"openstack/swift","branch":"master","topic":"674705","hashtags":[],"change_id":"If62da0b4b34b9d1396b5838bf79ff494679f1ae3","subject":"py3: decode stdout from backgrounded servers","status":"MERGED","created":"2019-08-05 21:51:29.000000000","updated":"2019-09-28 18:10:52.000000000","submitted":"2019-09-28 18:09:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674705-1569694187341-efb49f7a","meta_rev_id":"eb95a50749dcf839ed95b83b2acbfa29f3b9a88f","_number":674705,"virtual_id_number":674705,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-28 18:09:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-28 15:45:48.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-09-27 13:25:45.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2019-09-27 13:25:45.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 13:25:45.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-09-28 18:09:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e52603d5074a52936676fc2bf57bb73ccdb3b5e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-05 21:51:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f564feae3480a7b85610c2fe47f4e3eb1d8c61a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 23:20:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/05/674705/1/check/openstack-tox-docs/4c18e1d/html/ : SUCCESS in 6m 34s\n- grenade-py3 https://logs.opendev.org/05/674705/1/check/grenade-py3/64f85db/ : SUCCESS in 1h 25m 39s\n- tempest-integrated-object-storage https://logs.opendev.org/05/674705/1/check/tempest-integrated-object-storage/24fd412/ : SUCCESS in 49m 48s\n- swift-build-image https://logs.opendev.org/05/674705/1/check/swift-build-image/be23e0d/ : SUCCESS in 9m 31s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/05/674705/1/check/swift-tox-py27/2cb44f6/ : SUCCESS in 11m 19s\n- swift-tox-py36 https://logs.opendev.org/05/674705/1/check/swift-tox-py36/15f55c3/ : FAILURE in 15m 13s\n- swift-tox-py37 https://logs.opendev.org/05/674705/1/check/swift-tox-py37/c1a2626/ : FAILURE in 14m 16s\n- swift-tox-func https://logs.opendev.org/05/674705/1/check/swift-tox-func/b287138/ : SUCCESS in 19m 34s\n- swift-tox-func-encryption https://logs.opendev.org/05/674705/1/check/swift-tox-func-encryption/d7119dc/ : SUCCESS in 10m 27s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/05/674705/1/check/swift-tox-func-domain-remap-staticweb/e4cf108/ : SUCCESS in 12m 33s\n- swift-tox-func-ec https://logs.opendev.org/05/674705/1/check/swift-tox-func-ec/61ce70e/ : SUCCESS in 11m 54s\n- swift-tox-func-s3api https://logs.opendev.org/05/674705/1/check/swift-tox-func-s3api/5dfa8b7/ : SUCCESS in 11m 43s\n- swift-tox-func-py37 https://logs.opendev.org/05/674705/1/check/swift-tox-func-py37/ecb313f/ : SUCCESS in 19m 01s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/05/674705/1/check/swift-tox-func-domain-remap-staticweb-py37/3b42898/ : SUCCESS in 9m 07s\n- swift-tox-func-ec-py37 https://logs.opendev.org/05/674705/1/check/swift-tox-func-ec-py37/d2dd3ae/ : SUCCESS in 10m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/05/674705/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/169e0de/ : SUCCESS in 15m 26s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/05/674705/1/check/swift-probetests-centos-7/2a39521/ : SUCCESS in 47m 27s\n- swift-dsvm-functional https://logs.opendev.org/05/674705/1/check/swift-dsvm-functional/d8dbe23/ : SUCCESS in 24m 44s\n- swift-dsvm-functional-py3 https://logs.opendev.org/05/674705/1/check/swift-dsvm-functional-py3/cea4c10/ : SUCCESS in 27m 59s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/05/674705/1/check/swift-dsvm-functional-ipv6/9ab5e79/ : SUCCESS in 39m 34s\n- swift-tox-lower-constraints https://logs.opendev.org/05/674705/1/check/swift-tox-lower-constraints/cbd0b5c/ : SUCCESS in 11m 42s\n- openstack-tox-pep8 https://logs.opendev.org/05/674705/1/check/openstack-tox-pep8/cd28137/ : SUCCESS in 13m 05s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/05/674705/1/check/swift-multinode-rolling-upgrade/bf10603/ : SUCCESS in 17m 31s","accounts_in_message":[],"_revision_number":1},{"id":"32c740ef1b8896e8ef6dc9f6a29e442ce603fd57","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-06 17:58:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80d98ed24e9c135373ccc440fcc44ca798a738cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:53:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/05/674705/2/check/openstack-tox-docs/7688015/html/ : SUCCESS in 4m 43s\n- grenade-py3 https://logs.opendev.org/05/674705/2/check/grenade-py3/e27af1c/ : SUCCESS in 55m 09s\n- tempest-integrated-object-storage https://logs.opendev.org/05/674705/2/check/tempest-integrated-object-storage/837d68c/ : SUCCESS in 1h 01m 07s\n- swift-build-image https://logs.opendev.org/05/674705/2/check/swift-build-image/5f60be4/ : SUCCESS in 9m 49s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/05/674705/2/check/swift-tox-py27/ff6650d/ : SUCCESS in 13m 13s\n- swift-tox-py36 https://logs.opendev.org/05/674705/2/check/swift-tox-py36/78ca8b2/ : SUCCESS in 12m 47s\n- swift-tox-py37 https://logs.opendev.org/05/674705/2/check/swift-tox-py37/cc823a0/ : SUCCESS in 12m 16s\n- swift-tox-func https://logs.opendev.org/05/674705/2/check/swift-tox-func/06b255d/ : SUCCESS in 11m 00s\n- swift-tox-func-encryption https://logs.opendev.org/05/674705/2/check/swift-tox-func-encryption/71b1144/ : SUCCESS in 10m 28s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/05/674705/2/check/swift-tox-func-domain-remap-staticweb/4680dac/ : SUCCESS in 15m 47s\n- swift-tox-func-ec https://logs.opendev.org/05/674705/2/check/swift-tox-func-ec/102bc74/ : SUCCESS in 33m 49s\n- swift-tox-func-s3api https://logs.opendev.org/05/674705/2/check/swift-tox-func-s3api/52221eb/ : SUCCESS in 12m 45s\n- swift-tox-func-py37 https://logs.opendev.org/05/674705/2/check/swift-tox-func-py37/f7d0313/ : SUCCESS in 10m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/05/674705/2/check/swift-tox-func-domain-remap-staticweb-py37/996e026/ : SUCCESS in 13m 15s\n- swift-tox-func-ec-py37 https://logs.opendev.org/05/674705/2/check/swift-tox-func-ec-py37/2da4a60/ : SUCCESS in 9m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/05/674705/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/0f288fc/ : SUCCESS in 18m 26s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/05/674705/2/check/swift-probetests-centos-7/ba0717c/ : SUCCESS in 31m 50s\n- swift-dsvm-functional https://logs.opendev.org/05/674705/2/check/swift-dsvm-functional/9b9c58d/ : SUCCESS in 31m 17s\n- swift-dsvm-functional-py3 https://logs.opendev.org/05/674705/2/check/swift-dsvm-functional-py3/d6f91cc/ : SUCCESS in 31m 48s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/05/674705/2/check/swift-dsvm-functional-ipv6/ce90983/ : SUCCESS in 33m 55s\n- swift-tox-lower-constraints https://logs.opendev.org/05/674705/2/check/swift-tox-lower-constraints/448662c/ : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://logs.opendev.org/05/674705/2/check/openstack-tox-pep8/89e8e12/ : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/05/674705/2/check/swift-multinode-rolling-upgrade/199a85d/ : SUCCESS in 19m 08s","accounts_in_message":[],"_revision_number":2},{"id":"7df6fff027cc235180ddf7f90b4b0f6f69c25ed4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-14 02:26:13.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e26824bb102f82e8f848e5cf4e8559eba11b590e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-17 00:22:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57596056883b3e93ef1f9215561c43e4a400f4a6","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-09-26 22:12:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8041ed482d55be5c15d9894254cda0ee9874a15","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-27 05:06:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d2353315a8fe45e2b5ae0e326e6c8d5c62dfe8c2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-27 05:06:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8ff0255a1028e8253408a0773c68910699a0a866","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 06:25:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ade1bc2ec3446a78d2b708dfac15c0e : SUCCESS in 5m 30s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7736e195842848ca894c976397779f99 : SUCCESS in 1h 00m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a813c6d0f3ea472989454aa248ac59fc : SUCCESS in 1h 05m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5d958bd19b0c48f59bfe47bee483a946 : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3e25e02efcc8419cb4a7a887bdf2d1da : SUCCESS in 28m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fab6925f9a2447849213682adbe3e1a3 : SUCCESS in 12m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/72c4e56e7e974a15b28e077c9aef18bb : SUCCESS in 28m 30s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/c5b3dae99bf74fd4aa74dd6add94b80d : SUCCESS in 11m 03s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/c9b67e809791451390247672b83c9526 : SUCCESS in 26m 15s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/7b237ceaaad3474a8b415e8a901a1a4d : SUCCESS in 10m 18s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/32aeced6879c4c57b2864e9b55530c58 : SUCCESS in 18m 56s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/c396747296ad45ec9610ae3d8f193434 : SUCCESS in 14m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/eea0e4562ac74603a301efb24225dd6b : SUCCESS in 25m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dabf0ca4e6be4d7cae7b0f474b702b53 : SUCCESS in 11m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e50b4356f48f42b4b0356bb3eec3d333 : SUCCESS in 26m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3ec08f7fd03444d49b1a1e87be2f560f : SUCCESS in 11m 21s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f953d0c6ff9c451d93142c57c2a95ed7 : SUCCESS in 17m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bb79b0f7c5e54e4eadc852037414b6c9 : SUCCESS in 16m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fd7b69a1135142e184f90ad28ced2ff3 : SUCCESS in 52m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d61e602c5520415f971146d7e429129d : SUCCESS in 26m 58s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/faa7ecd4cedf418e91a2fe03fdf157b7 : TIMED_OUT in 50m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/873fffd9440f4e899318ee414405eae6 : SUCCESS in 27m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b96b52130a54d88a516b81a59e37ef1 : SUCCESS in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d5544b924214bd8ab3f62c01a54292b : SUCCESS in 5m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87acebb687aa45da8d66221217ab4cba : SUCCESS in 34m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdb42c74aee24df2b13db6bed307f2f3 : SUCCESS in 1h 01m 13s","accounts_in_message":[],"_revision_number":3},{"id":"75107ee7a275814a1ef612fc4e7b91dd865978e6","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-09-27 13:25:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"89caa5ac699ce658e81d40ce38c7a1dfa6a10d22","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-27 15:21:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"083b3324d6dd7bf85ac92980c68635072e4d7495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 17:14:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db4e70542c70492287001e388cb162a6 : SUCCESS in 7m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a32ba662fd0467ea5dd851d1e626fee : SUCCESS in 1h 06m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4fbd2746d66543499d7ae49568dfbeab : SUCCESS in 55m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52648f017dff4548adfe4648e3ce5819 : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be712c08229a42c4bbdeab54bccdbae5 : SUCCESS in 11m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2aa5a7e26eeb458386abdd490801157e : SUCCESS in 11m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3d543310aba8407d995bc163fb268309 : SUCCESS in 11m 26s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/06a73ddbdd99482a9a30df03fe254b8d : SUCCESS in 9m 38s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/a474bfc6b886446f839c745b5d74062e : SUCCESS in 12m 45s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/55aed5e4a8a2482d9ac264c8d542c233 : SUCCESS in 10m 15s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/84f5a52073414472bc02c79591e96190 : SUCCESS in 10m 07s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/fcf74b85affa49f6867d83a04bfca1e4 : SUCCESS in 12m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/336396d5366f43c59166ce6cb9366ad0 : SUCCESS in 10m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8daf7ee250224807a6ed56b98ff338e6 : SUCCESS in 9m 46s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8b78f2b81c5349919ca783ebf3c55552 : SUCCESS in 11m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e87c49d472f647ec8477eee278863e15 : SUCCESS in 12m 44s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/23735312bb4447aca8f98628a8111367 : SUCCESS in 15m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e4e836432ead45ff906f6cc0ea0a9074 : SUCCESS in 13m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/efe30326dc904c67a52c6eb6052035dd : SUCCESS in 31m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5d7d87659b34da6a2dfab3159121985 : SUCCESS in 28m 53s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fb6c01d0e31346fc80092a8ebfe250f8 : SUCCESS in 30m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/340db8b1843149499a7c5b13b169a8c3 : SUCCESS in 44m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d8f4d8d548a4a64a9cfe9c6204c42e4 : SUCCESS in 12m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/463f3ac9b8154c05a16d374945da4a33 : SUCCESS in 5m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a700800f4bf4446ab38d3eb4b8f29ae1 : SUCCESS in 23m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75c4b2433e8b4c009db7e00698e3a2e8 : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":3},{"id":"849f89f17667165f0e27e82784f2b6f03ff6695a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 17:14:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"72870ea82ab54b0c73871c28ae2256e18a19467f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 19:10:54.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea48d879cc54c6fb93f200b8062f89d : SUCCESS in 5m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/143efc8befd94c41b5175a1222307ca1 : SUCCESS in 1h 13m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/775358732468403e8e2fcbf714a51cc2 : SUCCESS in 1h 01m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/efa774ca03d04989ad159f3f5902c097 : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c3894126c240474bbde90ccbd357b150 : SUCCESS in 12m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a385e91913640f5bcb2757b630b803c : SUCCESS in 16m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e110edf629694f9f86a98cf877e0bcae : SUCCESS in 14m 53s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/a4fc02f7a6ca486daf068ccf4598de60 : SUCCESS in 9m 25s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/52b04a7b32654d5b99eb21f578c0483b : SUCCESS in 13m 36s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/c841f7fcdd8c4479b0e1796442ffb811 : SUCCESS in 9m 12s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/eaca648d764a47358576539bd82b32da : SUCCESS in 9m 25s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/da54790676f04439bcfbd9b3373a488d : SUCCESS in 12m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/42f605511e2c4cf890c6a374e471cbd1 : SUCCESS in 20m 59s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/17f980a48e0e42f3b75fb1c36301891b : SUCCESS in 11m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e6c04141c2e848b7aba6482f0bce63ee : SUCCESS in 12m 36s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7756e357a40748b0aa9306a3824ca4a1 : SUCCESS in 15m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3d5c4fe1714946588723df9a97d24909 : FAILURE in 30m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/93261c8ac3f3426ca7c428a173dfb24e : SUCCESS in 39m 31s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7cc9efaebd7642c68796a8bc761625e8 : SUCCESS in 36m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cbbf589bdff94fe08bb398699fe4daf1 : SUCCESS in 42m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6538acb1e5748adae2c5ea48da6335e : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c2903a8e4184b1580df9686ede698e8 : SUCCESS in 6m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b8bb8a73b8547e5914a5636511eab91 : SUCCESS in 1h 23m 44s","accounts_in_message":[],"_revision_number":3},{"id":"5a4b0e2371aec54ef21a2a3e4451dc1108f9ed84","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-27 19:13:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3358902ef6b8083a1b0f88eb0656f8fa61f9ff2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 20:43:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2942e82a8b94669aa2407e47e236153 : SUCCESS in 7m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3cffc8833dec4cb3abd6e79c621de2eb : SUCCESS in 1h 27m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ecfc2ddc8a894f32891cf715e3570fd7 : SUCCESS in 53m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/88855329c33f4fc68fa9804ae8285c3d : SUCCESS in 10m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2b196b7f9bef4c50b20f83d1e1255cfa : SUCCESS in 16m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/765ed8f2bf854398aa1d97f09b774dd0 : SUCCESS in 20m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/466d95a5ef4b43afaec05007dbe493d2 : SUCCESS in 12m 45s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/6f08c7b05f60428985804cbc7949139e : SUCCESS in 10m 26s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/15cefb0f54be4efaa74311a274709235 : SUCCESS in 8m 48s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/2400fbe2e89f44658f6f169e7a6a9066 : SUCCESS in 11m 16s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/e23f842cd881430080b999259316290b : SUCCESS in 10m 25s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/a9fcd6475d0b4dc5a2ad5ee705abda0d : SUCCESS in 17m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/50cbdb6868dd4a448ad7a11b90a01519 : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8605c8e6ecba4f9f925f8fb52ccbfecb : SUCCESS in 13m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/11f844b947394904b64cbdb4e840aaab : SUCCESS in 8m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d52c1795ff3345d4a5f3ab8e5346e940 : SUCCESS in 13m 55s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1a6bddadcbb046e0972a7a2b3d32c5fe : SUCCESS in 13m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0358726e01dc4629a05bac594c99fc2a : SUCCESS in 24m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9eaeb1cb5b834bb9ac27e489733732d0 : FAILURE in 38m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/98350af958c842febafa0c13b9e9e441 : SUCCESS in 28m 05s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/35cbadde916d4ea8ae19886e4bff6fc5 : SUCCESS in 29m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c426d6bb04924953aeee38e5f9691251 : SUCCESS in 25m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9c4bca09e0f4de5b8c43b8199da04bc : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/329107a22e134da499be36070c0869d8 : SUCCESS in 7m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/deea6441ac3a4a4b8314c6de36f76342 : SUCCESS in 16m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/198e79eb58e84d1783bddac998534ad3 : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":3},{"id":"49c591d2f400cee08dc843380c5960fbc035ef4e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-27 20:44:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6c7c4cbccfa01147e677d3a3b3c3c12ba61e5a88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 21:59:08.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f5f1ec143d4097b7410b8d9b92c780 : SUCCESS in 6m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d8f190f383aa468f800a213b6303dd76 : SUCCESS in 59m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5d026b763b840da8a211448e896a8a0 : SUCCESS in 51m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f3ab149114934cd483dad4307b477ac7 : SUCCESS in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/034f096cffdc4b46b6da2b91c0a742b4 : FAILURE in 13m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0101159d64a40c3b4a5f277cd927f65 : SUCCESS in 13m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/db05cd682da5451a96ec745cadf2bbc5 : SUCCESS in 26m 49s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/9d2be6d392c74db8bda820144c3a4984 : SUCCESS in 9m 35s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/4fd9365994804af4a8135e6dc45ce59a : SUCCESS in 10m 47s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/feeffb4e88b140c391a1aaf8d89373af : SUCCESS in 10m 14s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/7b25a5f568c04fc3a81eb82cd4b2102c : SUCCESS in 10m 35s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/8eea620394174ba6bbfcde7dd5a63e86 : SUCCESS in 14m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4309db755e51404fa637728000d45fc7 : SUCCESS in 10m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b7b584b62b6c4431a6fcb7e55b4682d3 : SUCCESS in 9m 44s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/46ea882fbd4c466bbc2a4402101e358a : SUCCESS in 8m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3ea95d73063f426e8d9231d78311599a : SUCCESS in 10m 34s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/53cadeeb814d4e76b105e6dc239a50cd : FAILURE in 16m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7769829b14f844d1b16a559487fecc99 : SUCCESS in 15m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2f718f4c66774e6abce842c849656fba : SUCCESS in 32m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/facfb83c1816478982c4780c4388f4e4 : SUCCESS in 31m 05s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f68332ed62114bcbb48334513affba5e : SUCCESS in 30m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a15ac3cb41b0407088b10fadc0ef34a7 : SUCCESS in 38m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90089bdf12f74c6e97a8732ce2c60874 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9edb3183641a4b1aa9802f27ca217181 : SUCCESS in 7m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d3c321d33ba3409dab1136f81c9d3a07 : SUCCESS in 14m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ad6645c71aa45f3836f3708720c700e : SUCCESS in 1h 08m 50s","accounts_in_message":[],"_revision_number":3},{"id":"6b6f588c1bc8eac81bc45a890121ed9d45318a63","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-27 22:05:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c037fcd0d41ba55b6931b1205da08d7dbbd87f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 23:21:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b326fa1c2ada4eb7b2de29072d72872a : SUCCESS in 6m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d162326d0e7642e0a9397d014be826ee : SUCCESS in 1h 02m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1a10bec781364a0f891e744d71a9a95f : SUCCESS in 54m 36s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c6a960c845e471688586dbeffc4d4d0 : SUCCESS in 8m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b772188c6b484f48bc63066cd0a22d11 : SUCCESS in 11m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a620a01105134f46b5e53f770dcbd307 : SUCCESS in 12m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/18b3f4fe77214580b37cde26b7b8039a : SUCCESS in 12m 46s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/a6d23a5690634ab39f13d9d5edecd9e9 : SUCCESS in 8m 43s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/494ca43c8f38460785b90a8a26ba5898 : SUCCESS in 10m 00s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/01e1dc4894de4fbda9f79c79b90918e9 : SUCCESS in 21m 33s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/9cabb28c1cba482a89ea6e75a26db485 : SUCCESS in 10m 54s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/96e1deb452e54a53ad91a25d84464c2b : SUCCESS in 27m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/87286c0d7f0243608b18c3cbd013c0dc : SUCCESS in 10m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f7dcd8df850b44d6a45c9ad519c13c6c : SUCCESS in 9m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/81d692251cc74fe888a38dae6539b1d2 : SUCCESS in 9m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/974752e1b7704abfbe829f6276352cc7 : SUCCESS in 9m 35s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3d01572b1ce2496da1852f7bde02c0a9 : SUCCESS in 14m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03f93f2b7b3149a9a8ab104e8bb04aa4 : SUCCESS in 13m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ec35a038f9b342229eacf67a2e9b5cef : SUCCESS in 28m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db0a4a32a8c44ebaaaee0cf1ab415ccb : SUCCESS in 29m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/71f2c5ce7b714dada6aa3cb995e6b1d0 : SUCCESS in 29m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e35baf08f9f14a62be9c25c5a80f2736 : SUCCESS in 27m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/138fc9295da04f719d6c29365fdc3d51 : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8576ced639e24771a3371e5df922c549 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e8b80082e204a6d96a9849e86532b61 : SUCCESS in 15m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57239fc5496b4dc3b2f8e17a45eb85b2 : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":3},{"id":"fb7ca5b1b9fd9600df37d94532ef8dab26d80ac8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 23:21:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"763d6954d439504a0a41a32baf1c7d7d97f34f00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 00:30:31.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53c287502024445b9bab3437d23d3507 : SUCCESS in 4m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d8698a6c8abc49e28a1a9f2cabb915ea : SUCCESS in 58m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bf19c0f616154c0b9edd7b83ea20c999 : SUCCESS in 54m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/013f90c827ce4e5a9571b6d2dd02da63 : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4bac459ec07a41a6bdc2297b828133fe : SUCCESS in 10m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9272745af4364bc489f4ec49464a41ed : SUCCESS in 12m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/90e812cdf40a4a6e94159c4bb4db860d : SUCCESS in 11m 36s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/044aed5ff3fb4da59a4bc08f6693760f : SUCCESS in 10m 26s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/e5e7bc21127848da81758fffa7b629c4 : SUCCESS in 9m 24s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/7b7e2b446b0c4dcf8347704509416a08 : SUCCESS in 10m 29s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/fa49d7a5e78c4713b5514d2f53557c0d : SUCCESS in 9m 35s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/f404668897b144dc8bbba88501dbea70 : SUCCESS in 12m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e876d402210443a3b61743f2ce18cf53 : SUCCESS in 9m 49s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/102bbe7f0b5f4944b2610d4c5e464b3e : SUCCESS in 9m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f1289090b92f4334bbfdb266b1ac2df6 : SUCCESS in 10m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0d673a7606a3438a8138643e59447ea9 : SUCCESS in 12m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/33c78336191743c193d2ad158a59b152 : FAILURE in 33m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3860ac66df74a059a60821aea185eb3 : SUCCESS in 29m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/702a92ede8a44787af3104f2c65bd819 : SUCCESS in 28m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/852ef995b03e4b20b7d9a75d57622d07 : SUCCESS in 30m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dfd3fcbb35b4a258842fb6195acd784 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f0baa6464f4ae18aa0dca9c58f7fb1 : SUCCESS in 5m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43fcdedbd8254836bbd7fc6a1cce36df : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":3},{"id":"374e8579d5cd296a98b131b14b04b7861a0dfd53","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 02:27:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2090cbde0cf05da6d0a60f3a9c82725ddcedde26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 03:40:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4f106838bc8450bb6349816e8b786df : SUCCESS in 10m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8a83600b0ca54cf0859df153fd059950 : SUCCESS in 59m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5257548807a84a818b7dd1798d55a262 : SUCCESS in 1h 00m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21e943476d164f2784087baf2f1ed77d : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83f385095dc24f068400c96d3933b641 : SUCCESS in 14m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3f3877420e43c1b277047e53ad0a2d : SUCCESS in 14m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c2cc1b6ed3844b00909d014a8c1b30cb : SUCCESS in 14m 28s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/7b661f2d116b4aab86017126ee46b8d1 : SUCCESS in 11m 41s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/b3fc18f1717d46909cf05277ccf794db : SUCCESS in 18m 14s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/0b5ac44ad0874700afd122e32d51725b : SUCCESS in 18m 07s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/fc62e17d1bd24245a23662a0627fe4b7 : SUCCESS in 12m 41s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/13461b9bf73946ca8ac078f74c5d38b6 : FAILURE in 15m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/53d4c00faca0493b9e4eee91432fd985 : SUCCESS in 13m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8f51a83d3e1e4b1e8ce720ef2ec4305f : SUCCESS in 19m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/bc78518148314629929519db420e4de8 : SUCCESS in 13m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b4706d3f65b648158d2371d7617fd0a6 : SUCCESS in 11m 22s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4733fb7fadf741a7a84f8400200bc4de : SUCCESS in 14m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b8adca878cdf46e885993ba2fe0fa371 : SUCCESS in 18m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eea9590ca04a47d88ac23e41b305cc4d : SUCCESS in 33m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8f7f1a6808e4b919b6e93f695f30e46 : SUCCESS in 37m 05s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fa986ada2e8c459ba392c9b6f7130c8c : SUCCESS in 29m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8886afb81f8e4a79b9ab04541b08184b : SUCCESS in 33m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91efdf401d6e4a6480c0e54dbd642a6d : SUCCESS in 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6328062c6d9442a780a1b58005df4ccb : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0d273b7c0ef48f38177c0d2f9fd0955 : SUCCESS in 22m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e98e70004ffd4024a22bc66e36bd7ecf : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":3},{"id":"c3d8c94e4ebf785844755d3c7f57567800af4565","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 03:42:03.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f3b72a49a50c6f3b0f68e7fcc335b7cef5d4bdee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 04:55:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/386ca9a36bc840a98c94cae799dcf552 : SUCCESS in 4m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d67079c780d94547a2e77b7fbe71062b : SUCCESS in 58m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4185856c8eef4d299af6cc21a3f7f836 : SUCCESS in 54m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/796198f223d84e2e8930fdeabf07bf40 : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d0876726d959475992209fea362b5280 : SUCCESS in 10m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/df4fae3928094bf09c4ea2a72c9a24f7 : SUCCESS in 11m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/160018dcff07432680cbffbd782b4298 : SUCCESS in 11m 52s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/3bd0e3b9cda048aba3510ee4524a2c19 : SUCCESS in 10m 30s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/e943d8628cbf46ed9e2c45aec6fb6679 : SUCCESS in 9m 18s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/aacd261519a74d7b88a34bff1fa1834b : SUCCESS in 9m 22s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/6c86aa52796c4acbba10334a8d9b1163 : SUCCESS in 10m 00s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/abe2f6ac8f994374a60b691852cc94b4 : SUCCESS in 13m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ff3aeafbe91143bfae225bdf6490572d : SUCCESS in 9m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6a9e217e05854f188c5eb5da006e1819 : SUCCESS in 13m 20s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e38a3ce010ad4c798bebab2afd6a83bc : SUCCESS in 9m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fd2f8c89503e4ad693399882db8af92a : SUCCESS in 11m 18s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a4dc7dbad67d4d2fb37f0bcace5b8333 : SUCCESS in 12m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e9b48ee0a5c4688b068691c4d685247 : SUCCESS in 13m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6659d710c7204ed1bd90c829324380e7 : FAILURE in 22m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/42bf9c2d9f9242d4be6c7a88f716fe54 : SUCCESS in 27m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/34198367d28e49858173e375b904a6b9 : SUCCESS in 28m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b41e7d846acf42a3a8422ccde6d602ba : SUCCESS in 27m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdceb5727d6a438f8f060adfcfde2a92 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb8fcbad1aa7470da36869722fc5767e : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b29db1543744f758cbdf411c6444f4d : SUCCESS in 19m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8ef5da498fc4bd18d62201103b8358d : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":3},{"id":"d775649d26282cd022c1307ad23bc2a7d9260624","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 05:11:08.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cb0dba4fb78d5754214c30e7d415be1d357e6d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 06:58:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d2d688a22142e496230f50994906f4 : SUCCESS in 6m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9507748191364bd39961ef3369fd351c : SUCCESS in 1h 01m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/535504983ced4fedbfc343e0e47be522 : FAILURE in 38m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5111396d2d1e4f8da090ba4b83905fca : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f960dcea7b74683bca27dec478b153c : SUCCESS in 12m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd6b62759b264dd0b85d3112497bc070 : SUCCESS in 12m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0c3d892fedbf49f5a3a4b7d414816af3 : SUCCESS in 11m 41s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/f31c4ef7392c4bc9b0f2f016cd5ab157 : SUCCESS in 9m 11s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/284120a5b19e48f6b913598eb9bc78de : SUCCESS in 11m 11s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/eadde9dde5d54da8afd0441fa07710b5 : SUCCESS in 15m 25s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/e9ab763a792341eab8c93491d8baaa80 : SUCCESS in 15m 40s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/a7bb7905d8574bf5a10e887d9facfaba : SUCCESS in 14m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6b5ba5e03f2243ae96587b6fa2ede1ee : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f1313132398e4b6ebf11fd71fff2dbc2 : SUCCESS in 11m 55s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f336279a69344f5ab2366f6a562f1cd9 : SUCCESS in 11m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a03e619e0ca54dd9989300e98e669706 : SUCCESS in 11m 22s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3ad28ed191a645019027743d986e05fa : SUCCESS in 15m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2faaf3c6d7594b5998587e771cfff1ec : SUCCESS in 17m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/354f7ee27f6f465e844cd9585dbb44d5 : FAILURE in 30m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c35f5044b94c4acc92271926541f39d3 : SUCCESS in 31m 48s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/21087d8760f14cf2a4208fb2bd36782c : SUCCESS in 35m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/36ec742119144b66bd3f202b5b29dfee : SUCCESS in 44m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e6118be5f194838818ef1d60af43630 : SUCCESS in 13m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/779bb314a2114a4eaa3c11349ee8f8df : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ffdb0118da544d8982e1488846468d9 : FAILURE in 26m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39acc1ac0fd141b7adb8333863e2cdf2 : SUCCESS in 1h 39m 15s","accounts_in_message":[],"_revision_number":3},{"id":"67ee3dfab04208c3ca5593a19d0d61917591ae32","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 15:45:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c2788af9e79ae30d456272d1d6969798e3b57edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 16:52:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/923d55b423d64ca7bf694e71036af710 : SUCCESS in 4m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1820924969574f11bea99af384723b99 : SUCCESS in 58m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cbfd9b4b0d254cd698318658fcec265e : SUCCESS in 51m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f50aeb085c84437e8da3a5b8690e03fc : SUCCESS in 8m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d136f1f8c80e46b9b29c469163e69ab9 : SUCCESS in 10m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1e27856cc8e94b39b4a0c9894a01e928 : SUCCESS in 11m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9e3983e243e54a449415ef0d331e0075 : SUCCESS in 11m 07s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/ec342562e6814efd941aee6790b9ad33 : SUCCESS in 8m 58s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/9a54b37bfd934e19a0819444c0c4cadd : SUCCESS in 9m 05s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/fa830f154dc54bf2a6360c636d2e4c84 : SUCCESS in 8m 51s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/2e1d8862463d4f3698666f607be92278 : SUCCESS in 9m 23s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e41235e9cb134fb2ab7e4eaf30a9b659 : SUCCESS in 12m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f45f6989a31544679bc0ac21c51c42f1 : SUCCESS in 10m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/876cfa90826040d5807615dc46ea093f : SUCCESS in 9m 24s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2a780958843545a3a89af49ec7bb45b8 : SUCCESS in 10m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4f121d8b0a0245aaa4452a3f76e1ccc0 : SUCCESS in 10m 18s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4f6badc017264fb98ec2f15bd97aa3c0 : SUCCESS in 11m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0ffa430395744177bbfca0c4b3f7298f : SUCCESS in 12m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b54d567cfff643f1b9b016d372caf8d4 : SUCCESS in 32m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bf7f1cc2b96346f7af89ab063f33a80d : SUCCESS in 28m 22s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/3f3f4b9cdda1418388a0f2e7c5376b27 : SUCCESS in 32m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fd6536d402ac48c08d5870f156bd75c6 : SUCCESS in 28m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6da2049fd9245cda2966f1e4002e7b8 : SUCCESS in 12m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373dc8a57df741df9f7681b841cbd87f : SUCCESS in 5m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a4587dc860e4639a2943594a917af98 : SUCCESS in 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16260c45e1d247e3910d18c43bfbfb07 : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":3},{"id":"3fa276e9a39b6bc0d32df8151d2df340ec1de3eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 16:52:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"de8b024450661f062b09968d2d6819eef49a9454","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 18:09:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"89a58382e3e93405f3338173a51571d1ccd6eb40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 18:09:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/528e99194ff44a6b924df6a7a923f5a4 : SUCCESS in 4m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8ee2b2a3557b4c528962390864644442 : SUCCESS in 59m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f345fa33b3224fbcbd8f40a54bcba3f9 : SUCCESS in 53m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c7a881fed57c4ba78d2bc06721884457 : SUCCESS in 8m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a4582834fffe4929863898b2c119b5f5 : SUCCESS in 12m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9f578f7a971843e6bb4e8cec62d2e023 : SUCCESS in 11m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/44245e932b054d78916141c5d56ae0d5 : SUCCESS in 11m 49s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/28c7764f4a4245448b8489888ddfc6e3 : SUCCESS in 13m 21s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/d7a3dfa709084c99b090da5e7bc34be1 : SUCCESS in 10m 24s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/6aa01a9bcacb4a62ad5d17092c1e53bc : SUCCESS in 9m 17s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/4b91675229d0410daae116d8fdc64ec6 : SUCCESS in 10m 20s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/0ec8bdd426224252994a69cfcbced863 : SUCCESS in 12m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0917b3f57be04f6f85b8b45489fefb1d : SUCCESS in 10m 45s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d2c557d3c2f240bd9944f234f1e65ef6 : SUCCESS in 9m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8abe34cf5fe9492d9c0d3e055cb4c92d : SUCCESS in 10m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2fbf0112b2634f55a8803399266780cb : SUCCESS in 12m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4bd63118d96b471c905a826ffc06797c : SUCCESS in 30m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15e71cec61734eb68e369b3e1b245745 : SUCCESS in 28m 46s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2dcc10cc99e748c7aa92dd145ac40ce3 : SUCCESS in 29m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95aa0470b8b94235835bb3519ffde90f : SUCCESS in 27m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2f5e096235f4c1099bea2ee0c0524a8 : SUCCESS in 25m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2191228bd44b47cc87edd70aed359f91 : SUCCESS in 5m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86aa2448bae94ec38e87d0210b91bc51 : SUCCESS in 1h 15m 41s","accounts_in_message":[],"_revision_number":3},{"id":"b80a3256a323f388280d7e49a7e2875e96e5c6d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 18:10:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4932a2a6677149ad8ff86e93f17f43f3 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ee786b7c09824b5faec19bafb12cd3e4 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ad7f7da32d6f90aa49873f1021d18cd54daef102","revisions":{"a8f8c12cc8db628a349dfd45a3552cbf5a04e7fc":{"kind":"REWORK","_number":1,"created":"2019-08-05 21:51:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/674705/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/674705/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/674705/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/674705/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/674705/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/674705/1"}}},"commit":{"parents":[{"commit":"489a35db825f58a204471ea9da9b9c61ca6cfa4c","subject":"Merge \"Make backport releases conform to existing CHANGELOG formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/489a35db825f58a204471ea9da9b9c61ca6cfa4c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-05 21:51:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-05 21:51:14.000000000","tz":-420},"subject":"py3: decode stdout from backgrounded servers","message":"py3: decode stdout from backgrounded servers\n\nOtherwise, when we go to print() it, we get a bunch of b\"\" strings.\n\nChange-Id: If62da0b4b34b9d1396b5838bf79ff494679f1ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8f8c12cc8db628a349dfd45a3552cbf5a04e7fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8f8c12cc8db628a349dfd45a3552cbf5a04e7fc"}]},"branch":"refs/heads/master"},"f3988382048b41d2b816e01789d80377a9c98757":{"kind":"REWORK","_number":2,"created":"2019-08-06 17:58:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/674705/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/674705/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/674705/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/674705/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/674705/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/674705/2"}}},"commit":{"parents":[{"commit":"489a35db825f58a204471ea9da9b9c61ca6cfa4c","subject":"Merge \"Make backport releases conform to existing CHANGELOG formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/489a35db825f58a204471ea9da9b9c61ca6cfa4c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-05 21:51:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 17:57:32.000000000","tz":-420},"subject":"py3: decode stdout from backgrounded servers","message":"py3: decode stdout from backgrounded servers\n\nOtherwise, when we go to print() it, we get a bunch of b\"\" strings.\n\nChange-Id: If62da0b4b34b9d1396b5838bf79ff494679f1ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3988382048b41d2b816e01789d80377a9c98757"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f3988382048b41d2b816e01789d80377a9c98757"}]},"branch":"refs/heads/master"},"ad7f7da32d6f90aa49873f1021d18cd54daef102":{"kind":"REWORK","_number":3,"created":"2019-09-27 05:06:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/674705/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/674705/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/674705/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/674705/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/674705/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/674705/3"}}},"commit":{"parents":[{"commit":"489a35db825f58a204471ea9da9b9c61ca6cfa4c","subject":"Merge \"Make backport releases conform to existing CHANGELOG formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/489a35db825f58a204471ea9da9b9c61ca6cfa4c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-05 21:51:14.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-27 05:04:38.000000000","tz":-420},"subject":"py3: decode stdout from backgrounded servers","message":"py3: decode stdout from backgrounded servers\n\nOtherwise, when we go to print() it, we get a bunch of b\"\" strings.\n\nChange-Id: If62da0b4b34b9d1396b5838bf79ff494679f1ae3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad7f7da32d6f90aa49873f1021d18cd54daef102"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad7f7da32d6f90aa49873f1021d18cd54daef102"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
