)]}'
{"id":"openstack%2Fswift~674716","triplet_id":"openstack%2Fswift~master~I5b337ef66a23fc989762801dd6a5ba1ed903f57b","project":"openstack/swift","branch":"master","topic":"py3-func-tests","hashtags":[],"change_id":"I5b337ef66a23fc989762801dd6a5ba1ed903f57b","subject":"py3: mostly port s3 func tests","status":"MERGED","created":"2019-08-06 00:37:03.000000000","updated":"2019-08-15 01:12:40.000000000","submitted":"2019-08-15 01:10:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674716-1565831429218-28760403","meta_rev_id":"2437ca8019c082c6f83be6f9c46923d751205765","_number":674716,"virtual_id_number":674716,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":2,"date":"2019-08-15 01:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-08-14 02:39:58.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2019-08-14 03:16:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-14 20:23:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 02:39:58.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-08-14 03:16:09.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-08-15 01:10:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2683feac31f997234a2b951b8e1c7b8cae401c7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-06 00:37:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d622b66c2215d64f3fed539ea08812395ddacc6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-06 00:55:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"296b1bd7c8ce488c5c3abd38c42dc767f1824608","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 02:01:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/16/674716/2/check/build-openstack-api-ref/95ba1f8/html/ : SUCCESS in 3m 40s\n- openstack-tox-docs https://logs.opendev.org/16/674716/2/check/openstack-tox-docs/fb850e9/html/ : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://logs.opendev.org/16/674716/2/check/build-openstack-releasenotes/fb8916d/html/ : SUCCESS in 6m 30s\n- grenade-py3 https://logs.opendev.org/16/674716/2/check/grenade-py3/04bfbdd/ : SUCCESS in 52m 25s\n- tempest-integrated-object-storage https://logs.opendev.org/16/674716/2/check/tempest-integrated-object-storage/4c191df/ : SUCCESS in 58m 38s\n- swift-build-image https://logs.opendev.org/16/674716/2/check/swift-build-image/a1f33ac/ : SUCCESS in 8m 14s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/16/674716/2/check/swift-tox-py27/9f942dc/ : SUCCESS in 12m 25s\n- swift-tox-py36 https://logs.opendev.org/16/674716/2/check/swift-tox-py36/303aed5/ : SUCCESS in 11m 12s\n- swift-tox-py37 https://logs.opendev.org/16/674716/2/check/swift-tox-py37/a15ac5f/ : SUCCESS in 11m 54s\n- swift-tox-func https://logs.opendev.org/16/674716/2/check/swift-tox-func/df71648/ : SUCCESS in 9m 02s\n- swift-tox-func-encryption https://logs.opendev.org/16/674716/2/check/swift-tox-func-encryption/05b0ec1/ : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/16/674716/2/check/swift-tox-func-domain-remap-staticweb/1df81c8/ : SUCCESS in 8m 57s\n- swift-tox-func-ec https://logs.opendev.org/16/674716/2/check/swift-tox-func-ec/dead5fc/ : SUCCESS in 9m 37s\n- swift-tox-func-s3api https://logs.opendev.org/16/674716/2/check/swift-tox-func-s3api/e046996/ : SUCCESS in 13m 13s\n- swift-tox-func-py37 https://logs.opendev.org/16/674716/2/check/swift-tox-func-py37/cf4f30b/ : SUCCESS in 10m 48s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/16/674716/2/check/swift-tox-func-domain-remap-staticweb-py37/32a085c/ : SUCCESS in 8m 25s\n- swift-tox-func-ec-py37 https://logs.opendev.org/16/674716/2/check/swift-tox-func-ec-py37/62b962f/ : SUCCESS in 8m 35s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/16/674716/2/check/swift-tox-func-s3api-py37/193edcf/ : FAILURE in 11m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/16/674716/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/33ea7f0/ : SUCCESS in 14m 00s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/16/674716/2/check/swift-probetests-centos-7/d63c366/ : SUCCESS in 31m 42s\n- swift-dsvm-functional https://logs.opendev.org/16/674716/2/check/swift-dsvm-functional/7a99010/ : SUCCESS in 27m 44s\n- swift-dsvm-functional-py3 https://logs.opendev.org/16/674716/2/check/swift-dsvm-functional-py3/7d7b384/ : SUCCESS in 27m 50s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/16/674716/2/check/swift-dsvm-functional-ipv6/a0d2d4c/ : SUCCESS in 27m 20s\n- swift-tox-lower-constraints https://logs.opendev.org/16/674716/2/check/swift-tox-lower-constraints/f15136d/ : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://logs.opendev.org/16/674716/2/check/openstack-tox-pep8/96932c3/ : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/16/674716/2/check/swift-multinode-rolling-upgrade/9ff96eb/ : SUCCESS in 17m 31s","accounts_in_message":[],"_revision_number":2},{"id":"a6a2a612d8b7e88f639144d290d39696b17dfea9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-06 16:50:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"114ba585a6a8cb5caf8b7635ab3b2d0372b53d7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:15:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/16/674716/3/check/build-openstack-api-ref/eca75bd/html/ : SUCCESS in 4m 06s\n- openstack-tox-docs https://logs.opendev.org/16/674716/3/check/openstack-tox-docs/0f9eabb/html/ : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://logs.opendev.org/16/674716/3/check/build-openstack-releasenotes/a60c606/html/ : SUCCESS in 8m 24s\n- grenade-py3 https://logs.opendev.org/16/674716/3/check/grenade-py3/8f5ce2e/ : SUCCESS in 58m 22s\n- tempest-integrated-object-storage https://logs.opendev.org/16/674716/3/check/tempest-integrated-object-storage/13c5c9a/ : SUCCESS in 1h 01m 04s\n- swift-build-image https://logs.opendev.org/16/674716/3/check/swift-build-image/c81c306/ : SUCCESS in 10m 12s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/16/674716/3/check/swift-tox-py27/49ac10a/ : SUCCESS in 10m 49s\n- swift-tox-py36 https://logs.opendev.org/16/674716/3/check/swift-tox-py36/ffbf35f/ : SUCCESS in 12m 33s\n- swift-tox-py37 https://logs.opendev.org/16/674716/3/check/swift-tox-py37/e2bbe2b/ : FAILURE in 13m 26s\n- swift-tox-func https://logs.opendev.org/16/674716/3/check/swift-tox-func/8335b65/ : SUCCESS in 14m 09s\n- swift-tox-func-encryption https://logs.opendev.org/16/674716/3/check/swift-tox-func-encryption/4b06904/ : SUCCESS in 10m 16s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/16/674716/3/check/swift-tox-func-domain-remap-staticweb/006e125/ : SUCCESS in 13m 45s\n- swift-tox-func-ec https://logs.opendev.org/16/674716/3/check/swift-tox-func-ec/f801d74/ : SUCCESS in 11m 12s\n- swift-tox-func-s3api https://logs.opendev.org/16/674716/3/check/swift-tox-func-s3api/3bbabf1/ : FAILURE in 17m 54s\n- swift-tox-func-py37 https://logs.opendev.org/16/674716/3/check/swift-tox-func-py37/493bc78/ : SUCCESS in 12m 32s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/16/674716/3/check/swift-tox-func-domain-remap-staticweb-py37/690031e/ : SUCCESS in 10m 34s\n- swift-tox-func-ec-py37 https://logs.opendev.org/16/674716/3/check/swift-tox-func-ec-py37/c4bbe26/ : SUCCESS in 10m 55s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/16/674716/3/check/swift-tox-func-s3api-py37/42781e2/ : SUCCESS in 10m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/16/674716/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/89a81ab/ : SUCCESS in 22m 59s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/16/674716/3/check/swift-probetests-centos-7/05fd067/ : SUCCESS in 30m 46s\n- swift-dsvm-functional https://logs.opendev.org/16/674716/3/check/swift-dsvm-functional/4d3ddc0/ : SUCCESS in 30m 11s\n- swift-dsvm-functional-py3 https://logs.opendev.org/16/674716/3/check/swift-dsvm-functional-py3/0d544a9/ : SUCCESS in 29m 32s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/16/674716/3/check/swift-dsvm-functional-ipv6/3c1659e/ : SUCCESS in 30m 55s\n- swift-tox-lower-constraints https://logs.opendev.org/16/674716/3/check/swift-tox-lower-constraints/e130ab1/ : SUCCESS in 11m 52s\n- openstack-tox-pep8 https://logs.opendev.org/16/674716/3/check/openstack-tox-pep8/9ab6598/ : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/16/674716/3/check/swift-multinode-rolling-upgrade/b552de2/ : SUCCESS in 23m 30s","accounts_in_message":[],"_revision_number":3},{"id":"c2fba9c4eb209b78be5664ad28c68937ad45a075","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-06 21:19:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1538bca1b556a45a79ff0b6e5536dc04552a949d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 00:43:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://logs.opendev.org/16/674716/3/check/build-openstack-api-ref/19dc5aa/html/ : SUCCESS in 2m 56s\n- openstack-tox-docs https://logs.opendev.org/16/674716/3/check/openstack-tox-docs/7a66a5e/html/ : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://logs.opendev.org/16/674716/3/check/build-openstack-releasenotes/d7ecacb/html/ : SUCCESS in 8m 59s\n- grenade-py3 https://logs.opendev.org/16/674716/3/check/grenade-py3/68651b4/ : SUCCESS in 1h 05m 57s\n- tempest-integrated-object-storage https://logs.opendev.org/16/674716/3/check/tempest-integrated-object-storage/dc8b405/ : SUCCESS in 1h 06m 42s\n- swift-build-image https://logs.opendev.org/16/674716/3/check/swift-build-image/f5c7786/ : SUCCESS in 12m 19s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/16/674716/3/check/swift-tox-py27/f30d2fd/ : SUCCESS in 16m 28s\n- swift-tox-py36 https://logs.opendev.org/16/674716/3/check/swift-tox-py36/27005b9/ : SUCCESS in 17m 29s\n- swift-tox-py37 https://logs.opendev.org/16/674716/3/check/swift-tox-py37/0545a40/ : SUCCESS in 12m 51s\n- swift-tox-func https://logs.opendev.org/16/674716/3/check/swift-tox-func/029351b/ : SUCCESS in 10m 18s\n- swift-tox-func-encryption https://logs.opendev.org/16/674716/3/check/swift-tox-func-encryption/6c27796/ : SUCCESS in 22m 30s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/16/674716/3/check/swift-tox-func-domain-remap-staticweb/5fa5b18/ : SUCCESS in 9m 32s\n- swift-tox-func-ec https://logs.opendev.org/16/674716/3/check/swift-tox-func-ec/d233930/ : SUCCESS in 11m 11s\n- swift-tox-func-s3api https://logs.opendev.org/16/674716/3/check/swift-tox-func-s3api/5911feb/ : SUCCESS in 12m 48s\n- swift-tox-func-py37 https://logs.opendev.org/16/674716/3/check/swift-tox-func-py37/e437eae/ : SUCCESS in 10m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/16/674716/3/check/swift-tox-func-domain-remap-staticweb-py37/9aee3b1/ : SUCCESS in 9m 14s\n- swift-tox-func-ec-py37 https://logs.opendev.org/16/674716/3/check/swift-tox-func-ec-py37/5dd4126/ : SUCCESS in 11m 11s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/16/674716/3/check/swift-tox-func-s3api-py37/7e5b686/ : SUCCESS in 11m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/16/674716/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/9ae4d95/ : SUCCESS in 13m 38s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/16/674716/3/check/swift-probetests-centos-7/3be85aa/ : SUCCESS in 32m 15s\n- swift-dsvm-functional https://logs.opendev.org/16/674716/3/check/swift-dsvm-functional/782475a/ : SUCCESS in 40m 58s\n- swift-dsvm-functional-py3 https://logs.opendev.org/16/674716/3/check/swift-dsvm-functional-py3/ce9e153/ : SUCCESS in 30m 12s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/16/674716/3/check/swift-dsvm-functional-ipv6/31b59a7/ : SUCCESS in 24m 51s\n- swift-tox-lower-constraints https://logs.opendev.org/16/674716/3/check/swift-tox-lower-constraints/4902d62/ : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://logs.opendev.org/16/674716/3/check/openstack-tox-pep8/1f93bb0/ : SUCCESS in 8m 14s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/16/674716/3/check/swift-multinode-rolling-upgrade/489fa12/ : SUCCESS in 21m 29s","accounts_in_message":[],"_revision_number":3},{"id":"af62e66b9ea2cb45776e9d60fc0b06c815f036b4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-07 23:19:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"94f8f0d71284e53822663431acdc7e970a1b5c1e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 02:38:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/16/674716/4/check/build-openstack-api-ref/f99b071/html/ : SUCCESS in 4m 12s\n- openstack-tox-docs https://logs.opendev.org/16/674716/4/check/openstack-tox-docs/a64f75d/html/ : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://logs.opendev.org/16/674716/4/check/build-openstack-releasenotes/13f5dc0/html/ : SUCCESS in 8m 15s\n- grenade-py3 https://logs.opendev.org/16/674716/4/check/grenade-py3/93ac2a9/ : SUCCESS in 1h 05m 56s\n- tempest-integrated-object-storage https://logs.opendev.org/16/674716/4/check/tempest-integrated-object-storage/5ee831b/ : SUCCESS in 57m 00s\n- swift-build-image https://logs.opendev.org/16/674716/4/check/swift-build-image/467df96/ : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/16/674716/4/check/swift-tox-py27/3c82b51/ : SUCCESS in 10m 58s\n- swift-tox-py36 https://logs.opendev.org/16/674716/4/check/swift-tox-py36/c98a10b/ : SUCCESS in 12m 24s\n- swift-tox-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-py37/8662b91/ : SUCCESS in 11m 38s\n- swift-tox-func https://logs.opendev.org/16/674716/4/check/swift-tox-func/680d53a/ : SUCCESS in 11m 43s\n- swift-tox-func-encryption https://logs.opendev.org/16/674716/4/check/swift-tox-func-encryption/72e7485/ : SUCCESS in 10m 31s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/16/674716/4/check/swift-tox-func-domain-remap-staticweb/39cb98e/ : SUCCESS in 10m 48s\n- swift-tox-func-ec https://logs.opendev.org/16/674716/4/check/swift-tox-func-ec/3106500/ : SUCCESS in 9m 44s\n- swift-tox-func-s3api https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api/54bfd84/ : SUCCESS in 12m 33s\n- swift-tox-func-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-py37/37301ff/ : POST_FAILURE in 10m 32s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-domain-remap-staticweb-py37/bc09ce5/ : SUCCESS in 9m 29s\n- swift-tox-func-ec-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-ec-py37/5745d54/ : SUCCESS in 8m 51s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api-py37/eaf7cfe/ : SUCCESS in 15m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/74b0dd9/ : SUCCESS in 17m 14s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/16/674716/4/check/swift-probetests-centos-7/001eb96/ : SUCCESS in 35m 01s\n- swift-dsvm-functional https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional/246ecb9/ : SUCCESS in 28m 13s\n- swift-dsvm-functional-py3 https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional-py3/8e2029b/ : SUCCESS in 33m 29s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional-ipv6/b3c4a67/ : SUCCESS in 30m 11s\n- swift-tox-lower-constraints https://logs.opendev.org/16/674716/4/check/swift-tox-lower-constraints/1a008f8/ : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://logs.opendev.org/16/674716/4/check/openstack-tox-pep8/6a426a0/ : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/16/674716/4/check/swift-multinode-rolling-upgrade/d819cd6/ : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":4},{"id":"ae161709eb8fc4fb986944fd4539462fbccbf432","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-08 15:11:31.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a58791953c8a615b7972ec11bf3b7a972e852cfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 20:55:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/16/674716/4/check/build-openstack-api-ref/8137b0f/html/ : SUCCESS in 4m 42s\n- openstack-tox-docs https://logs.opendev.org/16/674716/4/check/openstack-tox-docs/617f947/html/ : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://logs.opendev.org/16/674716/4/check/build-openstack-releasenotes/7b22d38/html/ : SUCCESS in 8m 24s\n- grenade-py3 https://logs.opendev.org/16/674716/4/check/grenade-py3/7e3fd93/ : SUCCESS in 1h 01m 17s\n- tempest-integrated-object-storage https://logs.opendev.org/16/674716/4/check/tempest-integrated-object-storage/3515e03/ : SUCCESS in 52m 45s\n- swift-build-image https://logs.opendev.org/16/674716/4/check/swift-build-image/da76015/ : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/16/674716/4/check/swift-tox-py27/ae4f4ff/ : SUCCESS in 11m 50s\n- swift-tox-py36 https://logs.opendev.org/16/674716/4/check/swift-tox-py36/e8877e8/ : SUCCESS in 15m 10s\n- swift-tox-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-py37/b055948/ : SUCCESS in 12m 49s\n- swift-tox-func https://logs.opendev.org/16/674716/4/check/swift-tox-func/d5124fc/ : SUCCESS in 11m 05s\n- swift-tox-func-encryption https://logs.opendev.org/16/674716/4/check/swift-tox-func-encryption/22e6813/ : SUCCESS in 10m 32s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/16/674716/4/check/swift-tox-func-domain-remap-staticweb/8c0a3d1/ : SUCCESS in 12m 16s\n- swift-tox-func-ec https://logs.opendev.org/16/674716/4/check/swift-tox-func-ec/7093cc6/ : SUCCESS in 12m 07s\n- swift-tox-func-s3api https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api/62d9c33/ : SUCCESS in 13m 27s\n- swift-tox-func-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-py37/3cacfb1/ : SUCCESS in 17m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-domain-remap-staticweb-py37/0faad5a/ : SUCCESS in 10m 03s\n- swift-tox-func-ec-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-ec-py37/2ed6926/ : SUCCESS in 10m 05s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api-py37/53af1ac/ : SUCCESS in 11m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/3b6ee93/ : SUCCESS in 13m 58s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/16/674716/4/check/swift-probetests-centos-7/3c939a6/ : FAILURE in 38m 40s\n- swift-dsvm-functional https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional/9e91bb3/ : SUCCESS in 34m 36s\n- swift-dsvm-functional-py3 https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional-py3/650fffa/ : SUCCESS in 29m 03s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional-ipv6/49171e6/ : SUCCESS in 29m 33s\n- swift-tox-lower-constraints https://logs.opendev.org/16/674716/4/check/swift-tox-lower-constraints/6a785d5/ : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://logs.opendev.org/16/674716/4/check/openstack-tox-pep8/b93e6b9/ : SUCCESS in 8m 21s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/16/674716/4/check/swift-multinode-rolling-upgrade/8299d19/ : SUCCESS in 19m 38s","accounts_in_message":[],"_revision_number":4},{"id":"60c8f487e675f9293efa860e110dab4a8721dc6d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-08 21:37:04.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a732871c463b22ba6f8e31f4cc10b5344db68bd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 01:23:01.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/16/674716/4/check/build-openstack-api-ref/9b40da7/html/ : SUCCESS in 5m 19s\n- openstack-tox-docs https://logs.opendev.org/16/674716/4/check/openstack-tox-docs/3428f98/html/ : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://logs.opendev.org/16/674716/4/check/build-openstack-releasenotes/3fb4cc5/html/ : SUCCESS in 7m 18s\n- grenade-py3 https://logs.opendev.org/16/674716/4/check/grenade-py3/0ae8eb1/ : SUCCESS in 1h 00m 18s\n- tempest-integrated-object-storage https://logs.opendev.org/16/674716/4/check/tempest-integrated-object-storage/d786204/ : SUCCESS in 56m 16s\n- swift-build-image https://logs.opendev.org/16/674716/4/check/swift-build-image/48d5d8d/ : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/16/674716/4/check/swift-tox-py27/47b5d96/ : SUCCESS in 15m 47s\n- swift-tox-py36 https://logs.opendev.org/16/674716/4/check/swift-tox-py36/fc7c289/ : SUCCESS in 12m 06s\n- swift-tox-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-py37/0dd9d0e/ : SUCCESS in 10m 42s\n- swift-tox-func https://logs.opendev.org/16/674716/4/check/swift-tox-func/0bf5adb/ : SUCCESS in 9m 28s\n- swift-tox-func-encryption https://logs.opendev.org/16/674716/4/check/swift-tox-func-encryption/f664e34/ : SUCCESS in 9m 56s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/16/674716/4/check/swift-tox-func-domain-remap-staticweb/72c92bd/ : SUCCESS in 12m 48s\n- swift-tox-func-ec https://logs.opendev.org/16/674716/4/check/swift-tox-func-ec/0fae9cd/ : SUCCESS in 10m 54s\n- swift-tox-func-s3api https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api/98d5041/ : SUCCESS in 12m 32s\n- swift-tox-func-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-py37/c20916c/ : SUCCESS in 9m 57s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-domain-remap-staticweb-py37/3473bb6/ : SUCCESS in 9m 08s\n- swift-tox-func-ec-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-ec-py37/900e47a/ : SUCCESS in 9m 26s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api-py37/db0cb56/ : SUCCESS in 9m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f939cf0/ : SUCCESS in 15m 05s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/16/674716/4/check/swift-probetests-centos-7/ef5e6ff/ : SUCCESS in 37m 33s\n- swift-dsvm-functional https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional/57b95db/ : TIMED_OUT in 47m 36s\n- swift-dsvm-functional-py3 https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional-py3/9eb8185/ : SUCCESS in 25m 19s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional-ipv6/38b9d9b/ : SUCCESS in 28m 52s\n- swift-tox-lower-constraints https://logs.opendev.org/16/674716/4/check/swift-tox-lower-constraints/1aeb2ab/ : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://logs.opendev.org/16/674716/4/check/openstack-tox-pep8/e05e204/ : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/16/674716/4/check/swift-multinode-rolling-upgrade/e144d6f/ : SUCCESS in 18m 41s","accounts_in_message":[],"_revision_number":4},{"id":"fd587a5744a342bc43fdf1359248107fec502a14","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-09 04:55:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"48a25a478be73d15433a78bb8362d4fe77f2fe24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 06:04:09.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/16/674716/4/check/build-openstack-api-ref/bff7d8c/html/ : SUCCESS in 4m 10s\n- openstack-tox-docs https://logs.opendev.org/16/674716/4/check/openstack-tox-docs/8bbd637/html/ : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://logs.opendev.org/16/674716/4/check/build-openstack-releasenotes/1aad861/html/ : SUCCESS in 7m 39s\n- grenade-py3 https://logs.opendev.org/16/674716/4/check/grenade-py3/5f3fbb0/ : SUCCESS in 1h 01m 01s\n- tempest-integrated-object-storage https://logs.opendev.org/16/674716/4/check/tempest-integrated-object-storage/243f3bf/ : SUCCESS in 55m 43s\n- swift-build-image https://logs.opendev.org/16/674716/4/check/swift-build-image/582e709/ : SUCCESS in 7m 35s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/16/674716/4/check/swift-tox-py27/796ca5e/ : SUCCESS in 11m 50s\n- swift-tox-py36 https://logs.opendev.org/16/674716/4/check/swift-tox-py36/6b5f11f/ : SUCCESS in 20m 05s\n- swift-tox-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-py37/17bc9f8/ : TIMED_OUT in 41m 01s\n- swift-tox-func https://logs.opendev.org/16/674716/4/check/swift-tox-func/8e5d761/ : SUCCESS in 9m 24s\n- swift-tox-func-encryption https://logs.opendev.org/16/674716/4/check/swift-tox-func-encryption/2ba442c/ : SUCCESS in 9m 47s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/16/674716/4/check/swift-tox-func-domain-remap-staticweb/54feace/ : SUCCESS in 10m 45s\n- swift-tox-func-ec https://logs.opendev.org/16/674716/4/check/swift-tox-func-ec/2cc3d67/ : SUCCESS in 25m 35s\n- swift-tox-func-s3api https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api/734d2c4/ : SUCCESS in 11m 43s\n- swift-tox-func-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-py37/f2a6f5f/ : SUCCESS in 10m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-domain-remap-staticweb-py37/10a1d7f/ : SUCCESS in 8m 57s\n- swift-tox-func-ec-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-ec-py37/511eae1/ : SUCCESS in 38m 05s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api-py37/7be24b2/ : SUCCESS in 12m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/16/674716/4/check/swift-tox-func-s3api-ceph-s3tests-tempauth/f5f18da/ : SUCCESS in 13m 43s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/16/674716/4/check/swift-probetests-centos-7/7b38b7f/ : SUCCESS in 49m 42s\n- swift-dsvm-functional https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional/599c3b7/ : SUCCESS in 28m 04s\n- swift-dsvm-functional-py3 https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional-py3/2a5673b/ : SUCCESS in 30m 27s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/16/674716/4/check/swift-dsvm-functional-ipv6/689b66e/ : SUCCESS in 44m 06s\n- swift-tox-lower-constraints https://logs.opendev.org/16/674716/4/check/swift-tox-lower-constraints/f5485ac/ : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://logs.opendev.org/16/674716/4/check/openstack-tox-pep8/6b8cb76/ : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/16/674716/4/check/swift-multinode-rolling-upgrade/d8702d5/ : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":4},{"id":"3e2ec3f8bc1ffd448d5de256965b5c57415fa9b2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-09 23:10:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3206eb9ae6bc2d13a3b5b71d9e4854648aa5a436","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 02:17:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/16/674716/5/check/build-openstack-api-ref/cabb5ca/html/ : SUCCESS in 3m 13s\n- openstack-tox-docs https://logs.opendev.org/16/674716/5/check/openstack-tox-docs/eeae8c0/html/ : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://logs.opendev.org/16/674716/5/check/build-openstack-releasenotes/6c8126e/html/ : SUCCESS in 8m 01s\n- grenade-py3 https://logs.opendev.org/16/674716/5/check/grenade-py3/8953168/ : SUCCESS in 1h 04m 35s\n- tempest-integrated-object-storage https://logs.opendev.org/16/674716/5/check/tempest-integrated-object-storage/5932f0a/ : SUCCESS in 55m 43s\n- swift-build-image https://logs.opendev.org/16/674716/5/check/swift-build-image/628efa0/ : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/16/674716/5/check/swift-tox-py27/26d498d/ : SUCCESS in 9m 54s\n- swift-tox-py36 https://logs.opendev.org/16/674716/5/check/swift-tox-py36/f7d664a/ : SUCCESS in 11m 05s\n- swift-tox-py37 https://logs.opendev.org/16/674716/5/check/swift-tox-py37/034677d/ : SUCCESS in 26m 33s\n- swift-tox-func https://logs.opendev.org/16/674716/5/check/swift-tox-func/04e6c8d/ : SUCCESS in 8m 34s\n- swift-tox-func-encryption https://logs.opendev.org/16/674716/5/check/swift-tox-func-encryption/26e767c/ : SUCCESS in 11m 41s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/16/674716/5/check/swift-tox-func-domain-remap-staticweb/34af198/ : SUCCESS in 9m 22s\n- swift-tox-func-ec https://logs.opendev.org/16/674716/5/check/swift-tox-func-ec/5261bf6/ : SUCCESS in 9m 50s\n- swift-tox-func-s3api https://logs.opendev.org/16/674716/5/check/swift-tox-func-s3api/af070e9/ : SUCCESS in 11m 30s\n- swift-tox-func-py37 https://logs.opendev.org/16/674716/5/check/swift-tox-func-py37/42f8e64/ : SUCCESS in 9m 28s\n- swift-tox-func-encryption-py37 https://logs.opendev.org/16/674716/5/check/swift-tox-func-encryption-py37/b010192/ : SUCCESS in 9m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/16/674716/5/check/swift-tox-func-domain-remap-staticweb-py37/bb129c4/ : SUCCESS in 10m 07s\n- swift-tox-func-ec-py37 https://logs.opendev.org/16/674716/5/check/swift-tox-func-ec-py37/2c53834/ : SUCCESS in 8m 31s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/16/674716/5/check/swift-tox-func-s3api-py37/b1ba604/ : SUCCESS in 8m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/16/674716/5/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b7d79c5/ : SUCCESS in 13m 12s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/16/674716/5/check/swift-probetests-centos-7/78512f8/ : FAILURE in 34m 01s\n- swift-dsvm-functional https://logs.opendev.org/16/674716/5/check/swift-dsvm-functional/03ec8fe/ : SUCCESS in 28m 32s\n- swift-dsvm-functional-py3 https://logs.opendev.org/16/674716/5/check/swift-dsvm-functional-py3/99815bf/ : SUCCESS in 31m 54s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/16/674716/5/check/swift-dsvm-functional-ipv6/ec0b13f/ : SUCCESS in 28m 21s\n- swift-tox-lower-constraints https://logs.opendev.org/16/674716/5/check/swift-tox-lower-constraints/123d2e9/ : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://logs.opendev.org/16/674716/5/check/openstack-tox-pep8/631e746/ : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/16/674716/5/check/swift-multinode-rolling-upgrade/e669371/ : SUCCESS in 21m 05s","accounts_in_message":[],"_revision_number":5},{"id":"9b8c8e31b82aedd8ae58ace00515ee6ed0586c5d","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-14 02:39:58.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2c90dedeb37c7638ec20a2f06b11b6e0dabd7877","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-08-14 03:16:09.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"90ba3eeda496688c57a2b99c15a327d941d9a27d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-14 20:23:54.000000000","message":"Patch Set 5: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"8fcb70360043ed473e2a9d6d28181f0ef7e435cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 21:59:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/86d498f468c640f5b7e3ad6149bdd3fd : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2533b036fd754ccb8bec20b057f6b38a : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41443735668745d4bdcb9525f3bb940d : SUCCESS in 7m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/81951e82d9f44f738e549623889cf1a9 : SUCCESS in 1h 00m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e6b0fd0f577457892b369afbc3030b7 : SUCCESS in 1h 00m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4cef522dc58646f3822b54da24e4f972 : SUCCESS in 8m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8dd0548f1cc14286a3399966dc3f4b65 : SUCCESS in 12m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2575b941b2844d7a8ddae3cf254110ac : SUCCESS in 12m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/37432aa831ed45a4b5d2adf61769f3e9 : SUCCESS in 10m 49s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/334be1b5fd7c405f89e9d6f9f4475df9 : SUCCESS in 8m 44s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/ce6fbe968bf5402f8c13fee34b8a0543 : SUCCESS in 11m 27s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/cecdbfa0a9e04e9f889191d8acf4061c : SUCCESS in 10m 21s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/e9e2368285964ea6aa2a4c8c0bbec5ef : SUCCESS in 11m 20s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/f28b6f7ea021441884c91b8d863b73ca : SUCCESS in 13m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5e1b025efa62418b8b2ec071179f1626 : SUCCESS in 11m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e4688647280742129b0bef3005f5b717 : SUCCESS in 11m 05s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/0f7d58428eb847a4a79999fd4419a175 : SUCCESS in 11m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3ed4e30d2d6f4922bddd03e935e9f4af : SUCCESS in 8m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/287a13d7b9a2423dbca401e14fe44679 : SUCCESS in 9m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0c944297194469eb7b65db0f48f3e1c : SUCCESS in 13m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e5c35e16076542ad9c45dbad1d1b6bfd : SUCCESS in 30m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e98d75310323472b88bb0f02cd8e9728 : SUCCESS in 38m 39s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e2b7ff149a584404b860e9fb4d6b2f3b : SUCCESS in 33m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ae41495530042d2950837c6704dc323 : SUCCESS in 30m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5335ca70a7704ad2b3c913af24935303 : SUCCESS in 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8b8289515e4a4eb621ec303acb44ad : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9387a4a5c5ee4856bb47b372a48ebc53 : SUCCESS in 16m 34s","accounts_in_message":[],"_revision_number":5},{"id":"542898e1d32dc6c129081dbbac522793100c80a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 22:00:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9c78306a55c002f4bc95d63ebc954912daec8878","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 01:10:28.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/474a0078b924443285558ffe08ff2384 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc3a8ba8f88342638d6c09540b469d94 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b23eda465ddb4c62a044480074f811f6 : SUCCESS in 11m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/eebb5543647a4a09a1f0578d764f8b4e : SUCCESS in 56m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d78d202ec27b4e7bb5264f2689bf469a : SUCCESS in 55m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4bb00b2f51014bcd954d42d04d911eb5 : SUCCESS in 10m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8e27eeb28e7043189aee955dc53bf2bc : SUCCESS in 11m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d4405ead941140798b979c0e272eb3d8 : SUCCESS in 10m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7d4803019aac4731b24ad0e6a2e9f65a : SUCCESS in 12m 46s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/0944a9e32b1344e6b89aef3dc00c3180 : SUCCESS in 9m 57s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/9de5cafda510435195a78facfdf48bf1 : SUCCESS in 10m 19s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/ca0f6e2ceea548d19c068fb134aa80d9 : SUCCESS in 9m 33s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/f7002c463d33417184a8b94c5f8f9014 : SUCCESS in 11m 31s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/10f80ddeff2a4d1f8bce592fe63ff104 : SUCCESS in 13m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/aa41a3412df247ffa5235600c0e4b9ce : SUCCESS in 11m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c2312734d136413085ec9b5b1922d8a3 : SUCCESS in 8m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c2aeb03adfc04a6b99a8dd81dd17224a : SUCCESS in 9m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/335a4ec91cef4de4b66b30d41beeb0ff : SUCCESS in 10m 58s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2f51e2d6ff18416c938ae680d48f9244 : SUCCESS in 32m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85e66705a9bf4ab89fe57190d2745039 : SUCCESS in 31m 13s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a4854b8313b0488a82590af795f03b05 : SUCCESS in 26m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9c86d8c4242c48f6b2a398011381e00f : SUCCESS in 27m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c0334093f9544198517291df1ea2608 : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f6a00807a9493a894d951b2892ad2b : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":5},{"id":"a1729cf8e18b570c7160e9cfa3759ec260ed0b8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 01:10:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"db06a5fa2fcb4d7cc52f6792f826f0372a3c4dab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 01:12:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/9e3c314ec3df45e583ad83636be4375d : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f05119c16fb33140a655b30e76734f481a479412","revisions":{"786b8320de19f0db1db2e044e32c6362bfdea3f4":{"kind":"REWORK","_number":1,"created":"2019-08-06 00:37:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/16/674716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/674716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/674716/1"}}},"commit":{"parents":[{"commit":"ab192415347223cfb49af7a62e124ef6f35779a6","subject":"Increase node_timeout for real sever unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab192415347223cfb49af7a62e124ef6f35779a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 00:33:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 00:36:43.000000000","tz":-420},"subject":"py3: mostly port s3 func tests","message":"py3: mostly port s3 func tests\n\ntest_bucket.py is proving somewhat problematic.\n\nChange-Id: I5b337ef66a23fc989762801dd6a5ba1ed903f57b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/786b8320de19f0db1db2e044e32c6362bfdea3f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/786b8320de19f0db1db2e044e32c6362bfdea3f4"}]},"branch":"refs/heads/master"},"644a1271be43115b5782b2df5bd634fee6f41474":{"kind":"REWORK","_number":2,"created":"2019-08-06 00:55:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/16/674716/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/674716/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/674716/2"}}},"commit":{"parents":[{"commit":"ab192415347223cfb49af7a62e124ef6f35779a6","subject":"Increase node_timeout for real sever unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab192415347223cfb49af7a62e124ef6f35779a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 00:33:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 00:55:15.000000000","tz":-420},"subject":"py3: mostly port s3 func tests","message":"py3: mostly port s3 func tests\n\ntest_bucket.py is proving somewhat problematic.\n\nChange-Id: I5b337ef66a23fc989762801dd6a5ba1ed903f57b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/644a1271be43115b5782b2df5bd634fee6f41474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/644a1271be43115b5782b2df5bd634fee6f41474"}]},"branch":"refs/heads/master"},"9b52dcbcaa79360b4024ae6cd182ffe84e2b727c":{"kind":"REWORK","_number":3,"created":"2019-08-06 16:50:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/16/674716/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/674716/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/674716/3"}}},"commit":{"parents":[{"commit":"ab192415347223cfb49af7a62e124ef6f35779a6","subject":"Increase node_timeout for real sever unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab192415347223cfb49af7a62e124ef6f35779a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 00:33:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 16:50:16.000000000","tz":-420},"subject":"py3: mostly port s3 func tests","message":"py3: mostly port s3 func tests\n\ntest_bucket.py is proving somewhat problematic.\n\nChange-Id: I5b337ef66a23fc989762801dd6a5ba1ed903f57b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b52dcbcaa79360b4024ae6cd182ffe84e2b727c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b52dcbcaa79360b4024ae6cd182ffe84e2b727c"}]},"branch":"refs/heads/master"},"e02a516bda9879e611cb8bf3fa34352e62d8cb76":{"kind":"REWORK","_number":4,"created":"2019-08-07 23:19:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/16/674716/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/674716/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/674716/4"}}},"commit":{"parents":[{"commit":"8d3d21e66846979d3bb4fbac136251da88f22310","subject":"Merge \"py3: port RBAC func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d3d21e66846979d3bb4fbac136251da88f22310"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 00:33:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-07 23:15:14.000000000","tz":-420},"subject":"py3: mostly port s3 func tests","message":"py3: mostly port s3 func tests\n\ntest_bucket.py is proving somewhat problematic.\n\nChange-Id: I5b337ef66a23fc989762801dd6a5ba1ed903f57b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e02a516bda9879e611cb8bf3fa34352e62d8cb76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e02a516bda9879e611cb8bf3fa34352e62d8cb76"}]},"branch":"refs/heads/master"},"f05119c16fb33140a655b30e76734f481a479412":{"kind":"REWORK","_number":5,"created":"2019-08-09 23:10:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/16/674716/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/674716/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/674716/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/674716/5"}}},"commit":{"parents":[{"commit":"394d4655fa17d0d4066fe8183570a5ce40ba42f5","subject":"Merge \"slo: Add X-Manifest-Etag to responses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/394d4655fa17d0d4066fe8183570a5ce40ba42f5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 00:33:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-09 23:09:55.000000000","tz":-420},"subject":"py3: mostly port s3 func tests","message":"py3: mostly port s3 func tests\n\ntest_bucket.py is proving somewhat problematic.\n\nChange-Id: I5b337ef66a23fc989762801dd6a5ba1ed903f57b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f05119c16fb33140a655b30e76734f481a479412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f05119c16fb33140a655b30e76734f481a479412"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
