)]}'
{"id":"openstack%2Fswift~674883","triplet_id":"openstack%2Fswift~master~I668b11251a3d1192d42dfe914010d089df413da7","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I668b11251a3d1192d42dfe914010d089df413da7","subject":"Use `is` to compare against sentinel object","status":"MERGED","created":"2019-08-06 17:15:05.000000000","updated":"2019-08-07 00:34:20.000000000","submitted":"2019-08-07 00:32:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674883-1565137973617-00db80d6","meta_rev_id":"5e66b517e31c7964b5b3c3e1caacdceea90cbb6d","_number":674883,"virtual_id_number":674883,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-08-07 00:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2019-08-06 17:17:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2019-08-06 17:17:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-06 17:17:17.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-08-07 00:32:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f49c60615ff985e386fb91ca00326f489aabcf8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-06 17:15:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41898a5b886f0958db49c80c4c042e654d245cff","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-08-06 17:17:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nBRILLIANT!","accounts_in_message":[],"_revision_number":1},{"id":"6c4a4a0b053d4fc8c0f3578070916b431fbf7fae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:48:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/83/674883/1/check/openstack-tox-docs/9cbef55/html/ : SUCCESS in 6m 08s\n- grenade-py3 https://logs.opendev.org/83/674883/1/check/grenade-py3/d77c9a5/ : SUCCESS in 1h 04m 48s\n- tempest-integrated-object-storage https://logs.opendev.org/83/674883/1/check/tempest-integrated-object-storage/807a655/ : SUCCESS in 58m 46s\n- swift-build-image https://logs.opendev.org/83/674883/1/check/swift-build-image/19ed307/ : SUCCESS in 11m 56s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/83/674883/1/check/swift-tox-py27/11678c5/ : SUCCESS in 15m 31s\n- swift-tox-py36 https://logs.opendev.org/83/674883/1/check/swift-tox-py36/38c349d/ : SUCCESS in 15m 42s\n- swift-tox-py37 https://logs.opendev.org/83/674883/1/check/swift-tox-py37/86210af/ : SUCCESS in 16m 02s\n- swift-tox-func https://logs.opendev.org/83/674883/1/check/swift-tox-func/61158dd/ : SUCCESS in 13m 52s\n- swift-tox-func-encryption https://logs.opendev.org/83/674883/1/check/swift-tox-func-encryption/aa8a330/ : SUCCESS in 12m 36s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/83/674883/1/check/swift-tox-func-domain-remap-staticweb/621bbee/ : SUCCESS in 12m 42s\n- swift-tox-func-ec https://logs.opendev.org/83/674883/1/check/swift-tox-func-ec/947f01f/ : SUCCESS in 15m 45s\n- swift-tox-func-s3api https://logs.opendev.org/83/674883/1/check/swift-tox-func-s3api/2904597/ : SUCCESS in 16m 09s\n- swift-tox-func-py37 https://logs.opendev.org/83/674883/1/check/swift-tox-func-py37/b192dbd/ : SUCCESS in 16m 12s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/83/674883/1/check/swift-tox-func-domain-remap-staticweb-py37/e1b2650/ : SUCCESS in 13m 29s\n- swift-tox-func-ec-py37 https://logs.opendev.org/83/674883/1/check/swift-tox-func-ec-py37/e01119e/ : SUCCESS in 12m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/83/674883/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/96b71a7/ : SUCCESS in 14m 50s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/83/674883/1/check/swift-probetests-centos-7/91baf1a/ : SUCCESS in 31m 38s\n- swift-dsvm-functional https://logs.opendev.org/83/674883/1/check/swift-dsvm-functional/7be6a2f/ : SUCCESS in 32m 11s\n- swift-dsvm-functional-py3 https://logs.opendev.org/83/674883/1/check/swift-dsvm-functional-py3/df673c7/ : SUCCESS in 32m 55s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/83/674883/1/check/swift-dsvm-functional-ipv6/2d9dcf5/ : SUCCESS in 29m 42s\n- swift-tox-lower-constraints https://logs.opendev.org/83/674883/1/check/swift-tox-lower-constraints/7339ec4/ : SUCCESS in 17m 34s\n- openstack-tox-pep8 https://logs.opendev.org/83/674883/1/check/openstack-tox-pep8/0a8d154/ : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/83/674883/1/check/swift-multinode-rolling-upgrade/448e54b/ : SUCCESS in 19m 13s","accounts_in_message":[],"_revision_number":1},{"id":"24fabdf5e29b939be526064728ed1b9f99cf1187","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 21:53:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87836850008035f650b3cee88d19e8cc4820cb07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 00:32:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/83/674883/1/gate/openstack-tox-docs/0107c0e/html/ : SUCCESS in 6m 08s\n- grenade-py3 https://logs.opendev.org/83/674883/1/gate/grenade-py3/78c2c73/ : SUCCESS in 57m 15s\n- tempest-integrated-object-storage https://logs.opendev.org/83/674883/1/gate/tempest-integrated-object-storage/3912b87/ : SUCCESS in 55m 22s\n- swift-upload-image https://logs.opendev.org/83/674883/1/gate/swift-upload-image/2f04e08/ : SUCCESS in 10m 46s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/83/674883/1/gate/swift-tox-py27/bbe555a/ : SUCCESS in 22m 18s\n- swift-tox-py36 https://logs.opendev.org/83/674883/1/gate/swift-tox-py36/7df3e8c/ : SUCCESS in 13m 36s\n- swift-tox-py37 https://logs.opendev.org/83/674883/1/gate/swift-tox-py37/2699efb/ : SUCCESS in 11m 28s\n- swift-tox-func https://logs.opendev.org/83/674883/1/gate/swift-tox-func/15f6b63/ : SUCCESS in 10m 02s\n- swift-tox-func-encryption https://logs.opendev.org/83/674883/1/gate/swift-tox-func-encryption/eb6da6e/ : SUCCESS in 12m 52s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/83/674883/1/gate/swift-tox-func-domain-remap-staticweb/2840734/ : SUCCESS in 9m 39s\n- swift-tox-func-ec https://logs.opendev.org/83/674883/1/gate/swift-tox-func-ec/aa14975/ : SUCCESS in 10m 53s\n- swift-tox-func-s3api https://logs.opendev.org/83/674883/1/gate/swift-tox-func-s3api/a120148/ : SUCCESS in 12m 30s\n- swift-tox-func-py37 https://logs.opendev.org/83/674883/1/gate/swift-tox-func-py37/8c89d34/ : SUCCESS in 9m 04s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/83/674883/1/gate/swift-tox-func-domain-remap-staticweb-py37/8a8b76f/ : SUCCESS in 8m 18s\n- swift-tox-func-ec-py37 https://logs.opendev.org/83/674883/1/gate/swift-tox-func-ec-py37/f15362a/ : SUCCESS in 10m 09s\n- swift-probetests-centos-7 https://logs.opendev.org/83/674883/1/gate/swift-probetests-centos-7/ec48cbf/ : SUCCESS in 33m 47s\n- swift-dsvm-functional https://logs.opendev.org/83/674883/1/gate/swift-dsvm-functional/b33f065/ : SUCCESS in 26m 28s\n- swift-dsvm-functional-py3 https://logs.opendev.org/83/674883/1/gate/swift-dsvm-functional-py3/0627e6a/ : SUCCESS in 32m 20s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/83/674883/1/gate/swift-dsvm-functional-ipv6/0f9c14a/ : SUCCESS in 27m 20s\n- swift-tox-lower-constraints https://logs.opendev.org/83/674883/1/gate/swift-tox-lower-constraints/8e8a682/ : SUCCESS in 13m 03s\n- openstack-tox-pep8 https://logs.opendev.org/83/674883/1/gate/openstack-tox-pep8/b6e1828/ : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1674e59b5518702c2fb3395abda69bb60fc69992","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 00:32:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5e66b517e31c7964b5b3c3e1caacdceea90cbb6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 00:34:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image https://logs.opendev.org/83/674883/1/promote/swift-promote-image/b72ee1c/ : SUCCESS in 57s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a10d936ee57c5eb331b56965fa79779e77978c84","revisions":{"a10d936ee57c5eb331b56965fa79779e77978c84":{"kind":"REWORK","_number":1,"created":"2019-08-06 17:15:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/83/674883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/674883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/674883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/674883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/674883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/674883/1"}}},"commit":{"parents":[{"commit":"ab192415347223cfb49af7a62e124ef6f35779a6","subject":"Increase node_timeout for real sever unittests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab192415347223cfb49af7a62e124ef6f35779a6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 17:14:47.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 17:14:47.000000000","tz":-420},"subject":"Use `is` to compare against sentinel object","message":"Use `is` to compare against sentinel object\n\n...since `\u003d\u003d` can have repercussions by way of __eq__()\n\nChange-Id: I668b11251a3d1192d42dfe914010d089df413da7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a10d936ee57c5eb331b56965fa79779e77978c84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a10d936ee57c5eb331b56965fa79779e77978c84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
