)]}'
{"id":"openstack%2Fswift~674901","triplet_id":"openstack%2Fswift~master~Iadb0a40092b8347eb5c04785cc14d1324cc9396f","project":"openstack/swift","branch":"master","topic":"674901","hashtags":[],"change_id":"Iadb0a40092b8347eb5c04785cc14d1324cc9396f","subject":"s3api: Translate 503 to S3-style 503s","status":"MERGED","created":"2019-08-06 18:12:03.000000000","updated":"2019-08-14 18:27:05.000000000","submitted":"2019-08-14 18:26:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674901-1565807161932-25fc65c6","meta_rev_id":"d4d57551d5157e8f1f36f49781a370a5c58bd527","_number":674901,"virtual_id_number":674901,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-08-14 18:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":2,"date":"2019-08-14 16:03:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":1,"date":"2019-08-14 16:03:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 16:03:54.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-08-14 18:26:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65e2f7b02a1ac3e81da48f4c79c07aaad5afc2aa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-06 18:12:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efbe0a19ad5df5dbd77d6348bf61de8581ba3dc1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-06 21:53:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ea8b3d253bfb1393714be63aa2806e9bb8ceae7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 01:00:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/01/674901/2/check/openstack-tox-docs/871d859/html/ : SUCCESS in 4m 29s\n- grenade-py3 https://logs.opendev.org/01/674901/2/check/grenade-py3/3439410/ : SUCCESS in 1h 10m 42s\n- tempest-integrated-object-storage https://logs.opendev.org/01/674901/2/check/tempest-integrated-object-storage/b5ef608/ : SUCCESS in 54m 14s\n- swift-build-image https://logs.opendev.org/01/674901/2/check/swift-build-image/00e82ce/ : SUCCESS in 7m 36s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/01/674901/2/check/swift-tox-py27/dbaaa7e/ : SUCCESS in 15m 49s\n- swift-tox-py36 https://logs.opendev.org/01/674901/2/check/swift-tox-py36/beac365/ : SUCCESS in 10m 57s\n- swift-tox-py37 https://logs.opendev.org/01/674901/2/check/swift-tox-py37/12a8f3a/ : SUCCESS in 12m 01s\n- swift-tox-func https://logs.opendev.org/01/674901/2/check/swift-tox-func/7a31097/ : SUCCESS in 9m 11s\n- swift-tox-func-encryption https://logs.opendev.org/01/674901/2/check/swift-tox-func-encryption/adf3b33/ : SUCCESS in 9m 42s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/01/674901/2/check/swift-tox-func-domain-remap-staticweb/31d8e22/ : SUCCESS in 8m 40s\n- swift-tox-func-ec https://logs.opendev.org/01/674901/2/check/swift-tox-func-ec/3de6870/ : SUCCESS in 23m 45s\n- swift-tox-func-s3api https://logs.opendev.org/01/674901/2/check/swift-tox-func-s3api/35912ec/ : SUCCESS in 12m 45s\n- swift-tox-func-py37 https://logs.opendev.org/01/674901/2/check/swift-tox-func-py37/c1d0de1/ : SUCCESS in 11m 34s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/01/674901/2/check/swift-tox-func-domain-remap-staticweb-py37/22049f6/ : SUCCESS in 8m 25s\n- swift-tox-func-ec-py37 https://logs.opendev.org/01/674901/2/check/swift-tox-func-ec-py37/2c94ef2/ : SUCCESS in 9m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/01/674901/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/e4486a4/ : SUCCESS in 16m 49s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/01/674901/2/check/swift-probetests-centos-7/6522903/ : SUCCESS in 35m 03s\n- swift-dsvm-functional https://logs.opendev.org/01/674901/2/check/swift-dsvm-functional/e82f7ab/ : SUCCESS in 27m 24s\n- swift-dsvm-functional-py3 https://logs.opendev.org/01/674901/2/check/swift-dsvm-functional-py3/66c7461/ : SUCCESS in 26m 29s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/01/674901/2/check/swift-dsvm-functional-ipv6/457e07a/ : SUCCESS in 28m 01s\n- swift-tox-lower-constraints https://logs.opendev.org/01/674901/2/check/swift-tox-lower-constraints/08f5fb6/ : SUCCESS in 13m 32s\n- openstack-tox-pep8 https://logs.opendev.org/01/674901/2/check/openstack-tox-pep8/a33697c/ : SUCCESS in 7m 13s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/01/674901/2/check/swift-multinode-rolling-upgrade/2f22d70/ : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ec85e2143df4cef7f3589e20bf5df4c9c32c5d61","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-14 16:03:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"a2a8aeb592e6f717c5d8f9a4a76e5d382fce44b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 16:04:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45e71b9bc13bc2ac53ef102907cf47af4a81dc67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 18:26:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754702bd09624a2199a0b7a49a6d5767 : SUCCESS in 4m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7932c32805b64e038f753ca939aaadc4 : SUCCESS in 1h 08m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4e046aeab4904864acf10017abb0255e : SUCCESS in 51m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ce7f56aed2834dc990691080a4dcfb53 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/379e465edd4840b0a9aa5156edd023b0 : SUCCESS in 12m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee1f578aa6ba4186a722b923226de01f : SUCCESS in 11m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/51dd43a3d7a447f0b9c924be420226e8 : SUCCESS in 10m 44s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/e76294eeb0694586ab4e308b3de0c04d : SUCCESS in 17m 05s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/09b7beccc1d141cb8af8446048a5f4e2 : SUCCESS in 9m 47s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/cbc1c8313f834338b36a17cef352dc45 : SUCCESS in 11m 48s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/7b6dd01122304989907b569cfaa383bc : SUCCESS in 12m 11s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/004f063737ab49aab9a73fca5f08d6b8 : SUCCESS in 13m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/192097581bf84fc2ac08d2e9cce43df9 : SUCCESS in 15m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/82a3c7e924434db3b9eaf3734e795785 : SUCCESS in 8m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/92e430851458450595b7ec6d771a8a20 : SUCCESS in 9m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/30dff79bfeb04ce7bb39c99421ea7b7a : SUCCESS in 35m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3ba7e516f7748f99946d08ce0be4a87 : SUCCESS in 28m 30s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/01c0059fdd0f4e59b3dd73310d0e3efd : SUCCESS in 34m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f387443d0a6442f7b2ccc927dd267239 : SUCCESS in 32m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df915ab130ed46109d6013bd47c4dc6b : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/515b22c5a8ff48b4b3342ada3b34c758 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":2},{"id":"e30f81cb542d37b48c32b1041d4ccd0a863fb5ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 18:26:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fc0a13f737f1edbaec17eaa301391662d4c8d967","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 18:27:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b9728a54c7554a1384eac0e38fccf721 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7bb7da6d2dc4267f8279170b6d326e81777d3c06","revisions":{"2b2043e8490517e9d1a97c0cbc4ab1ede2fe4f2e":{"kind":"REWORK","_number":1,"created":"2019-08-06 18:12:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/674901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/674901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/674901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/674901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/674901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/674901/1"}}},"commit":{"parents":[{"commit":"489a35db825f58a204471ea9da9b9c61ca6cfa4c","subject":"Merge \"Make backport releases conform to existing CHANGELOG formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/489a35db825f58a204471ea9da9b9c61ca6cfa4c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 18:11:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 18:11:52.000000000","tz":-420},"subject":"s3api: Translate 503 to S3-style 503s","message":"s3api: Translate 503 to S3-style 503s\n\n...instead of logging tracebacks about unexpected status codes.\n\nChange-Id: Iadb0a40092b8347eb5c04785cc14d1324cc9396f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b2043e8490517e9d1a97c0cbc4ab1ede2fe4f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b2043e8490517e9d1a97c0cbc4ab1ede2fe4f2e"}]},"branch":"refs/heads/master"},"7bb7da6d2dc4267f8279170b6d326e81777d3c06":{"kind":"REWORK","_number":2,"created":"2019-08-06 21:53:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/01/674901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/01/674901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/674901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/674901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/01/674901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/01/674901/2"}}},"commit":{"parents":[{"commit":"489a35db825f58a204471ea9da9b9c61ca6cfa4c","subject":"Merge \"Make backport releases conform to existing CHANGELOG formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/489a35db825f58a204471ea9da9b9c61ca6cfa4c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 18:11:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-06 21:52:54.000000000","tz":-420},"subject":"s3api: Translate 503 to S3-style 503s","message":"s3api: Translate 503 to S3-style 503s\n\n...instead of logging tracebacks about unexpected status codes.\n\nChange-Id: Iadb0a40092b8347eb5c04785cc14d1324cc9396f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bb7da6d2dc4267f8279170b6d326e81777d3c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bb7da6d2dc4267f8279170b6d326e81777d3c06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
