)]}'
{"id":"openstack%2Fswift~675227","triplet_id":"openstack%2Fswift~master~I4bb59edd87336597791416c4f2a096efe0e72fe3","project":"openstack/swift","branch":"master","topic":"py3-func-tests","hashtags":[],"change_id":"I4bb59edd87336597791416c4f2a096efe0e72fe3","subject":"py3: Finish porting func tests","status":"MERGED","created":"2019-08-07 23:19:28.000000000","updated":"2019-08-22 08:17:46.000000000","submitted":"2019-08-22 08:13:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675227-1566461621435-4a7959b6","meta_rev_id":"98f563eac0f42d5404db328b4f275d0280b3ac3f","_number":675227,"virtual_id_number":675227,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2019-08-22 08:13:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-08-22 01:00:09.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2019-08-22 01:00:09.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 01:00:09.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-08-22 08:13:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"078b19bae473b1d64606d4920ad5eac0719c45a4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-07 23:19:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b0a802439d2b0e7d6582059450088f3a873e4d3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 02:29:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/27/675227/1/check/build-openstack-api-ref/9f8609d/html/ : SUCCESS in 3m 38s\n- openstack-tox-docs https://logs.opendev.org/27/675227/1/check/openstack-tox-docs/acbe4a4/html/ : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://logs.opendev.org/27/675227/1/check/build-openstack-releasenotes/d6b20e9/html/ : SUCCESS in 8m 14s\n- grenade-py3 https://logs.opendev.org/27/675227/1/check/grenade-py3/6e60f33/ : SUCCESS in 55m 08s\n- tempest-integrated-object-storage https://logs.opendev.org/27/675227/1/check/tempest-integrated-object-storage/c3f69b3/ : SUCCESS in 51m 50s\n- swift-build-image https://logs.opendev.org/27/675227/1/check/swift-build-image/23a1970/ : SUCCESS in 10m 54s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/27/675227/1/check/swift-tox-py27/d28e090/ : SUCCESS in 11m 42s\n- swift-tox-py36 https://logs.opendev.org/27/675227/1/check/swift-tox-py36/3c7b37e/ : SUCCESS in 14m 48s\n- swift-tox-py37 https://logs.opendev.org/27/675227/1/check/swift-tox-py37/a9f69f1/ : SUCCESS in 14m 03s\n- swift-tox-func https://logs.opendev.org/27/675227/1/check/swift-tox-func/c2fcf3f/ : SUCCESS in 10m 20s\n- swift-tox-func-encryption https://logs.opendev.org/27/675227/1/check/swift-tox-func-encryption/5cbd392/ : SUCCESS in 8m 52s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/27/675227/1/check/swift-tox-func-domain-remap-staticweb/9457757/ : SUCCESS in 11m 28s\n- swift-tox-func-ec https://logs.opendev.org/27/675227/1/check/swift-tox-func-ec/73f6d76/ : SUCCESS in 11m 13s\n- swift-tox-func-s3api https://logs.opendev.org/27/675227/1/check/swift-tox-func-s3api/09ea24b/ : SUCCESS in 13m 07s\n- swift-tox-func-py37 https://logs.opendev.org/27/675227/1/check/swift-tox-func-py37/77adbdc/ : SUCCESS in 11m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/27/675227/1/check/swift-tox-func-domain-remap-staticweb-py37/d04ffe8/ : SUCCESS in 7m 23s\n- swift-tox-func-ec-py37 https://logs.opendev.org/27/675227/1/check/swift-tox-func-ec-py37/d6250b1/ : SUCCESS in 9m 22s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/27/675227/1/check/swift-tox-func-s3api-py37/2c65560/ : FAILURE in 13m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/27/675227/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/c6514f8/ : SUCCESS in 15m 10s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/27/675227/1/check/swift-probetests-centos-7/53464f7/ : SUCCESS in 35m 57s\n- swift-dsvm-functional https://logs.opendev.org/27/675227/1/check/swift-dsvm-functional/fda1b2d/ : SUCCESS in 33m 41s\n- swift-dsvm-functional-py3 https://logs.opendev.org/27/675227/1/check/swift-dsvm-functional-py3/841df2a/ : SUCCESS in 31m 02s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/27/675227/1/check/swift-dsvm-functional-ipv6/eba6103/ : SUCCESS in 29m 17s\n- swift-tox-lower-constraints https://logs.opendev.org/27/675227/1/check/swift-tox-lower-constraints/aa90016/ : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://logs.opendev.org/27/675227/1/check/openstack-tox-pep8/fe32904/ : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/27/675227/1/check/swift-multinode-rolling-upgrade/f267523/ : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b5bccd15f0706bedfdbf074b8cbbe19b4287805d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-08 15:11:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f0c2a31fb6088428197d9ce94898861b3861726f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-08 15:44:10.000000000","message":"Patch Set 1:\n\n:-/\n\nSo alternatively, we could move the monkey patch that\u0027s currently at https://github.com/openstack/swift/blob/2.22.0/test/functional/__init__.py#L73 ahead of the import at https://github.com/openstack/swift/blob/2.22.0/test/functional/__init__.py#L44-L45 ...\n\nBut as for *why* that suddenly works, I have no idea...","accounts_in_message":[],"_revision_number":1},{"id":"18db6630663070afe932673f8e512d85dc57e5fe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-08 16:54:17.000000000","message":"Patch Set 1:\n\nOh! Moving the monkey-patch is starting to make sense. boto(core) isn\u0027t importing directly from ssl, but rather from some places buried in urllib3. swift_test_client import from swiftclient which\u0027ll drag in requests and urllib3, so... they aren\u0027t currently getting greened.","accounts_in_message":[],"_revision_number":1},{"id":"59c7443b6d372f11a32befb5096de2ae0f840536","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-08 17:06:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"399e668457af5622337365e807bbfee04efe1ee5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 00:36:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/27/675227/2/check/build-openstack-api-ref/fd78934/html/ : SUCCESS in 3m 40s\n- openstack-tox-docs https://logs.opendev.org/27/675227/2/check/openstack-tox-docs/f9b8151/html/ : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://logs.opendev.org/27/675227/2/check/build-openstack-releasenotes/dfe7c45/html/ : SUCCESS in 9m 03s\n- grenade-py3 https://logs.opendev.org/27/675227/2/check/grenade-py3/9e2f0a2/ : SUCCESS in 1h 00m 06s\n- tempest-integrated-object-storage https://logs.opendev.org/27/675227/2/check/tempest-integrated-object-storage/4fd7b45/ : SUCCESS in 52m 59s\n- swift-build-image https://logs.opendev.org/27/675227/2/check/swift-build-image/44f67cf/ : SUCCESS in 9m 30s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/27/675227/2/check/swift-tox-py27/0f438c4/ : SUCCESS in 11m 00s\n- swift-tox-py36 https://logs.opendev.org/27/675227/2/check/swift-tox-py36/b181d2f/ : SUCCESS in 10m 32s\n- swift-tox-py37 https://logs.opendev.org/27/675227/2/check/swift-tox-py37/3e32e76/ : SUCCESS in 19m 48s\n- swift-tox-func https://logs.opendev.org/27/675227/2/check/swift-tox-func/4e9132c/ : SUCCESS in 10m 22s\n- swift-tox-func-encryption https://logs.opendev.org/27/675227/2/check/swift-tox-func-encryption/8584ae6/ : SUCCESS in 9m 51s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/27/675227/2/check/swift-tox-func-domain-remap-staticweb/6113cb8/ : SUCCESS in 8m 43s\n- swift-tox-func-ec https://logs.opendev.org/27/675227/2/check/swift-tox-func-ec/30b5776/ : SUCCESS in 9m 49s\n- swift-tox-func-s3api https://logs.opendev.org/27/675227/2/check/swift-tox-func-s3api/c4cfc02/ : SUCCESS in 12m 54s\n- swift-tox-func-py37 https://logs.opendev.org/27/675227/2/check/swift-tox-func-py37/5dd3cbd/ : SUCCESS in 11m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/27/675227/2/check/swift-tox-func-domain-remap-staticweb-py37/7126c31/ : SUCCESS in 10m 24s\n- swift-tox-func-ec-py37 https://logs.opendev.org/27/675227/2/check/swift-tox-func-ec-py37/4c085f5/ : SUCCESS in 8m 12s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/27/675227/2/check/swift-tox-func-s3api-py37/26ad462/ : FAILURE in 33m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/27/675227/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/7661890/ : SUCCESS in 13m 20s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/27/675227/2/check/swift-probetests-centos-7/d1d3262/ : FAILURE in 28m 45s\n- swift-dsvm-functional https://logs.opendev.org/27/675227/2/check/swift-dsvm-functional/f62dac4/ : SUCCESS in 28m 40s\n- swift-dsvm-functional-py3 https://logs.opendev.org/27/675227/2/check/swift-dsvm-functional-py3/fee2f87/ : SUCCESS in 32m 44s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/27/675227/2/check/swift-dsvm-functional-ipv6/748c10c/ : SUCCESS in 30m 05s\n- swift-tox-lower-constraints https://logs.opendev.org/27/675227/2/check/swift-tox-lower-constraints/08f9a5b/ : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://logs.opendev.org/27/675227/2/check/openstack-tox-pep8/58ca969/ : SUCCESS in 7m 50s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/27/675227/2/check/swift-multinode-rolling-upgrade/351b477/ : SUCCESS in 22m 19s","accounts_in_message":[],"_revision_number":2},{"id":"1856df46b7d62d24a234cc2865e97eb33c3cb41b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-09 04:54:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5e83b44e403ec473e94b10ab8e0f5f1561ded309","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 05:57:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://logs.opendev.org/27/675227/2/check/build-openstack-api-ref/70d79ce/html/ : SUCCESS in 4m 03s\n- openstack-tox-docs https://logs.opendev.org/27/675227/2/check/openstack-tox-docs/d89ea49/html/ : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://logs.opendev.org/27/675227/2/check/build-openstack-releasenotes/ec6cb82/html/ : SUCCESS in 7m 39s\n- grenade-py3 https://logs.opendev.org/27/675227/2/check/grenade-py3/99167e5/ : SUCCESS in 55m 36s\n- tempest-integrated-object-storage https://logs.opendev.org/27/675227/2/check/tempest-integrated-object-storage/05f2c98/ : SUCCESS in 55m 42s\n- swift-build-image https://logs.opendev.org/27/675227/2/check/swift-build-image/c1b625f/ : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/27/675227/2/check/swift-tox-py27/543a29d/ : SUCCESS in 23m 56s\n- swift-tox-py36 https://logs.opendev.org/27/675227/2/check/swift-tox-py36/f7da7b8/ : SUCCESS in 11m 48s\n- swift-tox-py37 https://logs.opendev.org/27/675227/2/check/swift-tox-py37/dedefca/ : SUCCESS in 14m 04s\n- swift-tox-func https://logs.opendev.org/27/675227/2/check/swift-tox-func/e00d9a8/ : SUCCESS in 9m 16s\n- swift-tox-func-encryption https://logs.opendev.org/27/675227/2/check/swift-tox-func-encryption/f40c6a3/ : SUCCESS in 13m 03s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/27/675227/2/check/swift-tox-func-domain-remap-staticweb/97a0ef7/ : SUCCESS in 40m 38s\n- swift-tox-func-ec https://logs.opendev.org/27/675227/2/check/swift-tox-func-ec/ba7556e/ : SUCCESS in 12m 19s\n- swift-tox-func-s3api https://logs.opendev.org/27/675227/2/check/swift-tox-func-s3api/3371f6e/ : SUCCESS in 15m 01s\n- swift-tox-func-py37 https://logs.opendev.org/27/675227/2/check/swift-tox-func-py37/a8f55fe/ : SUCCESS in 13m 19s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/27/675227/2/check/swift-tox-func-domain-remap-staticweb-py37/6e657c5/ : SUCCESS in 20m 34s\n- swift-tox-func-ec-py37 https://logs.opendev.org/27/675227/2/check/swift-tox-func-ec-py37/e61c5b2/ : SUCCESS in 10m 23s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/27/675227/2/check/swift-tox-func-s3api-py37/2615fdb/ : SUCCESS in 14m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/27/675227/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/2f9528b/ : SUCCESS in 14m 23s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/27/675227/2/check/swift-probetests-centos-7/ede956a/ : SUCCESS in 36m 25s\n- swift-dsvm-functional https://logs.opendev.org/27/675227/2/check/swift-dsvm-functional/41dac3b/ : SUCCESS in 27m 54s\n- swift-dsvm-functional-py3 https://logs.opendev.org/27/675227/2/check/swift-dsvm-functional-py3/c23f15b/ : SUCCESS in 30m 07s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/27/675227/2/check/swift-dsvm-functional-ipv6/ba8bb2f/ : SUCCESS in 30m 18s\n- swift-tox-lower-constraints https://logs.opendev.org/27/675227/2/check/swift-tox-lower-constraints/5bcbc2d/ : SUCCESS in 30m 50s\n- openstack-tox-pep8 https://logs.opendev.org/27/675227/2/check/openstack-tox-pep8/e522d2d/ : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/27/675227/2/check/swift-multinode-rolling-upgrade/9632a7d/ : SUCCESS in 17m 43s","accounts_in_message":[],"_revision_number":2},{"id":"d4abd0ff8a88ff9d4adc938620a06a4f5e16cd5c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-09 23:10:00.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5fcc12f574a1851b6672945c9396faf3de175ce5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 02:16:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://logs.opendev.org/27/675227/3/check/build-openstack-api-ref/978dabd/html/ : SUCCESS in 3m 16s\n- openstack-tox-docs https://logs.opendev.org/27/675227/3/check/openstack-tox-docs/3471e2b/html/ : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://logs.opendev.org/27/675227/3/check/build-openstack-releasenotes/a8f73ae/html/ : SUCCESS in 7m 43s\n- grenade-py3 https://logs.opendev.org/27/675227/3/check/grenade-py3/82a59d4/ : SUCCESS in 1h 03m 42s\n- tempest-integrated-object-storage https://logs.opendev.org/27/675227/3/check/tempest-integrated-object-storage/8a3c20a/ : SUCCESS in 54m 10s\n- swift-build-image https://logs.opendev.org/27/675227/3/check/swift-build-image/4e923ea/ : SUCCESS in 10m 17s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/27/675227/3/check/swift-tox-py27/ae52295/ : SUCCESS in 11m 15s\n- swift-tox-py36 https://logs.opendev.org/27/675227/3/check/swift-tox-py36/6e3d9ac/ : SUCCESS in 10m 49s\n- swift-tox-py37 https://logs.opendev.org/27/675227/3/check/swift-tox-py37/f44d7de/ : SUCCESS in 10m 23s\n- swift-tox-func https://logs.opendev.org/27/675227/3/check/swift-tox-func/2d6d48e/ : SUCCESS in 8m 57s\n- swift-tox-func-encryption https://logs.opendev.org/27/675227/3/check/swift-tox-func-encryption/846fd13/ : SUCCESS in 9m 13s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/27/675227/3/check/swift-tox-func-domain-remap-staticweb/0ac9e55/ : SUCCESS in 10m 20s\n- swift-tox-func-ec https://logs.opendev.org/27/675227/3/check/swift-tox-func-ec/3781abb/ : SUCCESS in 10m 12s\n- swift-tox-func-s3api https://logs.opendev.org/27/675227/3/check/swift-tox-func-s3api/c62ad6f/ : SUCCESS in 12m 36s\n- swift-tox-func-py37 https://logs.opendev.org/27/675227/3/check/swift-tox-func-py37/8490f3f/ : SUCCESS in 9m 48s\n- swift-tox-func-encryption-py37 https://logs.opendev.org/27/675227/3/check/swift-tox-func-encryption-py37/aa01b5d/ : SUCCESS in 8m 23s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/27/675227/3/check/swift-tox-func-domain-remap-staticweb-py37/e6f5b96/ : SUCCESS in 7m 46s\n- swift-tox-func-ec-py37 https://logs.opendev.org/27/675227/3/check/swift-tox-func-ec-py37/d2ab1b7/ : SUCCESS in 7m 23s\n- swift-tox-func-s3api-py37 https://logs.opendev.org/27/675227/3/check/swift-tox-func-s3api-py37/bb7e338/ : SUCCESS in 14m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/27/675227/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/0b3864a/ : SUCCESS in 12m 25s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/27/675227/3/check/swift-probetests-centos-7/9bf3894/ : SUCCESS in 35m 05s\n- swift-dsvm-functional https://logs.opendev.org/27/675227/3/check/swift-dsvm-functional/3157b76/ : SUCCESS in 39m 20s\n- swift-dsvm-functional-py3 https://logs.opendev.org/27/675227/3/check/swift-dsvm-functional-py3/c667855/ : SUCCESS in 27m 28s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/27/675227/3/check/swift-dsvm-functional-ipv6/91fbb8a/ : SUCCESS in 27m 34s\n- swift-tox-lower-constraints https://logs.opendev.org/27/675227/3/check/swift-tox-lower-constraints/e04eebd/ : SUCCESS in 20m 40s\n- openstack-tox-pep8 https://logs.opendev.org/27/675227/3/check/openstack-tox-pep8/ddddc00/ : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/27/675227/3/check/swift-multinode-rolling-upgrade/2f27bd6/ : SUCCESS in 16m 20s","accounts_in_message":[],"_revision_number":3},{"id":"c4bd804b67c550a53251879cb5ba505c1097d679","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-17 03:34:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"140a2f22cb43df37728249d13887b1c653656c3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 04:36:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fab787963ab9499e9620dcf34a0d86fd : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/416c38a946004c4e960c844f04fae8ee : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9962b52f44d4f79919a8f54ea25d858 : SUCCESS in 7m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5fe2faf016b84fffbc90f97a47cdde7e : SUCCESS in 1h 01m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f4a266ed5ff4a9aa933124b62296687 : SUCCESS in 50m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/56a145ffdce44eae923114bbc6541170 : SUCCESS in 11m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca6b405de3d842a79a34cf9d745a54ea : SUCCESS in 11m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f5f22e093bf4cbbb2e78bf6995289e6 : SUCCESS in 17m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0ee964ecbabf4b01a55a595b19f7f13d : SUCCESS in 12m 36s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/c3be9d913fbe41d7b9b4a6d6e156d7aa : SUCCESS in 16m 13s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/a3f4054ade104e26b0413e5187fa04cc : SUCCESS in 8m 58s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/cfd9248a08f549119eed6f9cee221f59 : SUCCESS in 16m 17s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/1c4e20e893fd419db462c4e5144ac193 : SUCCESS in 10m 04s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/dc259e90fc8b41c7a77ee7f61f5eaffe : SUCCESS in 15m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3e926a37db5e463392608c6885392a34 : SUCCESS in 19m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3b13d403324a46c1950c4f063c016e60 : SUCCESS in 17m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/59e0a2cabaed4adc8cd7f72d853e70cb : SUCCESS in 10m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e430c755a13743b499d5b48dec1f8acb : SUCCESS in 11m 35s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3da3e028bd1a4f09bbf74dd0882509d1 : SUCCESS in 20m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a5371fd9549477b9e9c54eb44cb6f7c : SUCCESS in 15m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/251726a5e4f74a9d9f6dcf6fec909792 : SUCCESS in 38m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bceda42dd7064e73a1cca4fb4f644d9e : SUCCESS in 26m 16s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/eb230fd06251449cb01b3c5f675d3281 : SUCCESS in 35m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1d22507b30454b56927185524eb8501e : SUCCESS in 31m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15bd99415abb4d4da02b1cc81f54416a : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/262b04d582504d2d93efa12eec990169 : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9a1d5436a67640e7b4a16e9ef39d3f6d : SUCCESS in 28m 39s","accounts_in_message":[],"_revision_number":4},{"id":"82dd1597e051d081f44185225c63969454f48099","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-22 01:00:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nwhat a nightmare","accounts_in_message":[],"_revision_number":4},{"id":"0f23f438b37086eae54c84e38f7c53de8f64ae9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 01:58:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ead8941140f77f481e2853a9e95b70d9625839ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:13:40.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/97c04e472b3e4fcfb458f4c4db516b89 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b69a53d2d9cb4ce993d9493572ab6c53 : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fa9d36898f74d6481c395fe7ae52b42 : SUCCESS in 10m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/82165bdb76f2443094c0dd8539ba9fe2 : SUCCESS in 51m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6fb7c57cd71849d68e385b14232c2f00 : SUCCESS in 59m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/04167e5d80534101898d7e940d4a201e : SUCCESS in 10m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a3f24538e5143ba85569154bc01da45 : SUCCESS in 10m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/583a87841d714b85a4717cb8b163d8b0 : SUCCESS in 13m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e8bef868cfaa458aae699c62ada584de : SUCCESS in 10m 35s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/3e8de8360ba242af8e543731b097a1a3 : SUCCESS in 10m 23s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/0dab160860c947469b5b27ccff5314fa : SUCCESS in 11m 02s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/ba9842897f634541a04fdad09edb3b07 : SUCCESS in 10m 00s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/9d5a97f6116547d081431c6c46787cca : SUCCESS in 10m 18s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/93b9e25b647942f2bcba597fd4c1119c : SUCCESS in 12m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c79d7a53ab544dec8599a09d8b70b210 : SUCCESS in 11m 13s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b8cadf8f760f40b2a00b143aa2ee0459 : SUCCESS in 10m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8b4565624cba496bab047f6d3bf9140a : SUCCESS in 14m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/e315efbea3904b26b0f58cb72764c9ee : SUCCESS in 13m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e2f0be6939494059bc119241940dae2d : SUCCESS in 29m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0bc8df3ec9d477bbd06162502eb0d25 : SUCCESS in 29m 52s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e9ea807ffdd94cd8a99dc444f3bc5641 : SUCCESS in 30m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/170c17303cd14576b1c1aa0a7b0fad03 : SUCCESS in 28m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98d1c8d971894763aae34e77d6d800b4 : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/130ac672b3964f02a4bcd5c085b1408c : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":4},{"id":"4225dd38f75c38ab87a86dda29ae7d75f9629701","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:13:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"6fe066b04baa7d3182b991ca427ef71467336e68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 08:17:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/7fd8934ed1e54d1da246da29f2633798 : SUCCESS in 1m 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af2ccd96ecc4c58a48f209b2604a795 : SUCCESS in 1m 19s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ac6c81a5dec64d9ba98309b3d79168f3 : SUCCESS in 1m 41s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"74db3670607d952e597011eb07676aedff521b41","revisions":{"f4a0e1bc222bf6dab4b0fec9e9abbfb73ba57dc1":{"kind":"REWORK","_number":1,"created":"2019-08-07 23:19:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/675227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/675227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/675227/1"}}},"commit":{"parents":[{"commit":"e02a516bda9879e611cb8bf3fa34352e62d8cb76","subject":"py3: mostly port s3 func tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e02a516bda9879e611cb8bf3fa34352e62d8cb76"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-07 23:16:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-07 23:19:16.000000000","tz":-420},"subject":"py3: Finish porting s3 func tests","message":"py3: Finish porting s3 func tests\n\nChange-Id: I4bb59edd87336597791416c4f2a096efe0e72fe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4a0e1bc222bf6dab4b0fec9e9abbfb73ba57dc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4a0e1bc222bf6dab4b0fec9e9abbfb73ba57dc1"}]},"branch":"refs/heads/master"},"3ce2c45ed6c6331a0ae9de9649eead71dda2a80c":{"kind":"REWORK","_number":2,"created":"2019-08-08 17:06:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/675227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/675227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/675227/2"}}},"commit":{"parents":[{"commit":"e02a516bda9879e611cb8bf3fa34352e62d8cb76","subject":"py3: mostly port s3 func tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e02a516bda9879e611cb8bf3fa34352e62d8cb76"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-07 23:16:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-08 16:57:11.000000000","tz":-420},"subject":"py3: Finish porting s3 func tests","message":"py3: Finish porting s3 func tests\n\nWe were (indirectly) importing swiftclient (and therefore requests and\nurllib3) before doing our eventlet monkey-patching. This would lead\nboto3 (which digs an SSLContext out of urllib3) to trip RecursionErrors\non py3 similar to\n\n   \u003e\u003e\u003e from ssl import SSLContext, PROTOCOL_SSLv23\n   \u003e\u003e\u003e import eventlet\n   \u003e\u003e\u003e eventlet.monkey_patch(socket\u003dTrue)\n   \u003e\u003e\u003e SSLContext(PROTOCOL_SSLv23).options |\u003d 0\n   Traceback (most recent call last):\n     File \"\u003cstdin\u003e\", line 1, in \u003cmodule\u003e\n     File \"/usr/lib/python3.6/ssl.py\", line 465, in options\n       super(SSLContext, SSLContext).options.__set__(self, value)\n     File \"/usr/lib/python3.6/ssl.py\", line 465, in options\n       super(SSLContext, SSLContext).options.__set__(self, value)\n     File \"/usr/lib/python3.6/ssl.py\", line 465, in options\n       super(SSLContext, SSLContext).options.__set__(self, value)\n     [Previous line repeated 330 more times]\n   RecursionError: maximum recursion depth exceeded while calling a Python object\n\nChange-Id: I4bb59edd87336597791416c4f2a096efe0e72fe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ce2c45ed6c6331a0ae9de9649eead71dda2a80c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ce2c45ed6c6331a0ae9de9649eead71dda2a80c"}]},"branch":"refs/heads/master"},"6c465101dc46fa86b0094ea51fbece4a39b714f6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-09 23:10:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/675227/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/675227/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/675227/3"}}},"commit":{"parents":[{"commit":"f05119c16fb33140a655b30e76734f481a479412","subject":"py3: mostly port s3 func tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f05119c16fb33140a655b30e76734f481a479412"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-07 23:16:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-09 23:09:55.000000000","tz":-420},"subject":"py3: Finish porting s3 func tests","message":"py3: Finish porting s3 func tests\n\nWe were (indirectly) importing swiftclient (and therefore requests and\nurllib3) before doing our eventlet monkey-patching. This would lead\nboto3 (which digs an SSLContext out of urllib3) to trip RecursionErrors\non py3 similar to\n\n   \u003e\u003e\u003e from ssl import SSLContext, PROTOCOL_SSLv23\n   \u003e\u003e\u003e import eventlet\n   \u003e\u003e\u003e eventlet.monkey_patch(socket\u003dTrue)\n   \u003e\u003e\u003e SSLContext(PROTOCOL_SSLv23).options |\u003d 0\n   Traceback (most recent call last):\n     File \"\u003cstdin\u003e\", line 1, in \u003cmodule\u003e\n     File \"/usr/lib/python3.6/ssl.py\", line 465, in options\n       super(SSLContext, SSLContext).options.__set__(self, value)\n     File \"/usr/lib/python3.6/ssl.py\", line 465, in options\n       super(SSLContext, SSLContext).options.__set__(self, value)\n     File \"/usr/lib/python3.6/ssl.py\", line 465, in options\n       super(SSLContext, SSLContext).options.__set__(self, value)\n     [Previous line repeated 330 more times]\n   RecursionError: maximum recursion depth exceeded while calling a Python object\n\nChange-Id: I4bb59edd87336597791416c4f2a096efe0e72fe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c465101dc46fa86b0094ea51fbece4a39b714f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c465101dc46fa86b0094ea51fbece4a39b714f6"}]},"branch":"refs/heads/master"},"74db3670607d952e597011eb07676aedff521b41":{"kind":"REWORK","_number":4,"created":"2019-08-17 03:34:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/27/675227/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/675227/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/675227/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/675227/4"}}},"commit":{"parents":[{"commit":"9b4b57a8801d86e656b754930e74ac4087c1de8a","subject":"Merge \"py3: port SLO func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b4b57a8801d86e656b754930e74ac4087c1de8a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-07 23:16:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-17 03:34:16.000000000","tz":-420},"subject":"py3: Finish porting func tests","message":"py3: Finish porting func tests\n\nWe were (indirectly) importing swiftclient (and therefore requests and\nurllib3) before doing our eventlet monkey-patching. This would lead\nboto3 (which digs an SSLContext out of urllib3) to trip RecursionErrors\non py3 similar to\n\n   \u003e\u003e\u003e from ssl import SSLContext, PROTOCOL_SSLv23\n   \u003e\u003e\u003e import eventlet\n   \u003e\u003e\u003e eventlet.monkey_patch(socket\u003dTrue)\n   \u003e\u003e\u003e SSLContext(PROTOCOL_SSLv23).options |\u003d 0\n   Traceback (most recent call last):\n     File \"\u003cstdin\u003e\", line 1, in \u003cmodule\u003e\n     File \"/usr/lib/python3.6/ssl.py\", line 465, in options\n       super(SSLContext, SSLContext).options.__set__(self, value)\n     File \"/usr/lib/python3.6/ssl.py\", line 465, in options\n       super(SSLContext, SSLContext).options.__set__(self, value)\n     File \"/usr/lib/python3.6/ssl.py\", line 465, in options\n       super(SSLContext, SSLContext).options.__set__(self, value)\n     [Previous line repeated 330 more times]\n   RecursionError: maximum recursion depth exceeded while calling a Python object\n\nChange-Id: I4bb59edd87336597791416c4f2a096efe0e72fe3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74db3670607d952e597011eb07676aedff521b41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74db3670607d952e597011eb07676aedff521b41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
