)]}'
{"id":"openstack%2Fswift~675244","triplet_id":"openstack%2Fswift~master~I1d6881eb21478dbfbeae20617b174f80a3afb60a","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I1d6881eb21478dbfbeae20617b174f80a3afb60a","subject":"py3: Fix testUtf8Container (test.functional.tests.TestContainer)","status":"ABANDONED","created":"2019-08-08 04:03:47.000000000","updated":"2019-09-20 18:32:45.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9cdd1f2b5ea1791720ec4b0ae0e4fb6a1826500c","_number":675244,"virtual_id_number":675244,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-09-20 14:18:02.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2019-08-14 03:29:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-19 22:27:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 03:29:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-19 22:27:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"02db6c7b319a4a9cad87a4515aeb9260ee334717","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-08 04:03:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8e4aa8e76d997ad717d5a33cc3c82aecbf3dcca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 05:11:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/44/675244/1/check/openstack-tox-docs/142004d/html/ : SUCCESS in 4m 15s\n- grenade-py3 https://logs.opendev.org/44/675244/1/check/grenade-py3/7b6d785/ : SUCCESS in 1h 05m 11s\n- tempest-integrated-object-storage https://logs.opendev.org/44/675244/1/check/tempest-integrated-object-storage/da238ad/ : SUCCESS in 59m 21s\n- swift-build-image https://logs.opendev.org/44/675244/1/check/swift-build-image/3d1d13d/ : SUCCESS in 10m 32s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/44/675244/1/check/swift-tox-py27/967930b/ : SUCCESS in 13m 24s\n- swift-tox-py36 https://logs.opendev.org/44/675244/1/check/swift-tox-py36/a110b0e/ : SUCCESS in 11m 36s\n- swift-tox-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-py37/7366108/ : SUCCESS in 14m 29s\n- swift-tox-func https://logs.opendev.org/44/675244/1/check/swift-tox-func/b52e2be/ : SUCCESS in 11m 13s\n- swift-tox-func-encryption https://logs.opendev.org/44/675244/1/check/swift-tox-func-encryption/b70f40c/ : SUCCESS in 11m 02s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/44/675244/1/check/swift-tox-func-domain-remap-staticweb/a63b472/ : SUCCESS in 12m 03s\n- swift-tox-func-ec https://logs.opendev.org/44/675244/1/check/swift-tox-func-ec/1a9cd9b/ : SUCCESS in 10m 50s\n- swift-tox-func-s3api https://logs.opendev.org/44/675244/1/check/swift-tox-func-s3api/0597c86/ : SUCCESS in 13m 48s\n- swift-tox-func-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-func-py37/b78b78a/ : SUCCESS in 12m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-func-domain-remap-staticweb-py37/dade1b2/ : SUCCESS in 7m 41s\n- swift-tox-func-ec-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-func-ec-py37/03f3df8/ : SUCCESS in 8m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/44/675244/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/b574c04/ : SUCCESS in 13m 07s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/44/675244/1/check/swift-probetests-centos-7/706e78a/ : FAILURE in 28m 19s\n- swift-dsvm-functional https://logs.opendev.org/44/675244/1/check/swift-dsvm-functional/e61f2ce/ : SUCCESS in 33m 00s\n- swift-dsvm-functional-py3 https://logs.opendev.org/44/675244/1/check/swift-dsvm-functional-py3/6e3df10/ : SUCCESS in 30m 57s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/44/675244/1/check/swift-dsvm-functional-ipv6/60d8646/ : SUCCESS in 30m 01s\n- swift-tox-lower-constraints https://logs.opendev.org/44/675244/1/check/swift-tox-lower-constraints/6f8ec7a/ : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://logs.opendev.org/44/675244/1/check/openstack-tox-pep8/3f87369/ : SUCCESS in 7m 35s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/44/675244/1/check/swift-multinode-rolling-upgrade/1afd97a/ : FAILURE in 24m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4121b58ce8ff3803a306f83f8786581806e9a884","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-08 14:45:07.000000000","message":"Patch Set 1:\n\nrecheck\n\nHuh. I thought I took out the dance in https://github.com/eventlet/eventlet/commit/f0bc79e1d973806866e2eb74db82adfda64ab1e5 ...","accounts_in_message":[],"_revision_number":1},{"id":"fb28510ab80b12471dfd74d46a2bce0cc88cf895","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 20:33:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/44/675244/1/check/openstack-tox-docs/6487f0d/html/ : SUCCESS in 5m 18s\n- grenade-py3 https://logs.opendev.org/44/675244/1/check/grenade-py3/3ee1d14/ : SUCCESS in 1h 02m 45s\n- tempest-integrated-object-storage https://logs.opendev.org/44/675244/1/check/tempest-integrated-object-storage/bb58d09/ : SUCCESS in 53m 48s\n- swift-build-image https://logs.opendev.org/44/675244/1/check/swift-build-image/1dc5077/ : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/44/675244/1/check/swift-tox-py27/1e79640/ : SUCCESS in 13m 37s\n- swift-tox-py36 https://logs.opendev.org/44/675244/1/check/swift-tox-py36/cfacacd/ : SUCCESS in 10m 01s\n- swift-tox-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-py37/f53aec7/ : SUCCESS in 13m 53s\n- swift-tox-func https://logs.opendev.org/44/675244/1/check/swift-tox-func/78d3495/ : SUCCESS in 8m 40s\n- swift-tox-func-encryption https://logs.opendev.org/44/675244/1/check/swift-tox-func-encryption/199c5fe/ : SUCCESS in 9m 37s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/44/675244/1/check/swift-tox-func-domain-remap-staticweb/dcc3b4e/ : SUCCESS in 7m 55s\n- swift-tox-func-ec https://logs.opendev.org/44/675244/1/check/swift-tox-func-ec/a92416e/ : SUCCESS in 10m 29s\n- swift-tox-func-s3api https://logs.opendev.org/44/675244/1/check/swift-tox-func-s3api/249dedf/ : SUCCESS in 11m 34s\n- swift-tox-func-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-func-py37/9676e53/ : SUCCESS in 12m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-func-domain-remap-staticweb-py37/1ef8ecd/ : SUCCESS in 9m 20s\n- swift-tox-func-ec-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-func-ec-py37/700fc99/ : SUCCESS in 8m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/44/675244/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/a3407a8/ : SUCCESS in 16m 21s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/44/675244/1/check/swift-probetests-centos-7/8f2a0f8/ : FAILURE in 37m 58s\n- swift-dsvm-functional https://logs.opendev.org/44/675244/1/check/swift-dsvm-functional/c21ceda/ : SUCCESS in 27m 42s\n- swift-dsvm-functional-py3 https://logs.opendev.org/44/675244/1/check/swift-dsvm-functional-py3/267f19b/ : TIMED_OUT in 48m 26s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/44/675244/1/check/swift-dsvm-functional-ipv6/112e839/ : SUCCESS in 25m 49s\n- swift-tox-lower-constraints https://logs.opendev.org/44/675244/1/check/swift-tox-lower-constraints/9102fb7/ : TIMED_OUT in 41m 15s\n- openstack-tox-pep8 https://logs.opendev.org/44/675244/1/check/openstack-tox-pep8/100ae3a/ : SUCCESS in 6m 48s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/44/675244/1/check/swift-multinode-rolling-upgrade/c515512/ : SUCCESS in 17m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1d58bd5f036a5941523476b5c9be092ecfc1706c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-08 20:45:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"423e64492a61aff906d0b0a880240947862062eb","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-08 22:59:29.000000000","message":"Patch Set 1:\n\n\u003e recheck\n \u003e \n \u003e Huh. I thought I took out the dance in https://github.com/eventlet/eventlet/commit/f0bc79e1d973806866e2eb74db82adfda64ab1e5\n \u003e ...\n\nFedora 30 ships with eventlet-0.24.1, which still has it. I didn\u0027t know that you fixed it, I should file a bug with Fedora to rebuild eventlet with the 0.25.0.","accounts_in_message":[],"_revision_number":1},{"id":"c5618ba2cc2b3208ab31e25a57b4b55212d28608","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 01:06:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/44/675244/1/check/openstack-tox-docs/891ab26/html/ : SUCCESS in 4m 22s\n- grenade-py3 https://logs.opendev.org/44/675244/1/check/grenade-py3/6d84459/ : SUCCESS in 1h 09m 14s\n- tempest-integrated-object-storage https://logs.opendev.org/44/675244/1/check/tempest-integrated-object-storage/0127457/ : SUCCESS in 54m 01s\n- swift-build-image https://logs.opendev.org/44/675244/1/check/swift-build-image/43c8666/ : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/44/675244/1/check/swift-tox-py27/b48a3c3/ : SUCCESS in 10m 56s\n- swift-tox-py36 https://logs.opendev.org/44/675244/1/check/swift-tox-py36/b3b6264/ : SUCCESS in 12m 19s\n- swift-tox-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-py37/b1c7756/ : SUCCESS in 10m 41s\n- swift-tox-func https://logs.opendev.org/44/675244/1/check/swift-tox-func/343a3a4/ : SUCCESS in 9m 48s\n- swift-tox-func-encryption https://logs.opendev.org/44/675244/1/check/swift-tox-func-encryption/04ac68b/ : SUCCESS in 20m 09s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/44/675244/1/check/swift-tox-func-domain-remap-staticweb/f328c90/ : SUCCESS in 14m 14s\n- swift-tox-func-ec https://logs.opendev.org/44/675244/1/check/swift-tox-func-ec/98c982e/ : SUCCESS in 10m 30s\n- swift-tox-func-s3api https://logs.opendev.org/44/675244/1/check/swift-tox-func-s3api/86733c3/ : SUCCESS in 19m 03s\n- swift-tox-func-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-func-py37/4deb09a/ : SUCCESS in 10m 43s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-func-domain-remap-staticweb-py37/1e32a99/ : SUCCESS in 7m 58s\n- swift-tox-func-ec-py37 https://logs.opendev.org/44/675244/1/check/swift-tox-func-ec-py37/b8cc3df/ : SUCCESS in 14m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/44/675244/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/8b413f4/ : SUCCESS in 16m 17s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/44/675244/1/check/swift-probetests-centos-7/7d2789a/ : SUCCESS in 35m 48s\n- swift-dsvm-functional https://logs.opendev.org/44/675244/1/check/swift-dsvm-functional/f5d8c55/ : SUCCESS in 31m 03s\n- swift-dsvm-functional-py3 https://logs.opendev.org/44/675244/1/check/swift-dsvm-functional-py3/a515e33/ : SUCCESS in 26m 28s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/44/675244/1/check/swift-dsvm-functional-ipv6/12be1eb/ : SUCCESS in 27m 27s\n- swift-tox-lower-constraints https://logs.opendev.org/44/675244/1/check/swift-tox-lower-constraints/5e9a217/ : SUCCESS in 11m 53s\n- openstack-tox-pep8 https://logs.opendev.org/44/675244/1/check/openstack-tox-pep8/23de8c8/ : SUCCESS in 7m 12s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/44/675244/1/check/swift-multinode-rolling-upgrade/36632ee/ : SUCCESS in 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"f1362a676a82455796b1fbdb798647e23437414c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-09 18:20:04.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7db61d18e7a7e69a9bbbcfcc72f1b42aa2951f4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 00:32:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/44/675244/2/check/openstack-tox-docs/1cee344/html/ : SUCCESS in 4m 31s\n- grenade-py3 https://logs.opendev.org/44/675244/2/check/grenade-py3/d06dce5/ : SUCCESS in 1h 03m 46s\n- tempest-integrated-object-storage https://logs.opendev.org/44/675244/2/check/tempest-integrated-object-storage/0f6f77c/ : FAILURE in 39m 59s\n- swift-build-image https://logs.opendev.org/44/675244/2/check/swift-build-image/07070c0/ : SUCCESS in 8m 24s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/44/675244/2/check/swift-tox-py27/0f6c4e8/ : SUCCESS in 12m 32s\n- swift-tox-py36 https://logs.opendev.org/44/675244/2/check/swift-tox-py36/10e84f8/ : SUCCESS in 13m 35s\n- swift-tox-py37 https://logs.opendev.org/44/675244/2/check/swift-tox-py37/6b5a15e/ : SUCCESS in 10m 17s\n- swift-tox-func https://logs.opendev.org/44/675244/2/check/swift-tox-func/382373d/ : SUCCESS in 9m 21s\n- swift-tox-func-encryption https://logs.opendev.org/44/675244/2/check/swift-tox-func-encryption/e610d31/ : SUCCESS in 10m 37s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/44/675244/2/check/swift-tox-func-domain-remap-staticweb/6f425fa/ : SUCCESS in 9m 17s\n- swift-tox-func-ec https://logs.opendev.org/44/675244/2/check/swift-tox-func-ec/6891575/ : SUCCESS in 10m 03s\n- swift-tox-func-s3api https://logs.opendev.org/44/675244/2/check/swift-tox-func-s3api/940fa2c/ : SUCCESS in 13m 58s\n- swift-tox-func-py37 https://logs.opendev.org/44/675244/2/check/swift-tox-func-py37/77569fa/ : SUCCESS in 11m 47s\n- swift-tox-func-encryption-py37 https://logs.opendev.org/44/675244/2/check/swift-tox-func-encryption-py37/7b22ed7/ : SUCCESS in 8m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/44/675244/2/check/swift-tox-func-domain-remap-staticweb-py37/4e63ff2/ : SUCCESS in 7m 13s\n- swift-tox-func-ec-py37 https://logs.opendev.org/44/675244/2/check/swift-tox-func-ec-py37/832984c/ : SUCCESS in 8m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/44/675244/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/dc301c7/ : SUCCESS in 13m 33s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/44/675244/2/check/swift-probetests-centos-7/fd7f40b/ : SUCCESS in 32m 42s\n- swift-dsvm-functional https://logs.opendev.org/44/675244/2/check/swift-dsvm-functional/744ad3a/ : SUCCESS in 29m 49s\n- swift-dsvm-functional-py3 https://logs.opendev.org/44/675244/2/check/swift-dsvm-functional-py3/f77b3bc/ : SUCCESS in 29m 18s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/44/675244/2/check/swift-dsvm-functional-ipv6/c728805/ : SUCCESS in 26m 02s\n- swift-tox-lower-constraints https://logs.opendev.org/44/675244/2/check/swift-tox-lower-constraints/ce738dd/ : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://logs.opendev.org/44/675244/2/check/openstack-tox-pep8/236a7de/ : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/44/675244/2/check/swift-multinode-rolling-upgrade/d998862/ : SUCCESS in 17m 00s","accounts_in_message":[],"_revision_number":2},{"id":"571e36b06421af28f757b982c94bceedebc9e786","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-10 00:33:47.000000000","message":"Patch Set 2:\n\nrecheck\n\nNo, this didn\u0027t break volume snapshots...","accounts_in_message":[],"_revision_number":2},{"id":"5be55593eeef4ceb5be63ee68c0e843844261ec2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-10 00:46:20.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"00830450f4eb89d6aa26d0275485859a0a5789a0","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-10 01:54:26.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"eb07554cfc661d6745e16d41b12b256db6c84a01","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-10 01:54:59.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1b77b23d9d18e0cf8d444618ac7d67e71024b353","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 02:56:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://logs.opendev.org/44/675244/3/check/openstack-tox-docs/1ea9e00/html/ : SUCCESS in 4m 00s\n- grenade-py3 https://logs.opendev.org/44/675244/3/check/grenade-py3/b66a30f/ : SUCCESS in 55m 40s\n- tempest-integrated-object-storage https://logs.opendev.org/44/675244/3/check/tempest-integrated-object-storage/78376d5/ : SUCCESS in 49m 57s\n- swift-build-image https://logs.opendev.org/44/675244/3/check/swift-build-image/248c9cd/ : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/44/675244/3/check/swift-tox-py27/889a7a7/ : SUCCESS in 12m 05s\n- swift-tox-py36 https://logs.opendev.org/44/675244/3/check/swift-tox-py36/1c75ed5/ : SUCCESS in 11m 27s\n- swift-tox-py37 https://logs.opendev.org/44/675244/3/check/swift-tox-py37/c6c745a/ : FAILURE in 17m 39s\n- swift-tox-func https://logs.opendev.org/44/675244/3/check/swift-tox-func/381a09f/ : SUCCESS in 8m 28s\n- swift-tox-func-encryption https://logs.opendev.org/44/675244/3/check/swift-tox-func-encryption/d3e0675/ : SUCCESS in 9m 17s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/44/675244/3/check/swift-tox-func-domain-remap-staticweb/8d17447/ : SUCCESS in 9m 52s\n- swift-tox-func-ec https://logs.opendev.org/44/675244/3/check/swift-tox-func-ec/d09d1df/ : SUCCESS in 9m 44s\n- swift-tox-func-s3api https://logs.opendev.org/44/675244/3/check/swift-tox-func-s3api/98e9fe6/ : SUCCESS in 14m 05s\n- swift-tox-func-py37 https://logs.opendev.org/44/675244/3/check/swift-tox-func-py37/dd806ec/ : SUCCESS in 11m 32s\n- swift-tox-func-encryption-py37 https://logs.opendev.org/44/675244/3/check/swift-tox-func-encryption-py37/6019499/ : SUCCESS in 14m 44s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/44/675244/3/check/swift-tox-func-domain-remap-staticweb-py37/8ec2a89/ : FAILURE in 9m 34s\n- swift-tox-func-ec-py37 https://logs.opendev.org/44/675244/3/check/swift-tox-func-ec-py37/d3b88b7/ : SUCCESS in 10m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/44/675244/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/aad372a/ : SUCCESS in 15m 44s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/44/675244/3/check/swift-probetests-centos-7/6b02bf1/ : FAILURE in 31m 55s\n- swift-dsvm-functional https://logs.opendev.org/44/675244/3/check/swift-dsvm-functional/bdeeafa/ : SUCCESS in 26m 08s\n- swift-dsvm-functional-py3 https://logs.opendev.org/44/675244/3/check/swift-dsvm-functional-py3/79315c4/ : SUCCESS in 32m 39s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/44/675244/3/check/swift-dsvm-functional-ipv6/0b3f37e/ : SUCCESS in 26m 02s\n- swift-tox-lower-constraints https://logs.opendev.org/44/675244/3/check/swift-tox-lower-constraints/b4c293d/ : SUCCESS in 13m 21s\n- openstack-tox-pep8 https://logs.opendev.org/44/675244/3/check/openstack-tox-pep8/1279844/ : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/44/675244/3/check/swift-multinode-rolling-upgrade/8511e01/ : SUCCESS in 16m 32s","accounts_in_message":[],"_revision_number":3},{"id":"7c3b2a5b3110241eabb78cc882b41fae4b99b8ad","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-14 02:23:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"64796cede2d36ae980eea4db7ff4598cc4647aa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 03:29:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985e75ec2da541d29aeb35c3a823ef06 : SUCCESS in 4m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b4d31267e1504f23817db92d98ab991e : SUCCESS in 1h 04m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/826586c8a74a40ac866aeb003501dc9b : SUCCESS in 48m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a03be9bcbbe9402dbfb76879c22dda4a : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9368a0c1fe664d008257c68a70c87704 : SUCCESS in 18m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c2fe2289474f1cadb186fb54c77319 : SUCCESS in 9m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ed68786bae02404dbc5fb23ed1a1b8ef : SUCCESS in 11m 30s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/e32d970f468545519e1d8be6307cd906 : SUCCESS in 8m 51s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/ffd601cf0517493fb1d4dc9acd99b328 : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/ef699fde140b436c986625ee0aae598c : SUCCESS in 8m 48s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/08efb22bdba24f909066c7113b6d3010 : SUCCESS in 12m 09s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/d58600259ce541dd9edb8d91edbcdb64 : SUCCESS in 11m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/107bd93d742841a9ab742c41abbb552a : SUCCESS in 9m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3bb4bc55e0cf4c0391efbe5953ff584c : SUCCESS in 8m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/cecd529b65f34193b6ac3ddd008a88cb : SUCCESS in 8m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2a04c9f51c6a46f4ad1898d0ec0cf461 : SUCCESS in 9m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13c4d73b9d744fbb8aa0c0a93f03a15c : SUCCESS in 15m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/17f3126aef5241c6a0ca8b76b4ee1631 : SUCCESS in 30m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3eaf5d60951f4bb4a7adeb87191377bb : SUCCESS in 25m 13s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0cba8cddee0e4c6facee29f06b7543a6 : SUCCESS in 28m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aede2b95f9274c7a9dfa537ba19a932c : SUCCESS in 27m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1acef5f098542e7a7e1fda59afcc32e : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12bd2e4ad99b4ef9965f4d3f2785f45a : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8804a09133d4094ad737ae25875a6d1 : SUCCESS in 18m 57s","accounts_in_message":[],"_revision_number":3},{"id":"c268d80bf4a321ae85d4a069299d733ba43c2302","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-19 22:27:25.000000000","message":"Patch Set 3: Code-Review+2\n\nSo what do we want to do with this? I\u0027m not really *opposed* to it... I know I\u0027ve added more than my fair share of code-scars as part of this transition... what\u0027s one more, right?\n\nBut I *am* worried about RH\u0027s customers trying to run ssync (at least) without https://github.com/eventlet/eventlet/commit/53001f5e9 in their eventlet...\n\nFeel free to approve if this really would make your life easier.","accounts_in_message":[],"_revision_number":3},{"id":"6f264875199868f91801cfcb688d6baab043c164","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-09-20 14:18:02.000000000","message":"Patch Set 3:\n\n\u003e But I *am* worried about RH\u0027s customers trying to run ssync (at\n \u003e least) without https://github.com/eventlet/eventlet/commit/53001f5e9\n \u003e in their eventlet...\n\nIs this part of an eventlet release? If so, we ought to update dependencies to it or 0.25, whichever is greater.","accounts_in_message":[],"_revision_number":3},{"id":"b44a597f7b3d67f07a0a7be23dd160822bb2a344","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-09-20 14:37:53.000000000","message":"Abandoned\n\nI filed a review in RDO for a better eventlet:\nhttps://review.rdoproject.org/r/22457","accounts_in_message":[],"_revision_number":3},{"id":"9cdd1f2b5ea1791720ec4b0ae0e4fb6a1826500c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-20 18:32:45.000000000","message":"Patch Set 3:\n\n\u003e Is this part of an eventlet release? If so, we ought to update dependencies to it or 0.25, whichever is greater.\n\nYes, it was first released as part of 0.25.0","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1beccac53a767e5bdc8561255d0381ebd1340683","revisions":{"14a2ce720cfaeff29579ed68c6773d85e3ae52dd":{"kind":"REWORK","_number":1,"created":"2019-08-08 04:03:47.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/44/675244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/675244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/675244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/675244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/675244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/675244/1"}}},"commit":{"parents":[{"commit":"8d3d21e66846979d3bb4fbac136251da88f22310","subject":"Merge \"py3: port RBAC func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d3d21e66846979d3bb4fbac136251da88f22310"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-08-08 03:57:17.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-08-08 04:03:47.000000000","tz":-300},"subject":"py3: Fix testUtf8Container (test.functional.tests.TestContainer)","message":"py3: Fix testUtf8Container (test.functional.tests.TestContainer)\n\nThis pops up when running functional tests on an old\nsystem that uses py2 against a new cluster running on py3.\nThe test tries to do a PUT with a UTF-8 encoded path,\nwhich succeeds, and then a garbage binary path, which...\nalso successds. But in case of containers, it\u0027s should\nthrow a 412. On py3 the eventlet quetly substitutes all\nsequences not valid in UTF-8 with replacement characters,\nso we receive a WSGI string that decodes into a valid UTF-8,\nand then we go on to create a container with wrong name.\n\nOur fix for now is to undo the damage, as part of general\nfix-ups that we do for eventlent already.\n\nChange-Id: I1d6881eb21478dbfbeae20617b174f80a3afb60a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14a2ce720cfaeff29579ed68c6773d85e3ae52dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14a2ce720cfaeff29579ed68c6773d85e3ae52dd"}]},"branch":"refs/heads/master"},"b37bc0364e553bd32e1d80861e691e57cf210306":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-09 18:20:04.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/44/675244/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/675244/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/675244/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/675244/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/675244/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/675244/2"}}},"commit":{"parents":[{"commit":"8d3d21e66846979d3bb4fbac136251da88f22310","subject":"Merge \"py3: port RBAC func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d3d21e66846979d3bb4fbac136251da88f22310"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-08-08 03:57:17.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-08-09 18:19:36.000000000","tz":0},"subject":"py3: Fix testUtf8Container (test.functional.tests.TestContainer)","message":"py3: Fix testUtf8Container (test.functional.tests.TestContainer)\n\nThis pops up when running functional tests on an old\nsystem that uses py2 against a new cluster running on py3.\nThe test tries to do a PUT with a UTF-8 encoded path,\nwhich succeeds, and then a garbage binary path, which...\nalso successds. But in case of containers, it\u0027s should\nthrow a 412. On py3 the eventlet quetly substitutes all\nsequences not valid in UTF-8 with replacement characters,\nso we receive a WSGI string that decodes into a valid UTF-8,\nand then we go on to create a container with wrong name.\n\nOur fix for now is to undo the damage, as part of general\nfix-ups that we do for eventlent already.\n\nN.B. This problem is fixed in eventlet 2.5.0. So, this\nis a temporary workaround until about 2021, when RHEL 8.3\ncomes around. The fix is designed to be harmless if we\nforget to remove it.\n\nChange-Id: I1d6881eb21478dbfbeae20617b174f80a3afb60a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b37bc0364e553bd32e1d80861e691e57cf210306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b37bc0364e553bd32e1d80861e691e57cf210306"}]},"branch":"refs/heads/master"},"1beccac53a767e5bdc8561255d0381ebd1340683":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-10 01:54:59.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/44/675244/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/675244/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/675244/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/675244/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/675244/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/675244/3"}}},"commit":{"parents":[{"commit":"8d3d21e66846979d3bb4fbac136251da88f22310","subject":"Merge \"py3: port RBAC func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d3d21e66846979d3bb4fbac136251da88f22310"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-08-08 03:57:17.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-08-10 01:54:56.000000000","tz":0},"subject":"py3: Fix testUtf8Container (test.functional.tests.TestContainer)","message":"py3: Fix testUtf8Container (test.functional.tests.TestContainer)\n\nThis pops up when running functional tests on an old\nsystem that uses py2 against a new cluster running on py3.\nThe test tries to do a PUT with a UTF-8 encoded path,\nwhich succeeds, and then a garbage binary path, which...\nalso successds. But in case of containers, it\u0027s should\nthrow a 412. On py3 the eventlet quetly substitutes all\nsequences not valid in UTF-8 with replacement characters,\nso we receive a WSGI string that decodes into a valid UTF-8,\nand then we go on to create a container with wrong name.\n\nOur fix for now is to undo the damage, as part of general\nfix-ups that we do for eventlent already.\n\nN.B. This problem is fixed in eventlet 0.25.0. So, this\nis a temporary workaround until about 2021, when RHEL 8.3\ncomes around. The fix is designed to be harmless if we\nforget to remove it.\n\nChange-Id: I1d6881eb21478dbfbeae20617b174f80a3afb60a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1beccac53a767e5bdc8561255d0381ebd1340683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1beccac53a767e5bdc8561255d0381ebd1340683"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
