)]}'
{"id":"openstack%2Fswift~675435","triplet_id":"openstack%2Fswift~stable%2Fstein~I35eb16f568c00d973eb089aedd0e5a5619c25a2f","project":"openstack/swift","branch":"stable/stein","hashtags":[],"change_id":"I35eb16f568c00d973eb089aedd0e5a5619c25a2f","subject":"Fix quarantine when object path is not a directory","status":"MERGED","created":"2019-08-08 18:23:10.000000000","updated":"2019-08-14 22:17:11.000000000","submitted":"2019-08-14 22:17:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675435-1565821031116-8c476a95","meta_rev_id":"6aead38c886bc533d5b537dcc82a5460bf5a73de","_number":675435,"virtual_id_number":675435,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":2,"date":"2019-08-14 22:17:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-14 20:30:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"date":"2019-08-08 18:23:10.000000000","_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-14 20:30:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 18:23:10.000000000","updated_by":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"reviewer":{"_account_id":25251,"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","username":"alecuyer"},"state":"REVIEWER"},{"updated":"2019-08-14 22:17:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aac54cc21378e08ca452514dc7812e7276438626","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-08 18:23:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a22a4a36523530695388961234e969a9db3aeac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 23:06:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/35/675435/1/check/openstack-tox-docs/e0873c1/html/ : SUCCESS in 4m 33s\n- swift-tox-py27 https://logs.opendev.org/35/675435/1/check/swift-tox-py27/4e1336c/ : SUCCESS in 10m 37s\n- swift-tox-py35 https://logs.opendev.org/35/675435/1/check/swift-tox-py35/08f5a43/ : SUCCESS in 10m 45s\n- swift-tox-py36 https://logs.opendev.org/35/675435/1/check/swift-tox-py36/deaa98d/ : SUCCESS in 8m 40s (non-voting)\n- swift-tox-py37 https://logs.opendev.org/35/675435/1/check/swift-tox-py37/0777706/ : SUCCESS in 10m 41s (non-voting)\n- swift-tox-func https://logs.opendev.org/35/675435/1/check/swift-tox-func/e0129a2/ : SUCCESS in 10m 19s\n- swift-tox-func-encryption https://logs.opendev.org/35/675435/1/check/swift-tox-func-encryption/386c49f/ : SUCCESS in 9m 55s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/35/675435/1/check/swift-tox-func-domain-remap-staticweb/3c35cf7/ : SUCCESS in 9m 47s\n- swift-tox-func-ec https://logs.opendev.org/35/675435/1/check/swift-tox-func-ec/0cd9782/ : SUCCESS in 9m 45s\n- swift-tox-func-s3api https://logs.opendev.org/35/675435/1/check/swift-tox-func-s3api/afddb7f/ : SUCCESS in 13m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/35/675435/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/72306df/ : SUCCESS in 14m 21s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/35/675435/1/check/swift-probetests-centos-7/fced4c7/ : SUCCESS in 30m 10s\n- swift-dsvm-functional https://logs.opendev.org/35/675435/1/check/swift-dsvm-functional/2f7ddae/ : SUCCESS in 28m 39s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/35/675435/1/check/swift-dsvm-functional-ipv6/b55eb5d/ : SUCCESS in 36m 02s\n- swift-tox-lower-constraints https://logs.opendev.org/35/675435/1/check/swift-tox-lower-constraints/2250a45/ : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://logs.opendev.org/35/675435/1/check/openstack-tox-pep8/77f2aee/ : SUCCESS in 8m 56s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/35/675435/1/check/swift-multinode-rolling-upgrade/6a3ca19/ : SUCCESS in 15m 42s","accounts_in_message":[],"_revision_number":1},{"id":"122cc46760535576e6e293e28de5086789fe3347","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-14 20:30:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8060c23f9b6a96463eb7fcfca0f1ff939b73492","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 20:30:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6dd04e145c566f9d5e31955cd3fba2cd132f53d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 22:17:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8755a19070134bfdb55e2ff8c94f9a2a : SUCCESS in 5m 35s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3eb6c6928ee045c2a4a6b3801ff2224d : SUCCESS in 18m 10s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/a1f7a70c224743d98313f2895e54ae7e : SUCCESS in 9m 15s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/2ca6cb72b2a9491bb4041ee329a77aa6 : SUCCESS in 8m 31s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/71066f18f5cf4317af70dade3e5db48a : SUCCESS in 11m 55s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/60134921997d4abd8e1a1e2f28fb5f1f : SUCCESS in 8m 33s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/53d3a5f6a4c54373a16cb3cf5319bfa4 : SUCCESS in 9m 20s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/8323f519da6e479fbb1d72804e279c39 : SUCCESS in 16m 41s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cafc04ccafe446b3b13a6198c894a44b : SUCCESS in 40m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b79693675581437c8b2c627a5005c11c : SUCCESS in 27m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61fcabce53244a5aa38b4c261da1e0cc : SUCCESS in 24m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd9f73f78380410ca55b297f7436c7ec : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a369abfa82c46d18873b40ebf5d7bc7 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6aead38c886bc533d5b537dcc82a5460bf5a73de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 22:17:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"43046cdd4325c88bc17fe0af1b317fbbff55787d","revisions":{"43046cdd4325c88bc17fe0af1b317fbbff55787d":{"kind":"REWORK","_number":1,"created":"2019-08-08 18:23:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/675435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/675435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/675435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/675435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/675435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/675435/1"}}},"commit":{"parents":[{"commit":"0dabd1883decb574dcf30c49b2cd2c196d12564e","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dabd1883decb574dcf30c49b2cd2c196d12564e"}]}],"author":{"name":"Alexandre Lécuyer","email":"alexandre.lecuyer@corp.ovh.com","date":"2019-04-18 14:35:07.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-08 18:22:57.000000000","tz":-420},"subject":"Fix quarantine when object path is not a directory","message":"Fix quarantine when object path is not a directory\n\nIn the diskfile module, there are a couple of cases where we would\nquarantine the suffix dir if a single object within is not a directory.\n\nChange the code so that we quarantine only the object.\n\nChange-Id: I35eb16f568c00d973eb089aedd0e5a5619c25a2f\n(cherry picked from commit 6008b0a83f862b2396ad6642b7863a30a258520b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43046cdd4325c88bc17fe0af1b317fbbff55787d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43046cdd4325c88bc17fe0af1b317fbbff55787d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
