)]}'
{"id":"openstack%2Fswift~675696","triplet_id":"openstack%2Fswift~master~Icb0d3e3708949127e9750f2022de8c74ce51ebe9","project":"openstack/swift","branch":"master","topic":"py3/func-keepalive","hashtags":[],"change_id":"Icb0d3e3708949127e9750f2022de8c74ce51ebe9","subject":"Allow Connection: value to be case-insensitive","status":"MERGED","created":"2019-08-09 20:07:07.000000000","updated":"2019-08-10 03:58:38.000000000","submitted":"2019-08-10 03:57:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675696-1565409454281-53e3b4f1","meta_rev_id":"53c9d04357cc6b8e69f050cb1b8de2b7e181b722","_number":675696,"virtual_id_number":675696,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-10 03:57:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-09 20:50:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-09 20:50:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 20:50:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-08-10 03:57:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e60689dbd5ef8d76017b6426919046689d6b191","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-09 20:07:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f60b047466b59776231ab383d90618527c7ab84b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-09 20:50:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSpec even calls it out as being case-insensitive: https://tools.ietf.org/html/rfc7230#section-6.1\n\nLGTM!","accounts_in_message":[],"_revision_number":1},{"id":"07475d9a2388e6da68056081f9d2ac985f1cac0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 00:25:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/96/675696/1/check/openstack-tox-docs/8103195/html/ : SUCCESS in 3m 49s\n- swift-tox-func https://logs.opendev.org/96/675696/1/check/swift-tox-func/f4acfd5/ : SUCCESS in 8m 23s\n- swift-tox-func-encryption https://logs.opendev.org/96/675696/1/check/swift-tox-func-encryption/3263414/ : SUCCESS in 9m 39s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/96/675696/1/check/swift-tox-func-domain-remap-staticweb/0a79ff6/ : SUCCESS in 10m 22s\n- swift-tox-func-ec https://logs.opendev.org/96/675696/1/check/swift-tox-func-ec/7240c86/ : SUCCESS in 11m 12s\n- swift-tox-func-s3api https://logs.opendev.org/96/675696/1/check/swift-tox-func-s3api/54f62e7/ : SUCCESS in 12m 25s\n- swift-tox-func-py37 https://logs.opendev.org/96/675696/1/check/swift-tox-func-py37/e755f88/ : SUCCESS in 10m 23s\n- swift-tox-func-encryption-py37 https://logs.opendev.org/96/675696/1/check/swift-tox-func-encryption-py37/374efb9/ : SUCCESS in 9m 57s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/96/675696/1/check/swift-tox-func-domain-remap-staticweb-py37/1e3ca6d/ : SUCCESS in 8m 53s\n- swift-tox-func-ec-py37 https://logs.opendev.org/96/675696/1/check/swift-tox-func-ec-py37/9bd7f85/ : SUCCESS in 9m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/96/675696/1/check/swift-tox-func-s3api-ceph-s3tests-tempauth/8cd1cfd/ : SUCCESS in 12m 16s (non-voting)\n- swift-dsvm-functional https://logs.opendev.org/96/675696/1/check/swift-dsvm-functional/26fb88c/ : SUCCESS in 29m 24s\n- swift-dsvm-functional-py3 https://logs.opendev.org/96/675696/1/check/swift-dsvm-functional-py3/aa39444/ : SUCCESS in 27m 45s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/96/675696/1/check/swift-dsvm-functional-ipv6/c291478/ : SUCCESS in 25m 51s\n- openstack-tox-pep8 https://logs.opendev.org/96/675696/1/check/openstack-tox-pep8/b7d6826/ : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/96/675696/1/check/swift-multinode-rolling-upgrade/6c37ee1/ : SUCCESS in 18m 52s","accounts_in_message":[],"_revision_number":1},{"id":"149e75e68fd3f573dd874049c5a7bf18fc2ff54d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 00:26:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f9295263c1e881a12efff9068a505189032804e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 03:57:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/96/675696/1/gate/openstack-tox-docs/89fb114/html/ : SUCCESS in 4m 34s\n- swift-upload-image https://logs.opendev.org/96/675696/1/gate/swift-upload-image/bdfb2c6/ : SUCCESS in 10m 58s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/96/675696/1/gate/swift-tox-py27/df22bda/ : SUCCESS in 14m 41s\n- swift-tox-py36 https://logs.opendev.org/96/675696/1/gate/swift-tox-py36/d692f30/ : SUCCESS in 10m 46s\n- swift-tox-py37 https://logs.opendev.org/96/675696/1/gate/swift-tox-py37/b696d8f/ : SUCCESS in 12m 06s\n- swift-tox-func https://logs.opendev.org/96/675696/1/gate/swift-tox-func/d4aabb1/ : SUCCESS in 10m 16s\n- swift-tox-func-encryption https://logs.opendev.org/96/675696/1/gate/swift-tox-func-encryption/8b9df41/ : SUCCESS in 11m 43s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/96/675696/1/gate/swift-tox-func-domain-remap-staticweb/1d58552/ : SUCCESS in 10m 17s\n- swift-tox-func-ec https://logs.opendev.org/96/675696/1/gate/swift-tox-func-ec/690ec80/ : SUCCESS in 17m 08s\n- swift-tox-func-s3api https://logs.opendev.org/96/675696/1/gate/swift-tox-func-s3api/ab6de4c/ : SUCCESS in 12m 42s\n- swift-tox-func-py37 https://logs.opendev.org/96/675696/1/gate/swift-tox-func-py37/2233925/ : SUCCESS in 17m 46s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/96/675696/1/gate/swift-tox-func-domain-remap-staticweb-py37/a71b5e4/ : SUCCESS in 12m 11s\n- swift-tox-func-ec-py37 https://logs.opendev.org/96/675696/1/gate/swift-tox-func-ec-py37/fc52c9c/ : SUCCESS in 10m 04s\n- swift-dsvm-functional https://logs.opendev.org/96/675696/1/gate/swift-dsvm-functional/8b5c6b5/ : SUCCESS in 26m 51s\n- swift-dsvm-functional-py3 https://logs.opendev.org/96/675696/1/gate/swift-dsvm-functional-py3/7cfb7ab/ : SUCCESS in 37m 03s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/96/675696/1/gate/swift-dsvm-functional-ipv6/74f6ad5/ : SUCCESS in 26m 51s\n- openstack-tox-pep8 https://logs.opendev.org/96/675696/1/gate/openstack-tox-pep8/afb1955/ : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"51aa5a04b3cabdd37f46fd655524b93b82eeaa99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 03:57:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7ff830329756f30bb237388b97351e6d595d88e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 03:58:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image https://logs.opendev.org/96/675696/1/promote/swift-promote-image/05e0d4f/ : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c90365f98d64a38aa6a76a7efcd803470be384bd","revisions":{"c90365f98d64a38aa6a76a7efcd803470be384bd":{"kind":"REWORK","_number":1,"created":"2019-08-09 20:07:07.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/96/675696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/675696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/675696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/675696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/675696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/675696/1"}}},"commit":{"parents":[{"commit":"8d3d21e66846979d3bb4fbac136251da88f22310","subject":"Merge \"py3: port RBAC func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d3d21e66846979d3bb4fbac136251da88f22310"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-08-09 20:06:47.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-08-09 20:06:47.000000000","tz":-300},"subject":"Allow Connection: value to be case-insensitive","message":"Allow Connection: value to be case-insensitive\n\nProxies and LBs parse the Connection: header and re-insert\nthe value with random casing. This occurs when we run functional\ntests against realistic clusters.\n\nChange-Id: Icb0d3e3708949127e9750f2022de8c74ce51ebe9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c90365f98d64a38aa6a76a7efcd803470be384bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c90365f98d64a38aa6a76a7efcd803470be384bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
