)]}'
{"id":"openstack%2Fswift~675710","triplet_id":"openstack%2Fswift~master~I8c2517599a6d1cdf68050dc7acb039f09875fd53","project":"openstack/swift","branch":"master","topic":"py3-func-tests","hashtags":[],"change_id":"I8c2517599a6d1cdf68050dc7acb039f09875fd53","subject":"py3: port test/functional/test_versioned_writes.py","status":"MERGED","created":"2019-08-09 23:00:13.000000000","updated":"2019-08-14 22:21:14.000000000","submitted":"2019-08-14 22:17:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675710-1565821022976-d21f4b70","meta_rev_id":"172461470efc1ef2971513748f6f5e1243eb108e","_number":675710,"virtual_id_number":675710,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":2,"date":"2019-08-14 22:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-08-10 03:05:27.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":1,"date":"2019-08-13 16:01:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-14 20:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-10 03:05:27.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-08-13 16:01:54.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-08-14 22:17:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e88807b9023b9d0bd02915d91f99716d5685bb2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-09 23:00:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"565a6fe8a6263dca0592787c8ccda36171558792","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-09 23:01:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3dc932415d85be564e7520c336e2ed4b65b01dd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 02:38:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://logs.opendev.org/10/675710/2/check/build-openstack-api-ref/11cd587/html/ : SUCCESS in 4m 55s\n- openstack-tox-docs https://logs.opendev.org/10/675710/2/check/openstack-tox-docs/503a367/html/ : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://logs.opendev.org/10/675710/2/check/build-openstack-releasenotes/52c78fa/html/ : SUCCESS in 7m 38s\n- grenade-py3 https://logs.opendev.org/10/675710/2/check/grenade-py3/1bca2b0/ : SUCCESS in 1h 34m 36s\n- tempest-integrated-object-storage https://logs.opendev.org/10/675710/2/check/tempest-integrated-object-storage/087d022/ : SUCCESS in 46m 53s\n- swift-build-image https://logs.opendev.org/10/675710/2/check/swift-build-image/c33caef/ : SUCCESS in 8m 43s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/10/675710/2/check/swift-tox-py27/ad8a419/ : SUCCESS in 11m 57s\n- swift-tox-py36 https://logs.opendev.org/10/675710/2/check/swift-tox-py36/2b93bab/ : SUCCESS in 16m 58s\n- swift-tox-py37 https://logs.opendev.org/10/675710/2/check/swift-tox-py37/3dcff3c/ : SUCCESS in 10m 58s\n- swift-tox-func https://logs.opendev.org/10/675710/2/check/swift-tox-func/519a71f/ : SUCCESS in 9m 19s\n- swift-tox-func-encryption https://logs.opendev.org/10/675710/2/check/swift-tox-func-encryption/02bb5f3/ : SUCCESS in 11m 29s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/10/675710/2/check/swift-tox-func-domain-remap-staticweb/9b0c5d0/ : SUCCESS in 9m 09s\n- swift-tox-func-ec https://logs.opendev.org/10/675710/2/check/swift-tox-func-ec/6dedb1c/ : SUCCESS in 9m 10s\n- swift-tox-func-s3api https://logs.opendev.org/10/675710/2/check/swift-tox-func-s3api/9add3ae/ : SUCCESS in 12m 18s\n- swift-tox-func-py37 https://logs.opendev.org/10/675710/2/check/swift-tox-func-py37/714a64b/ : SUCCESS in 9m 46s\n- swift-tox-func-encryption-py37 https://logs.opendev.org/10/675710/2/check/swift-tox-func-encryption-py37/e92f924/ : SUCCESS in 9m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/10/675710/2/check/swift-tox-func-domain-remap-staticweb-py37/e4a7289/ : SUCCESS in 9m 05s\n- swift-tox-func-ec-py37 https://logs.opendev.org/10/675710/2/check/swift-tox-func-ec-py37/38c19d0/ : SUCCESS in 9m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/10/675710/2/check/swift-tox-func-s3api-ceph-s3tests-tempauth/fcb4286/ : SUCCESS in 23m 45s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/10/675710/2/check/swift-probetests-centos-7/14f8a03/ : FAILURE in 44m 29s\n- swift-dsvm-functional https://logs.opendev.org/10/675710/2/check/swift-dsvm-functional/d0ef154/ : SUCCESS in 27m 35s\n- swift-dsvm-functional-py3 https://logs.opendev.org/10/675710/2/check/swift-dsvm-functional-py3/a4d30bf/ : SUCCESS in 36m 48s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/10/675710/2/check/swift-dsvm-functional-ipv6/2d88a74/ : SUCCESS in 42m 03s\n- swift-tox-lower-constraints https://logs.opendev.org/10/675710/2/check/swift-tox-lower-constraints/ce2ff9a/ : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://logs.opendev.org/10/675710/2/check/openstack-tox-pep8/7b1a35c/ : FAILURE in 5m 48s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/10/675710/2/check/swift-multinode-rolling-upgrade/e8cace8/ : SUCCESS in 16m 40s","accounts_in_message":[],"_revision_number":2},{"id":"12ec5996aa314035abd23d766096fa17d4b2eb8c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-10 03:05:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2bf75bb7935e2b6a7e9070634fb5413b96039e8c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-10 03:05:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c90c140628091d213aaee55349e47806ec3c3ad6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 04:11:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://logs.opendev.org/10/675710/3/check/build-openstack-api-ref/e3eebe8/html/ : SUCCESS in 3m 02s\n- openstack-tox-docs https://logs.opendev.org/10/675710/3/check/openstack-tox-docs/aad0a5e/html/ : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://logs.opendev.org/10/675710/3/check/build-openstack-releasenotes/4bbe2f3/html/ : SUCCESS in 7m 32s\n- grenade-py3 https://logs.opendev.org/10/675710/3/check/grenade-py3/b572089/ : SUCCESS in 1h 00m 00s\n- tempest-integrated-object-storage https://logs.opendev.org/10/675710/3/check/tempest-integrated-object-storage/c886b66/ : SUCCESS in 48m 38s\n- swift-build-image https://logs.opendev.org/10/675710/3/check/swift-build-image/86ec8f5/ : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py27 https://logs.opendev.org/10/675710/3/check/swift-tox-py27/21a4a2b/ : SUCCESS in 9m 57s\n- swift-tox-py36 https://logs.opendev.org/10/675710/3/check/swift-tox-py36/4ee6554/ : SUCCESS in 9m 47s\n- swift-tox-py37 https://logs.opendev.org/10/675710/3/check/swift-tox-py37/9d28e59/ : SUCCESS in 9m 00s\n- swift-tox-func https://logs.opendev.org/10/675710/3/check/swift-tox-func/5559932/ : SUCCESS in 8m 27s\n- swift-tox-func-encryption https://logs.opendev.org/10/675710/3/check/swift-tox-func-encryption/6d6786f/ : SUCCESS in 8m 00s\n- swift-tox-func-domain-remap-staticweb https://logs.opendev.org/10/675710/3/check/swift-tox-func-domain-remap-staticweb/8bc8ad7/ : SUCCESS in 8m 09s\n- swift-tox-func-ec https://logs.opendev.org/10/675710/3/check/swift-tox-func-ec/9097d93/ : SUCCESS in 9m 06s\n- swift-tox-func-s3api https://logs.opendev.org/10/675710/3/check/swift-tox-func-s3api/542be81/ : SUCCESS in 11m 02s\n- swift-tox-func-py37 https://logs.opendev.org/10/675710/3/check/swift-tox-func-py37/c1267e3/ : SUCCESS in 11m 08s\n- swift-tox-func-encryption-py37 https://logs.opendev.org/10/675710/3/check/swift-tox-func-encryption-py37/9d7bb85/ : SUCCESS in 10m 08s\n- swift-tox-func-domain-remap-staticweb-py37 https://logs.opendev.org/10/675710/3/check/swift-tox-func-domain-remap-staticweb-py37/f887a84/ : SUCCESS in 7m 15s\n- swift-tox-func-ec-py37 https://logs.opendev.org/10/675710/3/check/swift-tox-func-ec-py37/bce45f7/ : SUCCESS in 9m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://logs.opendev.org/10/675710/3/check/swift-tox-func-s3api-ceph-s3tests-tempauth/0b9df58/ : SUCCESS in 10m 59s (non-voting)\n- swift-probetests-centos-7 https://logs.opendev.org/10/675710/3/check/swift-probetests-centos-7/c61e660/ : SUCCESS in 29m 36s\n- swift-dsvm-functional https://logs.opendev.org/10/675710/3/check/swift-dsvm-functional/9444ef5/ : SUCCESS in 25m 18s\n- swift-dsvm-functional-py3 https://logs.opendev.org/10/675710/3/check/swift-dsvm-functional-py3/7ab64b4/ : SUCCESS in 28m 57s\n- swift-dsvm-functional-ipv6 https://logs.opendev.org/10/675710/3/check/swift-dsvm-functional-ipv6/a40cf1d/ : SUCCESS in 27m 20s\n- swift-tox-lower-constraints https://logs.opendev.org/10/675710/3/check/swift-tox-lower-constraints/7f4ad33/ : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://logs.opendev.org/10/675710/3/check/openstack-tox-pep8/bee2753/ : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://logs.opendev.org/10/675710/3/check/swift-multinode-rolling-upgrade/ec11907/ : SUCCESS in 16m 21s","accounts_in_message":[],"_revision_number":3},{"id":"0e37bf5d8b1caa7ecfb64f2fd961a664b70f274a","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-13 16:01:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fd97ed3105a5e8a875e89e6f7df45b326a6a15ad","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-14 20:24:39.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"29a5eb452711ad5c3446d3be2c2a30fea77d5a16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 20:25:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"925da29d9bd3ac5e1ec82a33a7f8137f562fc0ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 22:17:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/42ecaf0ac36a481fba172ff6bf2740fd : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409f87ad7e7e45f78609f483c8117bd4 : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fa25b22286444768509369182111fa0 : SUCCESS in 9m 09s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ee73d719049f4de9a88eeca5ebe806c1 : SUCCESS in 1h 02m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a6d3a6b787234adb8186c3402d37428e : SUCCESS in 56m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fec6de8785e14596943d537f70726e77 : SUCCESS in 8m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dd3e3b105fbe4632b5c948622572ed23 : SUCCESS in 12m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8e81202055bf4c449df69e4e33ea8b7a : SUCCESS in 13m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/15083ee39c5d48c39ed27615218c2475 : SUCCESS in 13m 07s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/294374a6aed74a268669309541c2077e : SUCCESS in 12m 01s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/3d446d66e6a8411ea0dfba7603defc4d : SUCCESS in 13m 03s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/a7b12371dd3642029c65b97aa762a124 : SUCCESS in 13m 03s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/414e4336a8dd470fa09fa412c227914f : SUCCESS in 12m 50s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/f380bc06efd74dff880c31a256804dc2 : SUCCESS in 17m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/34169a874e3548a59c090a0befe4d713 : SUCCESS in 14m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/82da8b08d4d749a5a4256ae6fed99d42 : SUCCESS in 9m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1144be7cd180445eb09a310841524dbf : SUCCESS in 8m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d7879ea5fb1d45a8a861dbd31ab71058 : SUCCESS in 30m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a7d1c53270444c40a0788e4a02d50542 : SUCCESS in 39m 00s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4f6bc81336624e109ca7de123d5a6f99 : SUCCESS in 42m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/956239ca6b9d4b5c9ea70b6490ef6f2a : SUCCESS in 36m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f92333253e554c08bf74c0df5c698916 : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1fc1cdebf6b492e8f01695cb1389e61 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"910c61f726a8b94f8c059a0b25c1363d38df9aba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 22:17:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7fe6e78653fe880450b4db057d052f366e48d46d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 22:21:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6f7d566bdac04108a3c609c9edcd6409 : SUCCESS in 1m 03s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"227fe30251984feb45b761390a4ec1d8f5cc894d","revisions":{"2a0f2d717a4332307b0cd7862e39111cd8102054":{"kind":"REWORK","_number":1,"created":"2019-08-09 23:00:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/675710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/675710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/675710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/675710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/675710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/675710/1"}}},"commit":{"parents":[{"commit":"394d4655fa17d0d4066fe8183570a5ce40ba42f5","subject":"Merge \"slo: Add X-Manifest-Etag to responses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/394d4655fa17d0d4066fe8183570a5ce40ba42f5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-09 22:59:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-09 22:59:59.000000000","tz":-420},"subject":"py3: port test/functional/test_versioned_writes.py","message":"py3: port test/functional/test_versioned_writes.py\n\nChange-Id: I8c2517599a6d1cdf68050dc7acb039f09875fd53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a0f2d717a4332307b0cd7862e39111cd8102054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a0f2d717a4332307b0cd7862e39111cd8102054"}]},"branch":"refs/heads/master"},"f1a2dbf76311dba1ccb9d6240322e34d18d2a657":{"kind":"REWORK","_number":2,"created":"2019-08-09 23:01:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/675710/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/675710/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/675710/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/675710/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/675710/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/675710/2"}}},"commit":{"parents":[{"commit":"394d4655fa17d0d4066fe8183570a5ce40ba42f5","subject":"Merge \"slo: Add X-Manifest-Etag to responses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/394d4655fa17d0d4066fe8183570a5ce40ba42f5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-09 22:59:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-09 23:00:58.000000000","tz":-420},"subject":"py3: port test/functional/test_versioned_writes.py","message":"py3: port test/functional/test_versioned_writes.py\n\nChange-Id: I8c2517599a6d1cdf68050dc7acb039f09875fd53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1a2dbf76311dba1ccb9d6240322e34d18d2a657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1a2dbf76311dba1ccb9d6240322e34d18d2a657"}]},"branch":"refs/heads/master"},"227fe30251984feb45b761390a4ec1d8f5cc894d":{"kind":"REWORK","_number":3,"created":"2019-08-10 03:05:01.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/10/675710/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/675710/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/675710/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/675710/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/675710/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/675710/3"}}},"commit":{"parents":[{"commit":"394d4655fa17d0d4066fe8183570a5ce40ba42f5","subject":"Merge \"slo: Add X-Manifest-Etag to responses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/394d4655fa17d0d4066fe8183570a5ce40ba42f5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-09 22:59:59.000000000","tz":-420},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2019-08-10 03:04:55.000000000","tz":-300},"subject":"py3: port test/functional/test_versioned_writes.py","message":"py3: port test/functional/test_versioned_writes.py\n\nChange-Id: I8c2517599a6d1cdf68050dc7acb039f09875fd53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/227fe30251984feb45b761390a4ec1d8f5cc894d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/227fe30251984feb45b761390a4ec1d8f5cc894d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
