)]}'
{"id":"openstack%2Fswift~676414","triplet_id":"openstack%2Fswift~master~I1d2d063904447a444ecb506b93ee698a5181b6b9","project":"openstack/swift","branch":"master","topic":"add_params_head_req","hashtags":[],"change_id":"I1d2d063904447a444ecb506b93ee698a5181b6b9","subject":"Add params to get_*_metadata internal client methods","status":"MERGED","created":"2019-08-14 14:15:22.000000000","updated":"2019-08-14 20:16:46.000000000","submitted":"2019-08-14 20:14:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676414-1565813694292-0fc9d827","meta_rev_id":"fefdc8dffa4a3860a2bd11ccb48698db19e98b91","_number":676414,"virtual_id_number":676414,"owner":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-14 15:47:37.000000000","_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-08-14 20:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-14 18:10:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-14 18:10:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 18:10:49.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-08-14 20:14:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e89a7c5f4722ec1fe967d94380368a31b4fdbf9","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-14 14:15:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2be412674b6a0d159d0ed807de772db6bc88dfdf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 15:25:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7896a059071346bcba572515f1d6494e : SUCCESS in 4m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/32684eed6e0b4646b5aa0ebf07e862ba : SUCCESS in 1h 03m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c9b2e9a24500409e97713320195e86f5 : SUCCESS in 1h 03m 47s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/22e7fdb8ea6b49c48080874d2baeea67 : SUCCESS in 9m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/97da24bf2b2b4b1db9973f183e918108 : SUCCESS in 11m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a8b46732283c4f3188c19804b3bb4200 : SUCCESS in 13m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0676a8a9f9f249c6a1a740b544971525 : TIMED_OUT in 41m 11s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/e9c22209b55d4fce8a04d840add7bf06 : SUCCESS in 9m 52s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/4f578e2b9e754c42ab8ee08541e22264 : SUCCESS in 11m 04s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/843e0d4b79c043aca59142e96f571844 : SUCCESS in 11m 04s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/c7dfce74676e4bb6844aadb5ee662c20 : SUCCESS in 12m 07s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/230f26ece67d41009b371aa1dad9fda9 : FAILURE in 12m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d1022a3994ec45efb98a51c504f9b887 : SUCCESS in 8m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/56e260d5983d4c07a367629dd397145a : SUCCESS in 7m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c59dc3d543cc40babf41b9fe1fe360c9 : SUCCESS in 9m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2452aeb8cde54412948a37147f8b944e : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a157aecbfe7a4dd2895abdaa5581ee3f : SUCCESS in 13m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d47d37c2949f4651b121b70a79aa96e1 : SUCCESS in 31m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4873463140cd49d8ab14eef81846e47f : SUCCESS in 31m 49s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bcd0928e63dd4af7a640eb14e8ca0356 : SUCCESS in 29m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c0e7092616304028a3f19dd946036778 : SUCCESS in 32m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6005371efc3a4dc0bcec8eba7432edc1 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdd1446859c34dee951db879956706c8 : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1e584b76007647b1ba9e1ce61d05505d : SUCCESS in 18m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4bed14a5cbb37401b92bc95728281e84dd777f29","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-14 15:47:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"58fc7c6d63ee3e377d51bad91aef2c57244b2eb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 17:34:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c843ec110b42a49d48623479d6b1d6 : SUCCESS in 4m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/90b37cff97fb4c718e1b92010fc93c1c : SUCCESS in 1h 05m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6e36abf3b7914b4eb4c6c71cccf38c8e : SUCCESS in 57m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fa0054af31ca4dc48fffc1c3fcf83c5c : SUCCESS in 10m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e1d3d13e42f4570a9a3f45c40eb7e26 : SUCCESS in 13m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f2af632228aa4de5a356e29a043a85fd : SUCCESS in 13m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/65ec3bb65e1d4a5db4dfde1dee5f1463 : SUCCESS in 12m 22s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/6e6d9f1dbbdc469f8d52e65de659289f : SUCCESS in 10m 08s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/b392ecb4019b421e843051e04632906c : SUCCESS in 11m 52s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/a8698b2fb9434e82bf8dcb3e15817fa8 : SUCCESS in 10m 07s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/e3d28b4c76bc4c25b2cbd371273cf56a : SUCCESS in 11m 13s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/80c85de484634c6f8eb4d88f4e904628 : SUCCESS in 12m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/36050cdee2f5472093448c9aa4aa92cd : SUCCESS in 11m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bcdeda0312564a238c6c88a8ae3470b4 : SUCCESS in 9m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b18a9f9898754137bb4fd1185f158c23 : SUCCESS in 9m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/94644d7ce7374996b455b264c0e90adc : SUCCESS in 8m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/97f64743078046108872a66a1a7ef92e : SUCCESS in 13m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bcdf53757bad4e4dbf7d2cd34378ae85 : SUCCESS in 31m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b7b70115a94458daf15d4af5ed12766 : SUCCESS in 27m 29s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/06fc22b8e82147548ddf08781c496f10 : SUCCESS in 29m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb3b20097f6f4e17912d8f1e89323d6a : SUCCESS in 30m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0846f8c9972f483d9ef672e1044570f8 : SUCCESS in 12m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c3fd1895bd740a683b188e9a024ad71 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b27d8bd149ae4193b741a9e0698af8e5 : SUCCESS in 20m 17s","accounts_in_message":[],"_revision_number":1},{"id":"94df787f100a105be9252079b775d7523978a379","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-14 18:10:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMight could use some tests that actually exercise the new plumbing, but it looks straight-forward enough.","accounts_in_message":[],"_revision_number":1},{"id":"b0b4d29764b10b1190f55ec3307a96d4d34da48f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 18:11:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab8698dd3761f91b922d03a264002878556c8678","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 20:14:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d468e7b2a0bf4155955a0244e12b3768 : SUCCESS in 4m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/25b8b7a50dfb40e9a44baacde5873288 : SUCCESS in 1h 05m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7629d4112dd940a3b53cbe8ea1c6fc44 : SUCCESS in 54m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7aa464d000b64870bc3e6fa1612d7fb7 : SUCCESS in 8m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/855ad95fbcff4652bf50937e448b38be : SUCCESS in 10m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe8063fcad79452f9dc70fc0af5ba18c : SUCCESS in 9m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f3aab4dd1b504cf3b2505264c2cd2f72 : SUCCESS in 11m 20s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/279ecd03da144bd4837be201f5501e83 : SUCCESS in 11m 08s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/60aa0ea6459649d99c74ca9a82c837c6 : SUCCESS in 9m 54s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/52e678eafc564f1c8b72ebceac784ffd : SUCCESS in 10m 20s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/302829c7c88e4dc39812170da40189f8 : SUCCESS in 10m 50s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/244dcba42980480a9cc9b2f46820fc45 : SUCCESS in 11m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6c85311a5b814bb68a290b609f37302a : SUCCESS in 12m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1dd13f22dbe6415bb1cab46fd0a2b57d : SUCCESS in 9m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b1a6e954f8094ee5a8f67274e2e2a15b : SUCCESS in 8m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ad4b98ea09424f8aa3d4b0f4380a39bb : SUCCESS in 33m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/23deff4ac756466d96d122ed9e5dc860 : SUCCESS in 28m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ed12622da51e4777ab2431e89e7fc265 : SUCCESS in 30m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/88cfe3ed7dfd4ec887eb55fd87f87ede : SUCCESS in 32m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c84f67a01475464a8ef0922429885752 : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98bfd4b78c4d4d099b429f6c784d3698 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9187736e25c456bd631092117e481876bce08096","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 20:14:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6ed9095424cb28a811e69b32107240be83713368","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 20:16:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d1ee947eef93435b8b42a0e1604f5271 : SUCCESS in 1m 17s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"47c7058259b9f69b51d92ee806988db6ca37f28c","revisions":{"47c7058259b9f69b51d92ee806988db6ca37f28c":{"kind":"REWORK","_number":1,"created":"2019-08-14 14:15:22.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/14/676414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/676414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/676414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/676414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/676414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/676414/1"}}},"commit":{"parents":[{"commit":"196113f93e8a7951f9c963d07bb8b023f3ccbd16","subject":"Merge \"Consolidate Container-Update-Override headers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/196113f93e8a7951f9c963d07bb8b023f3ccbd16"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-14 14:13:25.000000000","tz":120},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-14 14:13:25.000000000","tz":120},"subject":"Add params to get_*_metadata internal client methods","message":"Add params to get_*_metadata internal client methods\n\nChange-Id: I1d2d063904447a444ecb506b93ee698a5181b6b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47c7058259b9f69b51d92ee806988db6ca37f28c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47c7058259b9f69b51d92ee806988db6ca37f28c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
