)]}'
{"id":"openstack%2Fswift~676914","triplet_id":"openstack%2Fswift~master~Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4","project":"openstack/swift","branch":"master","topic":"fix-url-404","hashtags":[],"change_id":"Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4","subject":"Fixing broken links","status":"MERGED","created":"2019-08-16 11:44:26.000000000","updated":"2019-08-16 12:23:26.000000000","submitted":"2019-08-16 12:21:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676914-1565958117332-16970a7d","meta_rev_id":"8ea41523057ccc5118a348bbda25d1f0235d41de","_number":676914,"virtual_id_number":676914,"owner":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-08-16 12:21:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":2,"date":"2019-08-16 11:54:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":1,"date":"2019-08-16 11:54:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 11:54:06.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-08-16 12:21:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c6e97b5085e49509bdc872d2b59bb5e0c7dcd51","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 11:44:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a09d6a9267fd94b61ad58a1cd33969775a3d2b6","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 11:44:55.000000000","message":"Topic set to fix-urls-404","accounts_in_message":[],"_revision_number":1},{"id":"0eb989377538582427ce1410255bfd97ddd266f1","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 11:45:15.000000000","message":"Topic changed from fix-urls-404 to fix-url-404","accounts_in_message":[],"_revision_number":1},{"id":"f2f54bdf97fd07d83044615c1836d4c8c3bb2521","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 11:45:55.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"70012d9d15f598564eb3670a535398535d02703d","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-16 11:54:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8866d6bda1a75f1b0d050614fccd36ce2b64012d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:03:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16edd5b985c94f2ca8caccf6deab70a6 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f4aed9e9970111952d55a185de8d86312c6d6ddd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:04:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e3ec34381cef19c22c856dc2f3b7db20f1081ee2","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 12:08:52.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 155e60fcf0fb9e4da96bc122fd2c04760113aa7b","accounts_in_message":[],"_revision_number":2},{"id":"4c83adde0a17d155fe9c7efb49f5ba69c4363a61","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 12:09:09.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 62dd1b0f0c1fe65306fef460e9590e8fb5d6826a","accounts_in_message":[],"_revision_number":2},{"id":"41494f13f7ac392ab15f6230f42564728964afed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:21:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b5b733deb777cf67914bb1a67635816124a2237a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:21:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb9db0729eec44ebb369bd3960713c86 : SUCCESS in 5m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f62b6b8dca924145a9d169c8e74c08b1 : SUCCESS in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ea41523057ccc5118a348bbda25d1f0235d41de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:23:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6d8ac5602117475bb44f5f983d862e72 : SUCCESS in 1m 07s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0c16fd9536ac37c838c4229800abdb684a7bd4d4","revisions":{"976b509465b02ffbed7fc9717b615c14bee65e3d":{"kind":"REWORK","_number":1,"created":"2019-08-16 11:44:26.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/14/676914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/676914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/676914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/676914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/676914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/676914/1"}}},"commit":{"parents":[{"commit":"1f53d912a3fb0ac2ad3dc7af87bad971933c167a","subject":"Merge \"Leave less garbage in /var/tmp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f53d912a3fb0ac2ad3dc7af87bad971933c167a"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-08-16 11:44:13.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-08-16 11:44:13.000000000","tz":60},"subject":"Fixing broken links","message":"Fixing broken links\n\nChange-Id: Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/976b509465b02ffbed7fc9717b615c14bee65e3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/976b509465b02ffbed7fc9717b615c14bee65e3d"}]},"branch":"refs/heads/master"},"0c16fd9536ac37c838c4229800abdb684a7bd4d4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-16 11:45:55.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/14/676914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/14/676914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/676914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/676914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/14/676914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/14/676914/2"}}},"commit":{"parents":[{"commit":"1f53d912a3fb0ac2ad3dc7af87bad971933c167a","subject":"Merge \"Leave less garbage in /var/tmp\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f53d912a3fb0ac2ad3dc7af87bad971933c167a"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-08-16 11:44:13.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-08-16 11:45:52.000000000","tz":0},"subject":"Fixing broken links","message":"Fixing broken links\n\nSmall changes, but helpful, mostly.\n\nBackport: stein rocky\n\nChange-Id: Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c16fd9536ac37c838c4229800abdb684a7bd4d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c16fd9536ac37c838c4229800abdb684a7bd4d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
