)]}'
{"id":"openstack%2Fswift~676917","triplet_id":"openstack%2Fswift~stable%2Fstein~Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4","project":"openstack/swift","branch":"stable/stein","topic":"fix-url-404","hashtags":[],"change_id":"Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4","subject":"Fixing broken links","status":"MERGED","created":"2019-08-16 12:08:52.000000000","updated":"2019-08-16 19:12:23.000000000","submitted":"2019-08-16 19:12:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676917-1565982743135-28760403","meta_rev_id":"d51aac8543dd17053373fbfabbc9d47e6155627b","_number":676917,"virtual_id_number":676917,"owner":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-16 19:12:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-16 18:10:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-16 18:10:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 18:10:55.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-08-16 19:12:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9876b1757ad92e89b427f8467bbcebc649b14de7","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 12:08:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"636cdb52c211dc141b4354ca571cb632208ab550","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 12:09:00.000000000","message":"Topic changed from fix-url-404-stable/stein to fix-url-404","accounts_in_message":[],"_revision_number":1},{"id":"e5f1bc69887c4e2951d24f5cbb635c2eeb3bd872","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:15:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401b191ee36c49ca9ec277aee70becd4 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"86d4f7701002f83535bfee01b94337f02183fa52","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-16 18:10:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport, all docs, and working links! What\u0027s not to love!?","accounts_in_message":[],"_revision_number":1},{"id":"07192e965193d6b304ed8a984f8318366064af66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 18:12:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f80caae9ae3df4cc83f3eb113b0e7472026f129","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 19:12:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d902f8e6d74c159135174acd36f8b5 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d51aac8543dd17053373fbfabbc9d47e6155627b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 19:12:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"155e60fcf0fb9e4da96bc122fd2c04760113aa7b","revisions":{"155e60fcf0fb9e4da96bc122fd2c04760113aa7b":{"kind":"REWORK","_number":1,"created":"2019-08-16 12:08:52.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/17/676917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/676917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/676917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/676917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/676917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/676917/1"}}},"commit":{"parents":[{"commit":"99c7795ca8e65859ea741b75d9c7fda31576eae0","subject":"Merge \"s3api: Block ETag header from CompleteMultipartUpload requests\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c7795ca8e65859ea741b75d9c7fda31576eae0"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-08-16 11:44:13.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-08-16 12:08:52.000000000","tz":0},"subject":"Fixing broken links","message":"Fixing broken links\n\nSmall changes, but helpful, mostly.\n\nBackport: stein rocky\n\nChange-Id: Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/155e60fcf0fb9e4da96bc122fd2c04760113aa7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/155e60fcf0fb9e4da96bc122fd2c04760113aa7b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
