)]}'
{"id":"openstack%2Fswift~676919","triplet_id":"openstack%2Fswift~stable%2Frocky~Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4","project":"openstack/swift","branch":"stable/rocky","topic":"fix-url-404","hashtags":[],"change_id":"Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4","subject":"Fixing broken links","status":"MERGED","created":"2019-08-16 12:09:09.000000000","updated":"2019-08-16 22:44:16.000000000","submitted":"2019-08-16 22:44:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676919-1565995456740-0e67f363","meta_rev_id":"6e1195af4727fe8be49f5c3840a5db558de01622","_number":676919,"virtual_id_number":676919,"owner":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-16 22:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-16 18:11:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-16 18:11:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 18:11:17.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-08-16 22:44:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9a2d858d0a6adcedde77d2d52b9f7d422784d56","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 12:09:09.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9c05ff1688a0e66e912c7c4429af2497b1bd3de7","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 12:09:13.000000000","message":"Topic changed from fix-url-404-stable/rocky to fix-url-404","accounts_in_message":[],"_revision_number":1},{"id":"c378396449b3459b908b6a6e9eea9023a3753a68","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-08-16 12:11:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0637b83e920d7bcc3a9544bf8d78468ecdc631fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:46:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/15ea11483bfd47498ae41243b60f9237 : SUCCESS in 6m 07s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/37b4ec2edcfa459aa830d7755b896c9e : SUCCESS in 33m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b95994550624a3a9a706aeb01b0e9af : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":2},{"id":"6ac0b69a92ba6d00d67d8f14f3f29509ee16b3eb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-16 18:11:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d653b4c64cdaaac00dd92822991d9acb07cd0176","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 18:12:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"219a3809f27fb9e9f2d2992c01ec0f4289c5a470","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 19:12:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/387f1aad782a47e994406291c1f4e9f8 : SUCCESS in 6m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c7dd81a4fe344c8d99206f4adead935b : FAILURE in 29m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b1b067e15bc444a8ceb0d10f374bc5b : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"819dca53dc8129c1d26b81caec04ef16d745481b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-16 21:31:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d49aa5ad036294fcd94dcf90e3baca846ba2cbf3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 22:08:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d28ebdf6d0694b9886fc3da60d336c07 : SUCCESS in 8m 47s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/783eb77fb84141cf9139871856b77854 : SUCCESS in 31m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db85e5155b346c28713d812e553344f : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":2},{"id":"aa8795dd63c8985e3198be1110282928c078753c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 22:09:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ccc5ccb80e81cdbbfe6a5488d87cd62d8adc1ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 22:44:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6e1195af4727fe8be49f5c3840a5db558de01622","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 22:44:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8840d159418846a8a62458aa8bb87518 : SUCCESS in 5m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5c2de23652cd4104be09a21b96f719ac : SUCCESS in 30m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5163832f48e5416c8f3b139501e9e4cf : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6f3a53ee9c04f8ec2233308e54167978c228d786","revisions":{"62dd1b0f0c1fe65306fef460e9590e8fb5d6826a":{"kind":"REWORK","_number":1,"created":"2019-08-16 12:09:09.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/19/676919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/676919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/676919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/676919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/676919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/676919/1"}}},"commit":{"parents":[{"commit":"b26aad755eec43a589f34f0647ad41a3d7f61307","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b26aad755eec43a589f34f0647ad41a3d7f61307"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-08-16 11:44:13.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-08-16 12:09:09.000000000","tz":0},"subject":"Fixing broken links","message":"Fixing broken links\n\nSmall changes, but helpful, mostly.\n\nBackport: stein rocky\n\nChange-Id: Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62dd1b0f0c1fe65306fef460e9590e8fb5d6826a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/62dd1b0f0c1fe65306fef460e9590e8fb5d6826a"}]},"branch":"refs/heads/stable/rocky"},"6f3a53ee9c04f8ec2233308e54167978c228d786":{"kind":"REWORK","_number":2,"created":"2019-08-16 12:11:14.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/19/676919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/676919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/676919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/676919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/676919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/676919/2"}}},"commit":{"parents":[{"commit":"b26aad755eec43a589f34f0647ad41a3d7f61307","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b26aad755eec43a589f34f0647ad41a3d7f61307"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-08-16 11:44:13.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-08-16 12:10:59.000000000","tz":60},"subject":"Fixing broken links","message":"Fixing broken links\n\nSmall changes, but helpful, mostly.\n\nBackport: stein rocky\n\nChange-Id: Ic4b6524d7804d2f74b2973b6acdb9e2679209cd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f3a53ee9c04f8ec2233308e54167978c228d786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f3a53ee9c04f8ec2233308e54167978c228d786"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
