)]}'
{"id":"openstack%2Fswift~676959","triplet_id":"openstack%2Fswift~master~Iea734c42418ca3b28c5c99c0e2425aef2c895401","project":"openstack/swift","branch":"master","topic":"seg_container_policy","hashtags":[],"change_id":"Iea734c42418ca3b28c5c99c0e2425aef2c895401","subject":"Option to set derived container storage policy","status":"NEW","created":"2019-08-16 14:28:12.000000000","updated":"2019-10-30 02:23:11.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5ad2e3f123b765a79716566847e975b82481b145","_number":676959,"virtual_id_number":676959,"owner":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"date":"2019-10-09 14:42:42.000000000","_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":1,"date":"2019-09-05 06:43:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-10-07 22:05:42.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2019-10-30 02:23:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":-1,"date":"2019-10-07 22:25:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 23:39:37.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-09-05 06:43:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-07 22:05:42.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-10-07 22:25:44.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-10-30 02:23:11.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"844e4a25e7b19cb95c13b8fd1018c7004b5ea4e8","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-16 14:28:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3c3dd71fa1bfa3676c5f0b7c8c650c77f11a059","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 15:40:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e91c0daea3194d4daf6afa50f2d6c357 : SUCCESS in 6m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/99bb38a12c5d4e269457fc4d3fd0f95b : SUCCESS in 1h 08m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d9e38949b06542d8a44452648e8b2e2b : SUCCESS in 57m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/36730c72a512405091169fb1453edf3d : SUCCESS in 11m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6aafc7fcca564c6693008a0bd6886f15 : FAILURE in 17m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29fddc9bb4ab4e37a51bfd689ed4bc73 : FAILURE in 18m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7f4baa2f772d472fa7709778232aec28 : FAILURE in 14m 58s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/9b9a6d20000449649cf4ba7fff26be01 : SUCCESS in 10m 57s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/05057af1e117484d846f570ef5cb262c : SUCCESS in 11m 09s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/361ec2a85a6a4452836675295faafc57 : SUCCESS in 10m 46s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/0d68a9d2eb7e43a0b2c0418f60a1161a : SUCCESS in 11m 59s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/ac1c037cf3d1497ba9bdc3a62400d9ce : SUCCESS in 13m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ea823bc69045476ca6096e24f14d453c : SUCCESS in 13m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8da9d221db3d42de9bae8bb2b5588dd5 : SUCCESS in 11m 13s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8bf97cd521014714a9e3bae6076d8402 : SUCCESS in 8m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/447abc224ef748b2bf4efa930a4f1771 : SUCCESS in 9m 51s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/9ef26633bbfc46929bf42ef14b14439e : SUCCESS in 11m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/302ac6f7403e45b0aff6f0b8fed595e5 : SUCCESS in 17m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c441660a71c949d8a594983c71a9576c : FAILURE in 32m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c1c7ae0c63e041679b26f608fb8b75f7 : SUCCESS in 28m 41s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/269d0ecec50d49418be4ee76dfbe4ffa : SUCCESS in 25m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02ae5db33d384b59b26417e36e592819 : SUCCESS in 28m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bcebc737e74c4c7ab77ebd75207759a4 : FAILURE in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf7166acf0d45b59fb03e6a0ea8776e : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8892d9ed14df4bbcb8bf3f98732072bf : SUCCESS in 25m 22s","accounts_in_message":[],"_revision_number":1},{"id":"966ba6a059a7aac01bd130ee448e05c1f053161c","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-19 15:02:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"abd18c45b7a0a38388df2c9aed9cfeb690d00f71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:31:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2feebd0b1cf841458e63a770084434a7 : SUCCESS in 4m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/809aa9e0533244c7b3f8474c75899f7b : SUCCESS in 1h 10m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba2a17cb03da4f61a4db6091c1c6b07d : SUCCESS in 1h 18m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/76441191a88048ada528f7eafc9a4c56 : SUCCESS in 12m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/918e3a3b7f704c8fbefcf650d7316f33 : SUCCESS in 17m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1cc9d9a1ff7a4dbc983d7c0267625825 : SUCCESS in 18m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/24aced9ac1d346c4b810bc32f1979020 : SUCCESS in 13m 45s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/370046078f6a46d688ea81f2ebaf9c8a : SUCCESS in 10m 40s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/2750fc0d926c4dbf9af08c659465f33f : SUCCESS in 11m 25s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/4a942820b68a45f4b48b82d691f7dfb0 : SUCCESS in 12m 26s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/09d50abb7ec6418d9c70dcabd0603076 : SUCCESS in 11m 59s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/865b23580e9646308d0d0fdeeb218ca0 : FAILURE in 14m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/94264178b1584fd8a16a86c822d565e6 : SUCCESS in 13m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e90e16a5ed1e4a539b35aaff87b5c121 : SUCCESS in 10m 51s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/25cd2a3392f2484daab946fcdcca8104 : SUCCESS in 12m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6f00ab01ea564ea08b1a57dc34326e07 : SUCCESS in 11m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5696d81ef0834e75841e6a12e46b2212 : SUCCESS in 14m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3fd1f175430141c88aca003d6ddd39c8 : SUCCESS in 14m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/89da8829d75a49078a21476b62b2cd9a : SUCCESS in 34m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/513f5f744e804e1e83a6250fac4e6471 : SUCCESS in 35m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/83081dc43155498f9392c562d92c6259 : SUCCESS in 36m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/26513fda87af41cfa916a08685d0c5b2 : SUCCESS in 34m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c47f83a93462440992b2274bb9959268 : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcd844dfbd6042489441ba310fcf947b : SUCCESS in 7m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/79d94f27de874932906e79f883a7b243 : SUCCESS in 23m 02s","accounts_in_message":[],"_revision_number":2},{"id":"323c05fdc31265de963fbdabeee1d3e7479d2598","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-08-20 23:39:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nThe code looks okay, but I\u0027m still a little confused.","accounts_in_message":[],"_revision_number":2},{"id":"a98cae701ed30a33aabc7363dd60c0307a1b1a04","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-29 12:38:33.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"500b445aa83e67b28a4471612a5dea6bea571379","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-29 12:40:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"31ef73e17f57de64cfc8667961d71b5adab3afc4","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-29 13:14:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0002216d6b2dae27515444623fab92f902834f81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 15:44:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf96da4acf264c8eb51e4ae4e9da5cf3 : SUCCESS in 5m 19s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4ab34ffe1cbc484c9a04c058499792ac : SUCCESS in 1h 12m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1e61bb4bea643678b4dd374a35cf1b4 : SUCCESS in 59m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ff34bea1bd0b4b3caa11d4c9cf436977 : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fda3200f87584c28b2adb06f3b8a570a : SUCCESS in 15m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/91dfd45bab4445e6ae29cea86c41cb5b : SUCCESS in 14m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/29031754983d40fdb03751b6b67d807b : POST_FAILURE in 11m 40s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/eb7b8edee47a49278b4bee2d5f1a52be : SUCCESS in 12m 42s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/d99e655599e14fd0bcbd9fa1875c966b : SUCCESS in 11m 28s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/29e9abccbb9a4d8d964213a7fd12ae19 : SUCCESS in 10m 42s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/1d372da92bca4b87ad641495b868e4a4 : SUCCESS in 13m 12s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/df5505a67d064bc2a55439baa4893566 : SUCCESS in 14m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dd5421ce6ca844dabddadfa01282445f : FAILURE in 11m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d1c45fb375544e828d7a79c8619adf23 : SUCCESS in 13m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b1270b660b1b4798a38dd8644b204bdb : SUCCESS in 13m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4d33bcd2a46749e5a4c2d9b78e5c3a86 : SUCCESS in 14m 50s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/13b0b5028d224c8dba005c6db23eec7b : SUCCESS in 17m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cde44956cad8409c9fc70bd468ece47c : SUCCESS in 16m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3e534f2c0f8f4605b481bf2b08f41b79 : SUCCESS in 40m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f13dffe2f50b40efba5f568a88e93b5b : SUCCESS in 40m 02s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/27030a67f62949b9a6e8003b567d286f : SUCCESS in 26m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e66c600212db455ea3073090842c6f8a : SUCCESS in 29m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19a502c66879497f962a5578d0600cfa : SUCCESS in 16m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12afde19bb9e46aeb0448f9c84f17a98 : SUCCESS in 7m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d232b0fafdae4548897add51485a08ec : SUCCESS in 22m 23s","accounts_in_message":[],"_revision_number":3},{"id":"8de191d17954f91d8e219726b4103d9f61d4fadb","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-09-04 12:11:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"1754720a76e14c308824b7aa70ce2fce13c852ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:35:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abb5cb43c37c499fad6598651f56bac7 : SUCCESS in 5m 51s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c2702b1365b04a48aac57a991a917063 : SUCCESS in 1h 01m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5cd8f1bf4318499c858815c1e467fd01 : SUCCESS in 1h 22m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5b172b133e1e4b6a88435510a1e7f5b7 : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/004e74296fbb4028b030b0ff29feed98 : SUCCESS in 14m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ea6950a7d6cf48a2959ac1a90b908ba5 : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/73c7681ab9634f9a98a2366d7d7a17b0 : SUCCESS in 12m 59s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/528953e9f71342bcacf9c4493c61b821 : SUCCESS in 10m 45s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/742d7fdc696346ad9568956f51b85131 : SUCCESS in 10m 06s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/9bd13f7a4ee9411fa419e3a17848e63b : SUCCESS in 10m 30s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/975133d884eb4f6b8fc662112c8625d4 : SUCCESS in 12m 03s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/02bc64d3ff974303a9a4818be791e267 : SUCCESS in 13m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/44db976333734d2fb43a0b09fa7069a0 : SUCCESS in 11m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/559fc7e2891a46728146b5a906e12011 : SUCCESS in 11m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/de8570c84b8d4451a4294fcecd1229cd : SUCCESS in 14m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5365c11eb4c8454c8486db66f9fcd26a : SUCCESS in 12m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/50772d87c4e449e993eb695fd7930d4b : SUCCESS in 14m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/82905d7213734077bd4199e345d1efd8 : SUCCESS in 13m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5644af3e59a84a76933ad75110878ca6 : FAILURE in 33m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c148f6df2954aaaa6871a5feeb016a0 : SUCCESS in 30m 10s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e23124b4c41d429aa0696d54d6c85945 : SUCCESS in 29m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4e67c2de097846218f2046090b47c216 : SUCCESS in 30m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c42da8655f4a402bb5886ae019b3b3f0 : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe4c477cab8044189f495cca745d4209 : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/472ae3b1f7cc4fac96b88777cb574bac : SUCCESS in 18m 10s","accounts_in_message":[],"_revision_number":3},{"id":"e3775d068c7c20feb0f676c126be6db4c49c9987","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-09-04 19:43:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2e9fe88abd590fe9f938b65dd9bc2752656dac71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 06:43:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9676dabd2ebb4892bf538c402f7d8bc5 : SUCCESS in 7m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4b5b553ed24c457bbec3e30086c01210 : SUCCESS in 57m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/94e15200a5764a03b4caa021ee3200aa : SUCCESS in 1h 04m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f0a7c0ca5b7c49bf990ef635e0c1d807 : SUCCESS in 10m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e5e95fbc8b14a71b7905a2c0f2c0819 : SUCCESS in 11m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13bf4a42f21d483a913848ee796a8a0c : SUCCESS in 12m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9e34182553ec410ba10550f8e670ab93 : SUCCESS in 13m 51s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/1dbb74ac30054d12901a7ca672d16602 : SUCCESS in 9m 58s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/15328b11a88f4574b7fdd1f3061e6db1 : SUCCESS in 9m 21s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/c5f0bce3db4547f9a8cb91047a030ef8 : SUCCESS in 10m 16s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/33b0dbfc2c5a4d07a9468b5614565fd6 : SUCCESS in 10m 15s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/9dc719743bae4abeb563e8d40ae05ba1 : SUCCESS in 11m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2dbf49378d684acd872e573fc21c600f : SUCCESS in 11m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6e2c7181ab0448808ebcfd9b40b9989c : SUCCESS in 12m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4dd05cfba9764abaa3d93b1dc396c3a3 : SUCCESS in 10m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dbaad4f1faef4958a262c565b3f16794 : SUCCESS in 12m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/628cf30998a34496a811326d986d5e7f : SUCCESS in 14m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d8de4059d6d94f33a707ca42bc1d0bff : SUCCESS in 13m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/83b7e508d498460ebdee04e4c4d3022c : SUCCESS in 34m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/76bd48d974324b66b3cc10d790e64e84 : SUCCESS in 30m 35s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/74ee011121a34e25beda36bf92b1366c : SUCCESS in 38m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b8f446672ad466b95b8f87820c9b9d4 : SUCCESS in 36m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f171421c2684eb4b54265c514b35b6a : SUCCESS in 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df94253fe75c419987cc2e1a74aa631c : SUCCESS in 7m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f4a45c94135480f861f8aea2e665189 : SUCCESS in 19m 24s","accounts_in_message":[],"_revision_number":3},{"id":"4567d60517476b0a9b9792fb44f78048bd0e9056","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-10-07 20:44:09.000000000","message":"Patch Set 3: Code-Review+1\n\n(8 comments)\n\ni think we\u0027d have greater forward progress on this already if we took smaller steps\n\nstep #1 would be *just* fix lp bug #1832390 by always using the parent container\n\nafter that, next steps, we can figure out how we want to do api\u0027s and configuration for extra buckets\n\nI\u0027m struggling just because `aws s3api create-bucket` doesn\u0027t seem to have a --add-header option and I don\u0027t know how to get permissions right on a bucket I created using the swift API","accounts_in_message":[],"_revision_number":3},{"id":"72c6830f396424bf347187917217be680ad3e09c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-07 22:05:42.000000000","message":"Patch Set 3:\n\n(1 comment)\n\n\u003e I don\u0027t know how to get permissions right on a bucket I created using the swift API\n\nTry turning off s3_acl :-/\n\nThis all reminds me of recurrent requests to have per-account default policies... and even per-account sets of *allowed* policies....","accounts_in_message":[],"_revision_number":3},{"id":"23a04c6657d90827f75ab81ad78454840db5e4ce","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-10-07 22:25:44.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nThis seems to work like you might expect except for the default option is wrong\n\nAfter I installed this change I created a bucket with a non-default storage policy using boto3 with:\n\nhttps://gist.github.com/clayg/1df576ef1d247ffee734f4a64097a8ae\n\nBut when I created a multipart upload from the command line using:\n\naws s3api create-multipart-upload --bucket foo --key bar\n\nthe resulting foo+segments bucket still had the default storage policy!\n\nI had to go change the config option and restart my proxy to actually *fix* lp bug #1832390\n\nFWIW, I\u0027d be happy to advocate the option should go away, but I don\u0027t think we should merge it at all with the wrong default.\n\nI also don\u0027t love the mechanism for the api/metadata, but it might not be a blocker for me if we had a good fix!","accounts_in_message":[],"_revision_number":3},{"id":"3121baf541e05d00f6c32b2ad0b65f7da9907c82","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-10-09 14:42:42.000000000","message":"Patch Set 3:\n\n@Clay, @Tim: agree with reducing scope, I\u0027ve proposed https://review.opendev.org/#/c/687577/ to create the s3api segments bucket with the same policy as the primary container. We can then continue discussion on this patch for the second use-case where a user might want a primary container to be in one policy and the secondary in a different policy.","accounts_in_message":[],"_revision_number":3},{"id":"91c2906de9124dc14cf1cc35afa1942e6fbcfc16","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-10-30 02:23:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c13529beac75a52bb3b30e7a1d6f4e86ad69a51e","revisions":{"e7aef40f9a29c0bc38a7908e094d9883fffdf559":{"kind":"REWORK","_number":1,"created":"2019-08-16 14:28:12.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/59/676959/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/676959/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/676959/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/676959/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/676959/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/676959/1"}}},"commit":{"parents":[{"commit":"61881797c76cd67a7eaf5117eed5c6b3e3ab0229","subject":"Merge \"s3api: Translate 503 to S3-style 503s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61881797c76cd67a7eaf5117eed5c6b3e3ab0229"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-16 14:04:34.000000000","tz":120},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-16 14:27:47.000000000","tz":120},"subject":"Option to set derived container storage policy","message":"Option to set derived container storage policy\n\nCurrently, when a middleware or even a client needs to automatically\ncreate a container that is derived from a primary container, there\u0027s\nno indication (besides the \u0027default\u0027 policy) to which storage policy\nshould be set for the new container. For example: When the s3api\nmiddleware creates a \u0027foo+segments\u0027 during an MPU upload to to\ncontainer \u0027foo\u0027, there\u0027s no clear indication if the middleware\nshould use the cluster\u0027s storage-policy or foo\u0027s storage-policy.\n\nThis patch provides two new options. One is a configuration option\nthat can be set by the operator and the second is a new header\nthat can be set on a container. It is important to note that both\nof these options only serve as an indicator to middleware and\nclients devs. There\u0027s no way for Swift to enforce that a new\ncontainer is using the expected policy.\n\nThe configuration option `derived_container_policy_use_default`\ncan be set by cluster operators. If set to true, then the\nderived container should use the cluster\u0027s default policy. If\nset to False, the derived container should use the same policy\nas the primary container.\n\nThe header `X-Derived-Container-Storage-Policy` allow users\nto override and define exactly what policy should be created\nwhen a derived container is automatically created.\n\nAs part of this patch, the new options are also used in the s3api\nmiddleware to set the correct policy to be used during an\nMultipart Upload.\n\nChange-Id: Iea734c42418ca3b28c5c99c0e2425aef2c895401\nCloses-Bug: 1832390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7aef40f9a29c0bc38a7908e094d9883fffdf559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7aef40f9a29c0bc38a7908e094d9883fffdf559"}]},"branch":"refs/heads/master"},"d53486bca8173f64a876fae4a8f410d22a885198":{"kind":"REWORK","_number":2,"created":"2019-08-19 15:02:13.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/59/676959/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/676959/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/676959/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/676959/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/676959/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/676959/2"}}},"commit":{"parents":[{"commit":"61881797c76cd67a7eaf5117eed5c6b3e3ab0229","subject":"Merge \"s3api: Translate 503 to S3-style 503s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61881797c76cd67a7eaf5117eed5c6b3e3ab0229"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-16 14:04:34.000000000","tz":120},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-19 15:01:54.000000000","tz":120},"subject":"Option to set derived container storage policy","message":"Option to set derived container storage policy\n\nCurrently, when a middleware or even a client needs to automatically\ncreate a container that is derived from a primary container, there\u0027s\nno indication (besides the \u0027default\u0027 policy) to which storage policy\nshould be set for the new container. For example: When the s3api\nmiddleware creates a \u0027foo+segments\u0027 during an MPU upload to to\ncontainer \u0027foo\u0027, there\u0027s no clear indication if the middleware\nshould use the cluster\u0027s storage-policy or foo\u0027s storage-policy.\n\nThis patch provides two new options. One is a configuration option\nthat can be set by the operator and the second is a new header\nthat can be set on a container. It is important to note that both\nof these options only serve as an indicator to middleware and\nclients devs. There\u0027s no way for Swift to enforce that a new\ncontainer is using the expected policy.\n\nThe configuration option `derived_container_policy_use_default`\ncan be set by cluster operators. If set to true, then the\nderived container should use the cluster\u0027s default policy. If\nset to False, the derived container should use the same policy\nas the primary container.\n\nThe header `X-Derived-Container-Storage-Policy` allow users\nto override and define exactly what policy should be created\nwhen a derived container is automatically created.\n\nAs part of this patch, the new options are also used in the s3api\nmiddleware to set the correct policy to be used during an\nMultipart Upload.\n\nChange-Id: Iea734c42418ca3b28c5c99c0e2425aef2c895401\nCloses-Bug: 1832390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d53486bca8173f64a876fae4a8f410d22a885198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d53486bca8173f64a876fae4a8f410d22a885198"}]},"branch":"refs/heads/master"},"c13529beac75a52bb3b30e7a1d6f4e86ad69a51e":{"kind":"REWORK","_number":3,"created":"2019-08-29 13:14:48.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/59/676959/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/676959/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/676959/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/676959/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/676959/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/676959/3"}}},"commit":{"parents":[{"commit":"61881797c76cd67a7eaf5117eed5c6b3e3ab0229","subject":"Merge \"s3api: Translate 503 to S3-style 503s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61881797c76cd67a7eaf5117eed5c6b3e3ab0229"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-16 14:04:34.000000000","tz":120},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-29 13:14:35.000000000","tz":120},"subject":"Option to set derived container storage policy","message":"Option to set derived container storage policy\n\nCurrently, when a middleware or even a client needs to automatically\ncreate a container that is derived from a primary container, there\u0027s\nno indication (besides the \u0027default\u0027 policy) to which storage policy\nshould be set for the new container. For example: When the s3api\nmiddleware creates a \u0027foo+segments\u0027 during an MPU upload to to\ncontainer \u0027foo\u0027, there\u0027s no clear indication if the middleware\nshould use the cluster\u0027s storage-policy or foo\u0027s storage-policy.\n\nThis patch provides two new options. One is a configuration option\nthat can be set by the operator and the second is a new header\nthat can be set on a container. It is important to note that both\nof these options only serve as an indicator to middleware and\nclients devs. There\u0027s no way for Swift to enforce that a new\ncontainer is using the expected policy.\n\nThe configuration option `derived_container_policy_use_default`\ncan be set by cluster operators. If set to true, then the\nderived container should use the cluster\u0027s default policy. If\nset to False, the derived container should use the same policy\nas the primary container.\n\nThe header `X-Derived-Container-Storage-Policy` allow users\nto override and define exactly what policy should be created\nwhen a derived container is automatically created.\n\nAs part of this patch, the new options are also used in the s3api\nmiddleware to set the correct policy to be used during an\nMultipart Upload.\n\nChange-Id: Iea734c42418ca3b28c5c99c0e2425aef2c895401\nCloses-Bug: 1832390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c13529beac75a52bb3b30e7a1d6f4e86ad69a51e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c13529beac75a52bb3b30e7a1d6f4e86ad69a51e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
