)]}'
{"id":"openstack%2Fswift~677160","triplet_id":"openstack%2Fswift~stable%2Fqueens~I2a0695b59875a58c4ea74fc1d3a3622ad016f3c0","project":"openstack/swift","branch":"stable/queens","topic":"goodbye_42.3","hashtags":[],"change_id":"I2a0695b59875a58c4ea74fc1d3a3622ad016f3c0","subject":"Remove openSUSE 42.3 job","status":"MERGED","created":"2019-08-19 09:37:25.000000000","updated":"2019-08-19 20:42:27.000000000","submitted":"2019-08-19 20:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677160-1566247347493-ac68746e","meta_rev_id":"48ca061fa259e5222502943a97f4731fea7b5c69","_number":677160,"virtual_id_number":677160,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-19 20:42:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-19 16:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-19 16:12:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 16:12:49.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-08-19 20:42:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35cdd33efd1aee4094039710164bfe98f267bb04","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-19 09:37:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"57ab73180a72e90cf865789813afa60a52b7643c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-19 09:37:33.000000000","message":"Topic changed from goodbye_42.3-stable/queens to goodbye_42.3","accounts_in_message":[],"_revision_number":1},{"id":"21e2934dadb01cadcac4cb9ceda4f32d3c2d28bb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-19 09:37:39.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit fbfb89bebd9d30692293c71c5b9b0c55904320b7","accounts_in_message":[],"_revision_number":1},{"id":"58192c89778effbb30044b9efd6ebc5dd7701564","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 11:23:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/602a4ba7e22649db82d32c5ed2d60e15 : SUCCESS in 9m 25s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/08a1330180fd4a79b345ad2f701dde8f : SUCCESS in 12m 10s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/6b4c8aaf2eae4fd8b015369deeb4bb1c : SUCCESS in 3m 41s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/2b85b45e63b644fa95f20c46457b19df : SUCCESS in 12m 17s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/7bb9e3995d884df7a49c95fa9c978fa3 : SUCCESS in 8m 04s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/134dc9a49a964396b0d31477565cd3f2 : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6347196f791b4ccea7f67b72caffe612 : SUCCESS in 8m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3605841ad034ec98b79fdfb35aab4c8 : SUCCESS in 31m 59s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/f54148038ae641719ddda50953cab047 : SUCCESS in 1h 43m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b075b928c0d44632b13ca000fbf5a5d1 : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":1},{"id":"040f7b2a612ac5969a087ccd1bca4151ce40fb3b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-19 16:12:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f25ef0fb4734b195929a536d886830cf16c4779d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:13:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6c6433dec440674628fff2b8d906dbb241e22b34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 20:42:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e105e5f3535c491b9c087a9d6d9db9e5 : SUCCESS in 6m 25s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a37b355e3a2e4cfd9de376d1a09c7fbe : SUCCESS in 7m 58s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/0bae552f4b2b4859b83b9ce1ccde874f : SUCCESS in 6m 40s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/cb85ca2d0bd94fc79d12aca727b5d415 : SUCCESS in 13m 12s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/28e2dbf6a9734d60a7c125d74c27e529 : SUCCESS in 13m 30s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/bc3209970d1e4c259292d3e9977cea94 : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e182119b05304c5ab034fa9adfe5a17e : SUCCESS in 7m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b61889b286f4d4299e08129596548bc : SUCCESS in 28m 32s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/80ac1e220f1646b8bd9d80ee3e2a567b : SUCCESS in 1h 26m 33s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/be305d46d1a14a57b42374fd4acf953c : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":1},{"id":"48ca061fa259e5222502943a97f4731fea7b5c69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 20:42:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c8b310021d8b1e373c99594d153458109a52fd2","revisions":{"9c8b310021d8b1e373c99594d153458109a52fd2":{"kind":"REWORK","_number":1,"created":"2019-08-19 09:37:25.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/60/677160/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/677160/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/677160/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/677160/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/677160/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/677160/1"}}},"commit":{"parents":[{"commit":"503b55ab598b856441b92b942e2cfa990e940c12","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/503b55ab598b856441b92b942e2cfa990e940c12"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-08-19 09:36:02.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2019-08-19 09:37:25.000000000","tz":0},"subject":"Remove openSUSE 42.3 job","message":"Remove openSUSE 42.3 job\n\nopenSUSE 42.3 is dead, remove an experimental job.\n\nNote this was removed in master as part of\nIf32b164827d07c4b3efc88fa2794e38ca67c3b53.\n\nChange-Id: I2a0695b59875a58c4ea74fc1d3a3622ad016f3c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c8b310021d8b1e373c99594d153458109a52fd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c8b310021d8b1e373c99594d153458109a52fd2"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
