)]}'
{"id":"openstack%2Fswift~677824","triplet_id":"openstack%2Fswift~master~I6b4365c40ea8b606cb104cc92ab11cd40767aa78","project":"openstack/swift","branch":"master","topic":"bug/1840322","hashtags":[],"change_id":"I6b4365c40ea8b606cb104cc92ab11cd40767aa78","subject":"squash: maybe clean up some func tests?","status":"ABANDONED","created":"2019-08-21 20:44:34.000000000","updated":"2020-01-16 20:44:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d925ee58d1e38c11770c0544d65c8d6e5a703e6b","_number":677824,"virtual_id_number":677824,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2019-08-21 21:24:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-21 21:24:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4068445964fe365e02f677a9950b5965e4b25bb","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-21 20:44:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a547e19fcfc2f853b29b6097dcc0e3ddec801d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 21:24:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d28ddade90714fa697e282cc64b093e8 : SUCCESS in 6m 37s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/9beca2fb00574ab6b9191256423211a4 : SUCCESS in 9m 26s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/c5f9c3cee0c24d9b9449bd23aad1fe5f : SUCCESS in 10m 21s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/96ab62141c38424db6ee6277218fe858 : SUCCESS in 12m 21s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/28fc5d5426564a8a9322cb5617ee2ba9 : SUCCESS in 12m 24s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/7a59b424581d4df583a576771e56d035 : SUCCESS in 14m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/35bed4a192e24727bcc5c63ad4cfb706 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b2973de9b9fa4ef4855f45768e162e94 : SUCCESS in 10m 15s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/93df3a131197428ca1c8bd53d2ea9313 : SUCCESS in 11m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/107315a1afea48eb82b5cb26a4cb7223 : SUCCESS in 13m 57s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/41b2fce3d02c4316826f8be8b169b507 : SUCCESS in 14m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4fd4abd06d2045d0816955aa94d8f87e : SUCCESS in 18m 02s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/263aed96eef0470486bf8e770c6c33b5 : SUCCESS in 27m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fb2d903df3e14f598ee4ad20329f54f2 : SUCCESS in 29m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7113333177684ad1b7bb641d4921665f : SUCCESS in 34m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53189f5848b846568b350ec1d7ad8a8e : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b5a8fe3feb14132a6a1a2f46f858971 : SUCCESS in 21m 14s","accounts_in_message":[],"_revision_number":1},{"id":"3b4b85e00aa6bd1cf12336230281b9821c7193c7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-16 20:44:40.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.opendev.org/#/c/682382/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fccc5f2375c7a24bd9d45129cd19e98fa13dd54","revisions":{"2fccc5f2375c7a24bd9d45129cd19e98fa13dd54":{"kind":"REWORK","_number":1,"created":"2019-08-21 20:44:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/24/677824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/677824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/677824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/677824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/677824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/677824/1"}}},"commit":{"parents":[{"commit":"cbc9e052491a9dc36f6920489d6bb395d23d5533","subject":"symlink-backed versioned_writes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbc9e052491a9dc36f6920489d6bb395d23d5533"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-21 19:00:13.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-21 20:18:01.000000000","tz":-420},"subject":"squash: maybe clean up some func tests?","message":"squash: maybe clean up some func tests?\n\nChange-Id: I6b4365c40ea8b606cb104cc92ab11cd40767aa78\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fccc5f2375c7a24bd9d45129cd19e98fa13dd54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fccc5f2375c7a24bd9d45129cd19e98fa13dd54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
