)]}'
{"id":"openstack%2Fswift~677848","triplet_id":"openstack%2Fswift~feature%2Flosf~I4150fd0152a6a3fd3f31b86315f212316f27ec39","project":"openstack/swift","branch":"feature/losf","topic":"merge-from-master","hashtags":[],"change_id":"I4150fd0152a6a3fd3f31b86315f212316f27ec39","subject":"Merge branch \u0027master\u0027 into feature/losf","status":"MERGED","created":"2019-08-21 22:15:23.000000000","updated":"2019-08-22 05:20:22.000000000","submitted":"2019-08-22 05:17:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677848-1566451075405-b79180e0","meta_rev_id":"279417029642d646eb1a8f8b4bf91d48dc4776a4","_number":677848,"virtual_id_number":677848,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-08-22 05:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-08-21 23:13:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-21 23:13:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 05:17:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e846aab751abe4e2811c0dad8bae8b40f073bd3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-21 22:15:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16ee5c4ead3a99ab313f471d4f66c80f8a62b4a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 23:02:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29869f7a5a8b47499cd677c32a851714 : SUCCESS in 6m 07s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae796cf646f84fcdb1dd8ffaa6172145 : SUCCESS in 15m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/406262763b124aab902ebc93d6f11be7 : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/34562a4cd318489194e39c623ec03527 : SUCCESS in 12m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2eed06a927954a8c9e96ae2ddf5dadef : SUCCESS in 12m 23s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/55cf869c59f64e2d885cdd2632e2a91a : SUCCESS in 10m 18s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/82f6e46ed3904686a1947c3b95d5927f : SUCCESS in 11m 52s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/930c2cdc2849409ca63b6524c0f26986 : SUCCESS in 10m 22s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/ce81b0a5d9bb481196372bbfdac2c2ef : SUCCESS in 15m 43s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/3217a8b5282a4c0b84cafd7f348618a2 : SUCCESS in 14m 14s\n- swift-tox-func-losf https://zuul.opendev.org/t/openstack/build/912ec63dd3c14f7b86d0ac747f9667b7 : SUCCESS in 21m 19s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/05bcb781847449bfb0bf2b9155fd6462 : SUCCESS in 13m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6166619fd61946c480f92ea16b8fd89c : SUCCESS in 14m 13s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7bd2a9aa41714cb2b05e892fd5ffad2b : SUCCESS in 10m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/658e4f30c35a4570a872b6b35a567b3e : SUCCESS in 13m 46s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/371a2f28803b46c6b1d778987e438aad : SUCCESS in 12m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/397dd42b62244526907bd35b1fe52ed2 : SUCCESS in 16m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1c919547dfa446c9b2a2a1f2abdcb0b9 : SUCCESS in 41m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/78a9b0e54b7847e39d27e36e5c4e452b : SUCCESS in 30m 47s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d56dfb8ffed7486baa6e07b500b1bd36 : SUCCESS in 31m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/33b0962f34ba49baa1070b6be964f75d : SUCCESS in 33m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1616f07aa83c42838f82c88594c93e59 : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9fa2430b08740eebe3af5d1e3f136c6 : SUCCESS in 8m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/edd29ebac75144afa7ef2e49595f5a38 : SUCCESS in 21m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8a22e4b18f74e8a69789e3550225c7beda078d43","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-21 23:13:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"508ce56ccc16a4237af4cb0fbd257550e56c875a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 23:14:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d661787e0ee3b107ad2329cae8e5c5a64a90e95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 05:17:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a5dc79996f4ee0b7da03709d833f13 : SUCCESS in 5m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/24f5b437322b4e24b8d120ea62ea905f : SUCCESS in 18m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/34419dff5a9145f9a653472a2701e65d : SUCCESS in 11m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3968ad13433c422e921de5fe982e1831 : SUCCESS in 12m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8fd14fdb573f4e4cbea0eb05fb5ea419 : SUCCESS in 11m 43s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/4c81ca8e7cbd4a5faa7c36d922886451 : SUCCESS in 11m 14s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/6c84e85e182345ca89d06b516d1984f3 : SUCCESS in 11m 52s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/3ee163c56f704e5da152bc64774ea9fe : SUCCESS in 11m 21s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/fe332d9b6c2644c88be7ec190847e932 : SUCCESS in 10m 53s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/6c523e7305d04df68526b50530dce1ce : SUCCESS in 14m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7df5dc9160f4407f85e330f1980a1b31 : SUCCESS in 13m 10s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c3828214aebc4d6c80ed20616eabc2cd : SUCCESS in 11m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4ef07942706748cfb8136f6e8f4e0baf : SUCCESS in 11m 15s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ce7bfab278a54530b1c5b5da84717b74 : SUCCESS in 12m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2dcf900cf95d47c09dc5c7888137fe1e : SUCCESS in 34m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac70ae681f5e4b708ba721c86216451d : SUCCESS in 31m 36s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/da755623d3594aa887e07be46b8a02b3 : SUCCESS in 29m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/160c58644b5f4798b186a0d187f16caf : SUCCESS in 28m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9a629cb02dc40d78e0c1de4f20f4232 : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c27814aaa49448129d1af56a8b2cadb3 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fc87f894b1cafe284465998be560aff16ce21fb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 05:17:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"618b0d9c5877599e9b49bb6b03d5afb4a41f5c4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 05:20:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/357e6272e83948949a8eb26c11da07e9 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5c80dc3a2fa34addb32a73c0b4d5bcc0 : SUCCESS in 1m 09s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6204a2bc571a51e91f97ad4342b8ae216aaf6f36","revisions":{"6204a2bc571a51e91f97ad4342b8ae216aaf6f36":{"kind":"REWORK","_number":1,"created":"2019-08-21 22:15:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/677848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/677848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/677848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/677848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/677848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/677848/1"}}},"commit":{"parents":[{"commit":"ea5cbf633cfa56a418a79cde4ccfea0c1f6f3f92","subject":"Merge remote-tracking branch \u0027remotes/origin/master\u0027 into feature/losf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea5cbf633cfa56a418a79cde4ccfea0c1f6f3f92"}]},{"commit":"ff5ea003b3a1b37d8417aa17d3521237768dfe62","subject":"ec: log durability of frags that fail to reconstruct","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff5ea003b3a1b37d8417aa17d3521237768dfe62"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-21 22:13:12.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-21 22:13:12.000000000","tz":-420},"subject":"Merge branch \u0027master\u0027 into feature/losf","message":"Merge branch \u0027master\u0027 into feature/losf\n\nChange-Id: I4150fd0152a6a3fd3f31b86315f212316f27ec39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6204a2bc571a51e91f97ad4342b8ae216aaf6f36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6204a2bc571a51e91f97ad4342b8ae216aaf6f36"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
