)]}'
{"id":"openstack%2Fswift~678277","triplet_id":"openstack%2Fswift~master~I5dbe5bdca17e624963cb3a3daba3b240cbb4bec4","project":"openstack/swift","branch":"master","topic":"change_stack_test","hashtags":[],"change_id":"I5dbe5bdca17e624963cb3a3daba3b240cbb4bec4","subject":"Add func test for changing versionining modes","status":"MERGED","created":"2019-08-23 17:49:14.000000000","updated":"2019-09-24 23:42:23.000000000","submitted":"2019-09-24 23:41:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678277-1569368462326-5622d755","meta_rev_id":"f67e779e5242544ac3b7082b6c574d193a644048","_number":678277,"virtual_id_number":678277,"owner":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-24 23:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-24 17:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-24 17:17:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 17:17:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-09-24 23:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c037d8b0b8fdada777bdc74ce556c81acc7ca8e","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-23 17:49:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23c153a723943617092d5d5ff77ac1f01bda502d","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-08-23 17:50:20.000000000","message":"Patch Set 1: Workflow-1\n\nnot really to be merged, i\u0027m adding same test to symlink versioning.","accounts_in_message":[],"_revision_number":1},{"id":"20d28a9e9118a7d93b6dce13c915c8d5a410f1fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 19:55:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68bdf5025e954a09a19089139f1ba3d7 : SUCCESS in 6m 19s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/150b41c8e99e4b02a51a3f74cc19032f : SUCCESS in 10m 56s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/d6aeb0ff5ce74a6e973f659d9cdbcbad : SUCCESS in 10m 32s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/4d772787a46b4d9cbbd5b97a070be8be : SUCCESS in 10m 31s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/6c5644788ddc441d8f9b61cf109444e2 : SUCCESS in 10m 26s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/39f706d4c37d4c87b304d2d447a50d90 : SUCCESS in 13m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/120a35ca977c4656881eb334eb339906 : SUCCESS in 12m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f7adaa2290de4795a98188372eeca3ed : SUCCESS in 11m 14s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/dde549771d1a4053ad9ae4924603ed97 : SUCCESS in 11m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/00be262920ff4671b3665a5360393dc0 : SUCCESS in 11m 43s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/20e24ea7529d4097966d3c15b8a49122 : SUCCESS in 14m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/96f98c3779f24a85ac3a9f142fd3b5a3 : SUCCESS in 15m 07s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/411ef5d36c0c4e2f8ac5b35f7e3f4ca0 : SUCCESS in 26m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b47f4a35eae846ccb8fab5623f203b8c : SUCCESS in 30m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/32a35d631b8a4cb484fb6e85b3c21e16 : SUCCESS in 28m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/526c413d4ae642578d8c1983bf992755 : SUCCESS in 7m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9217adb8432e4d9fb27ecc113f472db4 : SUCCESS in 22m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f4a45ae04e80d28c9efe1a3d65783d5457573602","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-09-24 12:23:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ec4d594c726fd336e8a74cce6a65bc01c2d21ac","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-09-24 12:29:01.000000000","message":"Patch Set 2:\n\nwith the changes in plans to add the new mode separately from versioned_writes.py we should in fact review and look to merge this test.","accounts_in_message":[],"_revision_number":2},{"id":"1590e826433b7d399cedd019061e2283babb3f88","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-24 15:16:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nYeah, that all looks sane -- matches my expectation about how history and stack modes interoperate. Might be good to have a UTF8 variant (see also: https://review.opendev.org/#/c/684146/), but that could be done as a follow-up.","accounts_in_message":[],"_revision_number":2},{"id":"961aaab2db88b331b506ce84ffa3fe51cca89e89","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-24 17:16:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3661509b9e87e55ef5a3e8cd7c0d09c12cdaf6fd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-24 17:17:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nMinor pep8 fix.","accounts_in_message":[],"_revision_number":3},{"id":"2a753ab524b05927c7d2ce8dcf1b08ba935793d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 21:40:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/518c64055aa2468d8e2c07b1a7111d2f : SUCCESS in 6m 54s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/7d2ea85975314184a8f7507c78ed2e7a : SUCCESS in 10m 16s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/00289f11ba284f5998fc1f9a99c52378 : SUCCESS in 10m 45s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/2b7bfcfce5a24c609a946112d1d40214 : SUCCESS in 10m 09s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/46dc3ad536a74f0fa76349f5a99bd00e : SUCCESS in 11m 12s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/91655a40ecac45c9bfa6c07410eaa74a : SUCCESS in 12m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6d13b22894184fb9b07bef03c26209ef : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2ab1e14c1ce84f31a57aed73929fa3e2 : SUCCESS in 9m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/4971f6c640f04430ab3fa4ddc50d168f : SUCCESS in 9m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c87b824e0a544bb4b3879ac7dd42f9cd : SUCCESS in 9m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2a7d744622d24568bc55fdf1b63a600e : SUCCESS in 14m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dcb6307ac67448a08685f59a4f526701 : SUCCESS in 22m 30s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7cb44999a84843b182ce100ce2a852a7 : SUCCESS in 32m 28s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2b217433728b4db78f00a8172a3c66ec : SUCCESS in 29m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/454c753f7d604be890c7d2bbe08fd8a5 : SUCCESS in 29m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c817edfe954b6592e54aa548c6adb0 : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/50508793064e4437ae64e739a8060c8c : SUCCESS in 24m 29s","accounts_in_message":[],"_revision_number":3},{"id":"6d282baf02ff8ff4b5c6a80594c5435c98472118","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 21:41:21.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"94f915515e7cbd7f0a4c6e1c619a5547d9f5dba1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 23:41:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f83fba3359e64b15a9a93c8ac2d16906 : SUCCESS in 6m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0f9538a6643a4f98baf49a87ec44c403 : SUCCESS in 11m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/793e05cc4ade4428be4b05e456a16a58 : SUCCESS in 13m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a3fb099b5b984b119b418c32f828b9de : SUCCESS in 11m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4eb38100a5604a5285ab0cb47c486bda : SUCCESS in 11m 24s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/cdb023684e484ef182e2529a630e98d9 : SUCCESS in 8m 59s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/4793225b7c2f4cc9a8f64860fe8ba422 : SUCCESS in 10m 50s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/1afcf13b5f274b3ba1e4a38c39c46d31 : SUCCESS in 11m 45s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/6b4ff110c77a4c38a5d5925aad9bbd45 : SUCCESS in 10m 55s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/597da7c7f3814ab9a6f7e93e77818a26 : SUCCESS in 15m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/043b3a42e08d422d80f6ce745acacd9a : SUCCESS in 10m 40s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1aa79c46056c4a8e9487530d27cc779d : SUCCESS in 8m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e8c8936782cc47b8a75908da343d2807 : SUCCESS in 9m 25s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/95cb0c8ac5294c27a3f53063aa206c1f : SUCCESS in 12m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a6a94922b824ef181b018b4c9e45377 : SUCCESS in 29m 46s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0b2d9f675af54ed681a723f7249f114a : SUCCESS in 28m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/783f500e0bca4369a12b1a5d09924f87 : SUCCESS in 32m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5bf8cbafcd247dbaeb5e0c2184ccd88 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":3},{"id":"1ba608f5e424b033ce3041764a7065dcb758fff7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 23:41:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4a752feff637a9dd81440ef88656d25e18356509","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 23:42:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fad41414c9914311b45fd36f8a9ae2eb : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/93017d2f3e08417281aa1df80dbab22f : SUCCESS in 57s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6271d88f9ed5e98f989a6739a75b268537fe0521","revisions":{"0124472b8efb9ecd8d4093a5923f151587cd6216":{"kind":"REWORK","_number":1,"created":"2019-08-23 17:49:14.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/77/678277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/678277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/678277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/678277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/678277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/678277/1"}}},"commit":{"parents":[{"commit":"2cee9f750fd943c44b7db5ffa28c06c6f790dcfb","subject":"Merge \"py3: Finish porting func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2cee9f750fd943c44b7db5ffa28c06c6f790dcfb"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-23 17:14:37.000000000","tz":120},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-23 17:14:37.000000000","tz":120},"subject":"Add func test of changes versionining modes","message":"Add func test of changes versionining modes\n\nUsers are able to change versioning in a container\nfrom X-Versions-Location to X-History-Location, which affects\nhow DELETEs are handled. We have some unit tests that check this\nbehavior, but no functional tests.\n\nThis patch adds a functional test that helps us understand and\ndocument how changing modes affects the handling of DELETE\nrequests.\n\nChange-Id: I5dbe5bdca17e624963cb3a3daba3b240cbb4bec4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0124472b8efb9ecd8d4093a5923f151587cd6216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0124472b8efb9ecd8d4093a5923f151587cd6216"}]},"branch":"refs/heads/master"},"ddcebf413c79a66f2778550786ecd18253fe2a8b":{"kind":"REWORK","_number":2,"created":"2019-09-24 12:23:52.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/77/678277/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/678277/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/678277/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/678277/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/678277/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/678277/2"}}},"commit":{"parents":[{"commit":"2abdd2d70d6ab946e35bd5f32e387d5cfd59da3c","subject":"Merge \"bufferedhttp: ensure query params are properly quoted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2abdd2d70d6ab946e35bd5f32e387d5cfd59da3c"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-23 17:14:37.000000000","tz":120},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-09-24 12:23:35.000000000","tz":120},"subject":"Add func test of changes versionining modes","message":"Add func test of changes versionining modes\n\nUsers are able to change versioning in a container\nfrom X-Versions-Location to X-History-Location, which affects\nhow DELETEs are handled. We have some unit tests that check this\nbehavior, but no functional tests.\n\nThis patch adds a functional test that helps us understand and\ndocument how changing modes affects the handling of DELETE\nrequests.\n\nChange-Id: I5dbe5bdca17e624963cb3a3daba3b240cbb4bec4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddcebf413c79a66f2778550786ecd18253fe2a8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddcebf413c79a66f2778550786ecd18253fe2a8b"}]},"branch":"refs/heads/master"},"6271d88f9ed5e98f989a6739a75b268537fe0521":{"kind":"REWORK","_number":3,"created":"2019-09-24 17:16:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/77/678277/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/678277/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/678277/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/678277/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/678277/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/678277/3"}}},"commit":{"parents":[{"commit":"2abdd2d70d6ab946e35bd5f32e387d5cfd59da3c","subject":"Merge \"bufferedhttp: ensure query params are properly quoted\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2abdd2d70d6ab946e35bd5f32e387d5cfd59da3c"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-08-23 17:14:37.000000000","tz":120},"committer":{"name":"Tim Burke","email":"tim@swiftstack.com","date":"2019-09-24 17:16:48.000000000","tz":0},"subject":"Add func test for changing versionining modes","message":"Add func test for changing versionining modes\n\nUsers are able to change versioning in a container\nfrom X-Versions-Location to X-History-Location, which affects\nhow DELETEs are handled. We have some unit tests that check this\nbehavior, but no functional tests.\n\nThis patch adds a functional test that helps us understand and\ndocument how changing modes affects the handling of DELETE\nrequests.\n\nChange-Id: I5dbe5bdca17e624963cb3a3daba3b240cbb4bec4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6271d88f9ed5e98f989a6739a75b268537fe0521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6271d88f9ed5e98f989a6739a75b268537fe0521"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
