)]}'
{"id":"openstack%2Fswift~678674","triplet_id":"openstack%2Fswift~master~I0156d0059092ed414b296c65fb70fc18533b074a","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I0156d0059092ed414b296c65fb70fc18533b074a","subject":"tests/py3: Improve header casing","status":"MERGED","created":"2019-08-26 20:55:44.000000000","updated":"2019-09-14 00:12:06.000000000","submitted":"2019-09-14 00:09:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678674-1568419796743-4596321f","meta_rev_id":"b29a31d153984a99bbc2fcfe5e644059e2a79825","_number":678674,"virtual_id_number":678674,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-09-14 00:09:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2019-09-13 20:18:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2019-09-13 20:18:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-13 20:18:03.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-09-14 00:09:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"235b50cac0af15652439588d9b2b5a254e008d2a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-26 20:55:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ea10e65a9c693a98339734f91ee4496faf14425","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 21:37:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2b0c48034542fb84cb3b54ee030274 : SUCCESS in 6m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dcfa23ace0264c51898957916c4b6df9 : SUCCESS in 9m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c665f9653e6547be98ab9de649686b93 : FAILURE in 12m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3abb861ce32a4f12bb3be1e265ba5f82 : FAILURE in 11m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/15c654434720440c948d7be3fa886d9f : FAILURE in 12m 25s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/5d1312517dbc4496a6199af1b2b33ae8 : SUCCESS in 10m 05s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/4839818042f34743a83737330ce496ad : SUCCESS in 10m 58s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/7411fdbd758d4e3fabe2b53d725c5038 : SUCCESS in 11m 34s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/640beee285e1456f842f62badb0beb6e : SUCCESS in 11m 01s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/f639a67e1caa4cabad2f16a80a5b64b4 : SUCCESS in 13m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2430164d1f8c4302b1e8a746088660bf : SUCCESS in 10m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ff73b5d7df8a4827ba8934e813a143ba : SUCCESS in 12m 31s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7b7d863712cc418ba58b7cfe9ea22dd9 : SUCCESS in 11m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f35c99e1bb994230b89e1bf29b924252 : SUCCESS in 11m 01s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2b2dbfd80ec04097aa137362facecbd5 : SUCCESS in 13m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/35a7a748ad3b40aebc33506df62c6b4e : SUCCESS in 33m 19s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c15d68f35ecc4e28b429fd7a0ca8dff3 : SUCCESS in 34m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7dda3ed16e4a495398c95903fcd32b9e : SUCCESS in 34m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e84a3d995ee49a09f0669c6c6f92871 : FAILURE in 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3183f07d3f7348278e7279ea83bb8bff : FAILURE in 7m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90ad0259647b4d24ad24aa8235c118a7 : SUCCESS in 18m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b4aadb0cc77feb203a9d3813b12aaad20540567e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-26 21:44:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5572d62a2066660474189d45710408c7dc19753","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 22:32:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60933098296f46e98f327e0136cf734d : SUCCESS in 6m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e4145f229bf4d20afb9a2962b9bc90e : SUCCESS in 10m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ee70b78cc9e2423c8c2b010f5258c754 : SUCCESS in 13m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80ba9f7eab034c79b04fa6be6d34ef47 : SUCCESS in 13m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f1952305a234f9fa49b880bf373dd97 : SUCCESS in 13m 29s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/9887c2173a0740b89deeae14ba251b18 : SUCCESS in 12m 22s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/f9f8986f0549460f82d13a90f957b321 : SUCCESS in 12m 35s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/ad46681d959b48cc88a497874f9766a6 : SUCCESS in 11m 47s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/100e107b95354b10ae6f6dd8751078a2 : SUCCESS in 12m 01s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/83cc8bafb2434a77a28c23f3b27510e7 : SUCCESS in 14m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/90a4fa2912d449e59b9f261a47c5ff1d : SUCCESS in 12m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/883dd72606454b7380f7349f262f51a9 : SUCCESS in 11m 11s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5ba7b5ebb927460a9f1d2613db15d518 : SUCCESS in 11m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/712290353df04d03bab4dd3b11005d5d : SUCCESS in 12m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/445e86068cd24450b46f25be97682c77 : SUCCESS in 14m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25a0e9d7b7984a1882d036a8cd0e15b1 : SUCCESS in 30m 03s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c73587cdf6be42e8965869b030cd97a3 : SUCCESS in 30m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e317adc6ef9492b81ac81ada65f9149 : SUCCESS in 27m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2dcd311ce244d7bb489c7fdc4706f93 : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76b9443be9844585848d58c1b61e5f4b : SUCCESS in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f55a3f0df094048b7862c03427a714f : SUCCESS in 21m 58s","accounts_in_message":[],"_revision_number":2},{"id":"9241ee965bed65c07cf28562f72167ed5fc2a595","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-09-13 20:18:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nseems to better match what we do in wsgi.run_server :shipit:","accounts_in_message":[],"_revision_number":2},{"id":"9b175b92ccd0932fea58c00cedfce29296b8b305","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 20:18:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"454127d06d4684858786fc4dbcce69789819adda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 00:09:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c925ce5e55549e90a3d7c2ce7043238aefea10e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 00:09:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69f8cf673079436092854de017eaf301 : SUCCESS in 5m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f3a427e0fbc24d7d922f4ad046b32116 : SUCCESS in 10m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a50cdd48b5be4bda810636c78eb5e18b : SUCCESS in 13m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b4b231a965443109bfce2f94dbda1ab : SUCCESS in 18m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3f2b5cc58bcd412597005dbdfe7f0014 : SUCCESS in 12m 50s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/dcc9c76402524f408cb822eeb09283a0 : SUCCESS in 12m 02s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/b5ac2fdffae440df9c9c887de68d0dd2 : SUCCESS in 10m 31s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/489c20abf1044e1398fb5698ebb64836 : SUCCESS in 8m 55s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/6a4da793848a4e1aaa0fb660e3ccdb3f : SUCCESS in 9m 59s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e736e624dc33473780de9c83f5528b85 : SUCCESS in 11m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1bd1d316e4a44269963b1db84cb8c2b4 : SUCCESS in 8m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ad1bf0fea67741159ea2525991c2304a : SUCCESS in 9m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5bd86126ce8d43fb8a19fec59e01fd39 : SUCCESS in 13m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3dbeae44642b43038cd6758cd5209be5 : SUCCESS in 14m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2b2c2a3518640a698eaed1a10069442 : SUCCESS in 26m 59s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d34911942e5743b7b27e381c70d53150 : SUCCESS in 29m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/91f6bd2a689b4fdbaf962a416d546590 : SUCCESS in 26m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20c6f3781f2b4caf8e54319a278918b3 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d8a4bc5311e4eb58991ba997357d517 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b29a31d153984a99bbc2fcfe5e644059e2a79825","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 00:12:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de89237310c409bbea595bb276016db : SUCCESS in 57s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/efc3d36a15594945a11ea55c4787c723 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d83b9b95e32038390dbdc66d93c36c929dbce2a","revisions":{"5b4b5cf0d80e6569799d5b973730b36aad23e4ef":{"kind":"REWORK","_number":1,"created":"2019-08-26 20:55:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/74/678674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/678674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/678674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/678674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/678674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/678674/1"}}},"commit":{"parents":[{"commit":"f70520239c7fa2d22e4b9ee604bfaf7123b39095","subject":"Merge \"Fix symlink docstring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f70520239c7fa2d22e4b9ee604bfaf7123b39095"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-15 21:33:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-26 20:55:26.000000000","tz":-420},"subject":"tests/py3: Improve header casing","message":"tests/py3: Improve header casing\n\nPreviously, our unit tests with socket servers would let eventlet\ncapitalize headers on the way out, which\n\n- isn\u0027t something we want to have eventlet do, because it\n- breaks unicode-in-header-names on py3, so it\n- is already disabled in swift.common.wsgi.run_server() for real servers.\n\nInclude a test to make sure we don\u0027t forget about it in the future.\n\nChange-Id: I0156d0059092ed414b296c65fb70fc18533b074a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b4b5cf0d80e6569799d5b973730b36aad23e4ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b4b5cf0d80e6569799d5b973730b36aad23e4ef"}]},"branch":"refs/heads/master"},"4d83b9b95e32038390dbdc66d93c36c929dbce2a":{"kind":"REWORK","_number":2,"created":"2019-08-26 21:44:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/74/678674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/678674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/678674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/678674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/678674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/678674/2"}}},"commit":{"parents":[{"commit":"2545372055922abd681ef665f9040590d2f5806c","subject":"py3: Switch swift-dsvm-functional-py3 to run tests under py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2545372055922abd681ef665f9040590d2f5806c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-15 21:33:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-26 21:44:05.000000000","tz":-420},"subject":"tests/py3: Improve header casing","message":"tests/py3: Improve header casing\n\nPreviously, our unit tests with socket servers would let eventlet\ncapitalize headers on the way out, which\n\n- isn\u0027t something we want to have eventlet do, because it\n- breaks unicode-in-header-names on py3, so it\n- is already disabled in swift.common.wsgi.run_server() for real servers.\n\nInclude a test to make sure we don\u0027t forget about it in the future.\n\nChange-Id: I0156d0059092ed414b296c65fb70fc18533b074a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d83b9b95e32038390dbdc66d93c36c929dbce2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d83b9b95e32038390dbdc66d93c36c929dbce2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
