)]}'
{"id":"openstack%2Fswift~678962","triplet_id":"openstack%2Fswift~master~I37a085886afc5e28e8e12532e1cbc62acde73570","project":"openstack/swift","branch":"master","topic":"bug/1840322","hashtags":[],"change_id":"I37a085886afc5e28e8e12532e1cbc62acde73570","subject":"WIP: Add two more versioning modes that feature a new naming scheme","status":"ABANDONED","created":"2019-08-27 21:31:25.000000000","updated":"2020-01-16 20:44:29.000000000","total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e33db84350b851d892b5ba5f36b51b6dc16c9364","_number":678962,"virtual_id_number":678962,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":-1,"date":"2019-09-11 06:15:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 06:15:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c871b38dfd23ae4eaf3f6bdf054984077d0b8ff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-27 21:31:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc318ece1f6645ea657c54c364f158adb80cb324","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 00:44:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd1c006adaa5404da13c1f08839fa152 : SUCCESS in 5m 23s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/31293d577f414ae29591b8526b0b0e01 : SUCCESS in 1h 02m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/050c730c9e884e89963e7aa3b2d38459 : SUCCESS in 50m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2bb417a9a00347a7900610cdeb5ccac9 : SUCCESS in 11m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d7fe6a5202ca45f5922e30ee9cbf0a61 : SUCCESS in 14m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/48062feeac044be58e5f8165dd5abfc7 : SUCCESS in 15m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cceb0cb5ec724806ba30d2754de5d118 : SUCCESS in 11m 14s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/1af09ebddbd745f1bf202d5a04b9aa34 : SUCCESS in 10m 13s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/57c51e603fea49e786dce2ba9d33548e : SUCCESS in 11m 53s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/382cb7b2278a4f02b7a19b92d274bd0a : SUCCESS in 11m 38s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/a2ccda6a17ba4eaa92b8d84e15ddbcf3 : SUCCESS in 11m 38s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/ef27345406854b0d99ad2dc9acb28992 : FAILURE in 15m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c1e31b98bbd24d409f63579307a26f7f : SUCCESS in 11m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c46b9554f9f1485cb71d2d5f6dd8861c : SUCCESS in 12m 30s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1e1206d1c3754232a0afe2845c50e107 : SUCCESS in 12m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bea7126600d547d0ace023bd839ec048 : SUCCESS in 10m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/3c22dfc35ce7465da333745ba52a9588 : FAILURE in 15m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f0f1882324ed41ff974822337286a9be : SUCCESS in 14m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f02a3cfa567343ec8937537c21aff1a5 : SUCCESS in 38m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e64933e5de7b42fb84e80e0ee362ee3e : SUCCESS in 28m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/92199ac9d0a74768a073d78afe487337 : SUCCESS in 30m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e3d503673e54e258f11919b273fe637 : SUCCESS in 29m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2367f626250b4005a7628dd2889ac978 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96ac6fecbde64eaaa145204de7d8973a : SUCCESS in 7m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/696e824f3c854462b1fda84b5a7d7019 : SUCCESS in 21m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f3ad0419a2b1bcfa81239f20fada00e625355d87","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-30 20:54:38.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cd4403b19ad7536bbd993e626d9884fd7dee3042","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-04 23:36:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f8b6c1ed5063c61ff286357372f37c798432b77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 20:42:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88f744fcc3d044d78b9d3a5598268a9d : SUCCESS in 5m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bf6a6d1c0f5d4b47af37470fa3df92cf : SUCCESS in 1h 00m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e14a277e40d9441fb84f7f3090c4419a : SUCCESS in 55m 52s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a15e906fc66e4d0ab249fed92af5efe6 : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/715bc33478724160b349fa3e9a1aa0e0 : FAILURE in 12m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/79f2b3e7227541e689a1d17d15c76e74 : FAILURE in 13m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9e19b6f4bc9a4cabb3c72af7738d9aa2 : FAILURE in 14m 46s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/05a55c31487647cfbb28fcb7d2088ac3 : FAILURE in 12m 02s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/29d37d11068f4b8f85ab17e3378c478a : FAILURE in 13m 30s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/33ca09ca6b304581b9a49fcd6b21b53d : FAILURE in 13m 01s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/7f881a6727a74e65a78664f4b07bdb14 : FAILURE in 14m 32s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/1fba32d6239546f8a330676a66338ef2 : FAILURE in 15m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f6b1294dca3b4e32838fd26d50eeee95 : SUCCESS in 13m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/da4a80171e16453887bff68e88ac9e46 : SUCCESS in 11m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b04efba0c6fc44baa4873d60c689f345 : SUCCESS in 13m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e37ae1577e7249449297255b3a96a466 : SUCCESS in 15m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2e53330325a64033807a275a1080f3fe : SUCCESS in 18m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9b135ca6ed4b4a82bcf7355e9e359875 : SUCCESS in 16m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f35572ba9ec5422bb346bb4d1d2c63f3 : SUCCESS in 32m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed39201790084505a43eb812c73388ff : FAILURE in 31m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/f40ba060a9054f67a3a9b881ae4ca573 : SUCCESS in 32m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/855cb87d128242f9b76c3caa08bcfd33 : FAILURE in 27m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc52a6e958f947729d7e384b39ad68ff : FAILURE in 15m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1948552687d6475fb6b3e33bddf0facd : FAILURE in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55a1b12d9704481f92b3f4c5efa01c88 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":2},{"id":"5802c520d040e2097690458394cda91dedbd9209","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-05 22:39:52.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c7448d83ae2848332298b9b1f32420e8d0ddc258","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-11 05:09:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"af6ca7c3512e0fdbc0dffe1694dc41da91db2569","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 06:15:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5f7927dd0040ac85d407662fe8ebc1 : SUCCESS in 5m 56s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/21711676e926466292fa99b6f4effc7a : SUCCESS in 1h 01m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6a07f9b13831498ca1349791a4b48de9 : SUCCESS in 55m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d641031b6dca4e2f8f298a16c729907a : SUCCESS in 12m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27273c781fa64082929ab71d6c68d640 : FAILURE in 10m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/89a1488a7a9f4b18a2e3522d68e5bec3 : FAILURE in 12m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d6309dd362c246d892e2c9eb3601b6f0 : FAILURE in 12m 38s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/5deffe188f304c9c8d4d7de6395c42ef : FAILURE in 21m 57s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/2e433815fe504c7aa8f77faada6da1dd : FAILURE in 21m 40s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/8b0d7087ac69499eaf961da1861ae6a6 : FAILURE in 9m 42s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/d3ac388fbfdf42b38421b49d1eaaea93 : FAILURE in 10m 54s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/5d9f32b34b83414c81c8558817550841 : FAILURE in 13m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7a20b06feb274a4caea86b03bb5b6108 : SUCCESS in 11m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c9215768f4f545078bd3a2de2ea8148a : SUCCESS in 13m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e5f9d6b32a024b2ebaaaf5a5276f747a : SUCCESS in 22m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/17d9cf5e07fa4f95a9a446aad9fee1bf : SUCCESS in 14m 15s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5c89bfd2d7a8464cb026e710354a3a5a : SUCCESS in 17m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f911ffd2bbd465580dab98a54e3ea29 : SUCCESS in 25m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/55c71239c7ac4eac80e1eaa1b545d336 : FAILURE in 31m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f0feea99edfb483ebbfb015b4a710cbd : FAILURE in 30m 14s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/405c10f430b1465a8f2230510b9e3799 : SUCCESS in 26m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/866a426b54974f4698aa625bc3e78500 : FAILURE in 24m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0c2fdf68d5c408a96542e7db4b7987c : FAILURE in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79744f1c6e1844359bde243cb9cef18d : FAILURE in 7m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bc2cc7101e3c461785f2ba6f03039255 : SUCCESS in 18m 24s","accounts_in_message":[],"_revision_number":3},{"id":"d9543fa03298fdbf5949f130792c40aca4d0b5ff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-16 20:44:29.000000000","message":"Abandoned\n\nAbandoned in favor of https://review.opendev.org/#/c/682382/","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8fe1eecadc06ded046d1066a2644e125949bdbda","revisions":{"9dc1950b0210391b075ed056c8c20396777bc151":{"kind":"REWORK","_number":1,"created":"2019-08-27 21:31:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/678962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/678962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/678962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/678962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/678962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/678962/1"}}},"commit":{"parents":[{"commit":"77ebb1754dbb521fc2b7cfe024d30c909090c8ac","subject":"symlink-backed versioned_writes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77ebb1754dbb521fc2b7cfe024d30c909090c8ac"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-27 18:26:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-27 21:19:39.000000000","tz":-420},"subject":"WIP: Add another versioning mode with a new naming scheme","message":"WIP: Add another versioning mode with a new naming scheme\n\nThis lets us have an index on (name, timestamp DESC) instead of the\nless-useful (len(name), name, timestamp) index we had before.\n\nAlso add some new APIs specific to this new versioning scheme:\n\n  GET    /v1/a/c?versions[\u0026marker\u003do][\u0026version_marker\u003dv]\n  GET    /v1/a/c/o?version\u003dv\n  HEAD   /v1/a/c/o?version\u003dv\n  POST   /v1/a/c/o?version\u003dv\n  DELETE /v1/a/c/o?version\u003dv\n\nThe idea is to more-closely mirror S3\u0027s feature set. I *think* COPY\nshould Just Work?\n\nStill need to handle DELETE of the current version, though. And tests;\nso many tests.\n\nChange-Id: I37a085886afc5e28e8e12532e1cbc62acde73570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dc1950b0210391b075ed056c8c20396777bc151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9dc1950b0210391b075ed056c8c20396777bc151"}]},"branch":"refs/heads/master"},"1b1913af93b200de1998d6bc86450ec59e3c4aa6":{"kind":"REWORK","_number":2,"created":"2019-09-04 23:36:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/678962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/678962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/678962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/678962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/678962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/678962/2"}}},"commit":{"parents":[{"commit":"77ebb1754dbb521fc2b7cfe024d30c909090c8ac","subject":"symlink-backed versioned_writes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/77ebb1754dbb521fc2b7cfe024d30c909090c8ac"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-27 18:26:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-04 23:33:53.000000000","tz":-420},"subject":"WIP: Add another versioning mode with a new naming scheme","message":"WIP: Add another versioning mode with a new naming scheme\n\nThis lets us have an index on (name, timestamp DESC) instead of the\nless-useful (len(name), name, timestamp) index we had before.\n\nAlso add some new APIs specific to this new versioning scheme:\n\n  GET    /v1/a/c?versions[\u0026marker\u003do][\u0026version_marker\u003dv]\n  GET    /v1/a/c/o?version\u003dv\n  HEAD   /v1/a/c/o?version\u003dv\n  POST   /v1/a/c/o?version\u003dv\n  DELETE /v1/a/c/o?version\u003dv\n  COPY   /v1/a/c/o?version\u003dv -H \"Destination: ...\"\n\nThe idea is to more-closely mirror S3\u0027s feature set.\n\nStill needs more tests (particularly unit), but at least there are some\nfunc tests demonstrating the new API.\n\nChange-Id: I37a085886afc5e28e8e12532e1cbc62acde73570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b1913af93b200de1998d6bc86450ec59e3c4aa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b1913af93b200de1998d6bc86450ec59e3c4aa6"}]},"branch":"refs/heads/master"},"8fe1eecadc06ded046d1066a2644e125949bdbda":{"kind":"REWORK","_number":3,"created":"2019-09-11 05:09:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/678962/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/678962/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/678962/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/678962/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/678962/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/678962/3"}}},"commit":{"parents":[{"commit":"c5cab23ee349d6214b83d5b0c726c1fc4359aed8","subject":"symlink-backed versioned_writes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5cab23ee349d6214b83d5b0c726c1fc4359aed8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-27 18:26:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-11 03:37:40.000000000","tz":-420},"subject":"WIP: Add two more versioning modes that feature a new naming scheme","message":"WIP: Add two more versioning modes that feature a new naming scheme\n\nThe new namin scheme lets us have an index on (name, timestamp DESC)\ninstead of the less-useful (len(name), name, timestamp) index we had\nbefore.\n\nOne mode acts like history-mode but with the new naming scheme. The\nother is comparable to disabling versioning, but tracks a location where\nversions may previously have been stored.\n\nAlso add some new APIs specific to this new versioning scheme:\n\n  GET    /v1/a/c?versions[\u0026marker\u003do][\u0026version_marker\u003dv]\n  GET    /v1/a/c/o?version\u003dv\n  HEAD   /v1/a/c/o?version\u003dv\n  POST   /v1/a/c/o?version\u003dv\n  DELETE /v1/a/c/o?version\u003dv\n  COPY   /v1/a/c/o?version\u003dv -H \"Destination: ...\"\n\nThe idea is to more-closely mirror S3\u0027s feature set.\n\nStill needs more tests (particularly unit), but at least there are some\nfunc tests demonstrating the new API.\n\nChange-Id: I37a085886afc5e28e8e12532e1cbc62acde73570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fe1eecadc06ded046d1066a2644e125949bdbda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8fe1eecadc06ded046d1066a2644e125949bdbda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
