)]}'
{"id":"openstack%2Fswift~679531","triplet_id":"openstack%2Fswift~master~Ia9a0ac28bde4b5dcbf6e979c131e61297577c120","project":"openstack/swift","branch":"master","topic":"bug/1805539","hashtags":[],"change_id":"Ia9a0ac28bde4b5dcbf6e979c131e61297577c120","subject":"diskfile: Add some argument validation","status":"MERGED","created":"2019-08-31 04:50:20.000000000","updated":"2019-09-05 05:45:35.000000000","submitted":"2019-09-05 05:41:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679531-1567662108498-0552f13f","meta_rev_id":"301941aba0b2581f3afd3606bc4b09a697bac4da","_number":679531,"virtual_id_number":679531,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":2,"date":"2019-09-05 05:41:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-04 17:09:42.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":2,"date":"2019-09-04 11:52:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"all":[{"value":1,"date":"2019-09-04 11:52:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-04 11:52:10.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-09-05 05:41:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc33ef194accd1a5e5c5be5745f8322bac4143bc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-08-31 04:50:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35485a3736ce3d875b8827da060bdd068ed0abae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 05:48:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/751905d64aa54994baa93234d54fb8ad : SUCCESS in 4m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8efe00bba04a4cc8911a1b3ca21719a2 : SUCCESS in 55m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eca969bcf7d84b47b605d950bcf6b203 : SUCCESS in 57m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a480563afb94221aeecbc418ff78e02 : SUCCESS in 9m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ca34e377d2c400d9d01b4fb72a6cf0b : SUCCESS in 11m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c9d843d976d4c43957d42363f671b26 : SUCCESS in 11m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/04a595d375dd457894867ee0d7be1c10 : SUCCESS in 11m 57s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/8d7285f47ebf4cb2ab60239604af0745 : SUCCESS in 10m 19s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/a510f987f02d45b7b1c467d80fb743ee : SUCCESS in 12m 14s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/7969d972fc7b44dd93c4b9df1a757fa6 : SUCCESS in 9m 15s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/884f565e42ec4026ba8df71e7c05db7a : SUCCESS in 10m 13s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/48db4d82a5ab49ada2e56e44b99f95bd : SUCCESS in 12m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/776c4a2b5c5d4e8faeecb247b4a37927 : SUCCESS in 11m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b1fe7da1ce634dc3b156a7e89716337a : SUCCESS in 10m 13s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3e0cbfb388d84c7a9874761ccf2e46b1 : SUCCESS in 9m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2a090d8ec2e7437b9ccc933d90841cba : SUCCESS in 10m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/af07e6ca68544300bc7e142169512a64 : SUCCESS in 12m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5aeb5c6eb7ac4950bf3fcb3f35b7d3a9 : SUCCESS in 16m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c2132a36e1e94f37ab37c5cf2d96b2cf : SUCCESS in 30m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ffe7c12546845068a23f8dd7a942b34 : SUCCESS in 27m 14s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a08d44856ef648758c2d07e320bc78d9 : SUCCESS in 29m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ac387282bd55443a87b0d3d8e11ce3c6 : SUCCESS in 29m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3965e14a89fe458d89a65fce4926c842 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b1d675220ea41ca8ac9ad40d7dffb05 : SUCCESS in 6m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d6740555e23434bb46f80d5d6b3b6ad : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7e283b876ed8d280aa62fad736aae6948f41a7de","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-09-04 11:52:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"90efeec57b94d5fffef3ea45daeb10426891428b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:42:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0a5e6b97958b3b4b6e1476f0c510d0653a700d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 15:58:22.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"49e39cda767e7ea1212b202734bd897dea802640","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-04 17:09:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"66f27550a646b3cc64b9306501c48eb36ba45cd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:11:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8fa452c8ccf4d05a2b1eb5c68fbccaa : SUCCESS in 5m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/819ea6eeba404a4583a301aac6be3624 : SUCCESS in 1h 00m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/027d56412ef145f5b5886864df3e39ee : SUCCESS in 58m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d9667a6d04948639ab978383dd40854 : SUCCESS in 10m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b90d798fb44b4693ba8d56a6caca6310 : SUCCESS in 12m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6f5cb6e498a43a181f8b1a86cc0cb4a : SUCCESS in 12m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/befc03e26bef496dbc6fc617a196c51e : SUCCESS in 12m 48s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/5259981f17e043759e6ee0b21e91a5a7 : SUCCESS in 10m 03s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/dee05263826e43888823e1ec72d7034b : SUCCESS in 10m 13s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/075b5c71575c4033ba1e0d86c014972a : SUCCESS in 10m 01s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/03e42704b1314cafbcaf269c6b7a57b6 : SUCCESS in 11m 03s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e6727cd4f84d4678b62366b431a3b0c4 : SUCCESS in 12m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/008490ecd3a642939d40f068c8cfe714 : SUCCESS in 10m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d5a80efbb16d4bdc925ff5aa496858c9 : SUCCESS in 11m 26s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ee3b8f288e7c4ea2b08c89b9775dbdef : SUCCESS in 9m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0ad846b0f85e40b5aeab1fcebcd9c609 : SUCCESS in 11m 23s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/87b78fe1764b4b998606dd75552bd37e : SUCCESS in 12m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/91d1dadf24354a21b80d1332956d3455 : SUCCESS in 16m 24s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fde2995b115e49e3834e706c0552edba : SUCCESS in 34m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d7b3139233074b469aa04246194c4c5b : SUCCESS in 33m 23s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/88200ae73441489580db448106e21d8b : SUCCESS in 31m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/58219da204c44fcb85302af00162ec71 : SUCCESS in 30m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8347909b85e40a9b6f4c56759dcc200 : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a33da9b9e84c23958902165d763664 : SUCCESS in 7m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2666291c53bd4362a9ccf13ab6ad4981 : SUCCESS in 18m 34s","accounts_in_message":[],"_revision_number":1},{"id":"961c61b6a3e97dcf31ba46cb0c3d764fc7542baf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 02:57:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47d447dd62227d40facc4cecf810e1aafc9a740a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 05:41:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba404a00dcea4842847e9f5e4fa8931b : SUCCESS in 6m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3725fa8b82dd488c889f364f8fc633be : SUCCESS in 1h 02m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/14dd707e01a64203989c828e49beca3a : SUCCESS in 59m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/109237c4be20464d81df1b07f6337c3d : SUCCESS in 10m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3562069fdb8542e3bdb30516427e04c9 : SUCCESS in 12m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e8e576f0118346fe8486e898d46744f9 : SUCCESS in 13m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e3c60a4508bb49879155cc99419c6c50 : SUCCESS in 11m 54s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/893a73d72b774e729ac26d49790a799a : SUCCESS in 10m 23s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/53a734cc4fd140b8b9d0243be4e1d6e4 : SUCCESS in 10m 44s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/78b9dd291bb14c5c9818498dcdfd156f : SUCCESS in 11m 20s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/aa14433323b94b37abe8e953b71c36fa : SUCCESS in 11m 29s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/4a63c65dc94540e899576b5c3a875cc5 : SUCCESS in 13m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/513dfa5b495f476d94022bc88ac4c347 : SUCCESS in 10m 45s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b1ef45f2c80e49788e867623b9ddbad6 : SUCCESS in 10m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2c861c0f7e6c4dc0be4b45918f6c3c40 : SUCCESS in 10m 51s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7b85d883ebc84a7d98eba777b5881c0d : SUCCESS in 13m 20s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/62c9cda802254336b1ea4ca8a6f7197c : SUCCESS in 32m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2cc6d2337bc34df4a03ba09adeb56c2c : SUCCESS in 29m 22s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/821aa22a865043d5954ce46f84205d41 : SUCCESS in 32m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c68c27ca99584efa84ce57ee63e46138 : SUCCESS in 30m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09c81d0be31c48f4816169bc8b402deb : SUCCESS in 13m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e52dcb60e9bb4032a397cc8fe2f53ea5 : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9cd9aad031f2b5d5f9bc5d14ded8ff29a149274b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 05:41:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cda27998a443a7bba4b196113f0a0649fea205b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 05:45:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385562495d8b480f92586c2d1586d37f : SUCCESS in 1m 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b07012163dab440d8e4c25cdee6ae716 : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c71bb2506310438b011818a44449daea500863fd","revisions":{"c71bb2506310438b011818a44449daea500863fd":{"kind":"REWORK","_number":1,"created":"2019-08-31 04:50:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/679531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/679531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/679531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/679531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/679531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/679531/1"}}},"commit":{"parents":[{"commit":"eed76d8bed446518bff2ca4af18259f7637c430e","subject":"Fix a potential bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eed76d8bed446518bff2ca4af18259f7637c430e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-31 04:40:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-08-31 04:49:48.000000000","tz":-420},"subject":"diskfile: Add some argument validation","message":"diskfile: Add some argument validation\n\nEither all or none of account, container, and object should be provided.\nIf we get some but not all, that\u0027s indicating some kind of a coding bug;\nthere\u0027s a chance it may be benign, but it seems safer to fail early and\nloudly.\n\nChange-Id: Ia9a0ac28bde4b5dcbf6e979c131e61297577c120\nRelated-Change: Ic2e29474505426dea77e178bf94d891f150d851b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c71bb2506310438b011818a44449daea500863fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c71bb2506310438b011818a44449daea500863fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
