)]}'
{"id":"openstack%2Fswift~681656","triplet_id":"openstack%2Fswift~master~I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc","project":"openstack/swift","branch":"master","topic":"bug/1840322","hashtags":[],"change_id":"I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc","subject":"versioned_writes: checks for SLO object before copy","status":"MERGED","created":"2019-09-12 07:39:56.000000000","updated":"2019-09-18 21:12:41.000000000","submitted":"2019-09-18 21:10:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681656-1568841055338-08dddfb7","meta_rev_id":"484be586d275a9ab478a3800ef3a685bf3bfbc32","_number":681656,"virtual_id_number":681656,"owner":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},{"value":2,"date":"2019-09-18 21:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"all":[{"value":1,"date":"2019-09-15 16:14:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-18 18:00:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-18 18:00:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 18:00:26.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-09-18 21:10:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bf456b81a6957dd53417dca54f5cfcd8d7fa3ac6","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-12 07:39:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1357eafc581fb49fad4b33fa1f86cf8ffc4ee5c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 13:18:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbaf53a024474419b70b36123aaa2b3c : FAILURE in 3m 43s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9084c9c14bda4517ac180bb4de249e26 : FAILURE in 1h 06m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de5c323acd74484ea98acda10b951ed2 : FAILURE in 13m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/62c8aca11d5a412fbf7f4ef1be6fee98 : SUCCESS in 8m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a680623458848f1a3de3b77997808fd : FAILURE in 10m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26e70607699f422cb33f9c4717f836c6 : FAILURE in 8m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ef54f6e787bf418bbd85329359227f06 : FAILURE in 9m 03s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/bbdf9278a39240ec9b7f54fe45707ec4 : FAILURE in 4m 00s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/381b29e42ad4453490e4c1b64d299f2d : FAILURE in 3m 30s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/c69abe402d384e3ba2b3d8d65c50e9e3 : FAILURE in 4m 25s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/890d8836f9f84a119220d58a46b140a9 : FAILURE in 3m 52s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/4ae15574b4b54eb3afc8b97fcb9ebc8c : FAILURE in 3m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b18dbb627be8432b8ecd9a17031e1907 : FAILURE in 3m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1dbbc18fcf7045ec8bc1e747c0ad864b : FAILURE in 3m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ea522024aac4422faf5ce3f2d2194170 : FAILURE in 3m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/79b8b6e8553a4a818aa208579bfd8028 : FAILURE in 6m 03s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f953289eea1348c088668aa2c0d3565e : FAILURE in 4m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/679c7ac1f2194557908bcb9dd41aa022 : POST_FAILURE in 6m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f93fb60213b14a41afa9bd41f53a15f4 : FAILURE in 58m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c66a15caf11344099ad386565e38295c : RETRY_LIMIT in 25m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/2679f04334ee43c88a0dafbf432de9bf : RETRY_LIMIT in 18m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d2b7c1c036bf40988ad51e5d88a65673 : RETRY_LIMIT in 21m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5d8ebed6ae9470dbbf290618b886751 : FAILURE in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eed42c894e7e4fbc9c154435dcba1ad7 : FAILURE in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a718b926dc34239ba6f80598811e618 : SUCCESS in 17m 16s","accounts_in_message":[],"_revision_number":1},{"id":"c1665c258d61a98562d62184c5c63058640b6ae2","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-12 15:36:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"473a71185605e27ce1d2cbee525f70d6b925ec4e","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-12 15:42:28.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"151edc221cff8939692afdb9e2ce0c3faf4777c1","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-12 15:45:15.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"17d781c95672c542258d58a2db7cd6e26201a593","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 02:49:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849f930b8f804319a27abd0caca00d53 : SUCCESS in 5m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8286ab713ba74542a51e4364490306c9 : SUCCESS in 1h 03m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/55d623afd1454c8faf54227688cc4efe : SUCCESS in 1h 00m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/66e5f7b5ac1441d98a6ac9274df0a7ce : SUCCESS in 9m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/089f8a7107aa4ddebb6b9ea65d942eae : SUCCESS in 11m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/488fa8b735f8486ba41bdd86cac45cf3 : SUCCESS in 11m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad90434ecc2e4815b19210a7e120256a : SUCCESS in 10m 29s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/fed401c0c510417cab9b777c1f61dae4 : SUCCESS in 8m 38s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/e64c1d64ce6d4ec8bc935d6756361596 : SUCCESS in 9m 42s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/b95fe54b546f4127aae4ecae7e6b26a1 : SUCCESS in 8m 32s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/982417f7e520404d845e168c4fb06ad7 : SUCCESS in 10m 01s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/c969d8163982404c973a8c5ac5ce0ce9 : SUCCESS in 12m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cc4b7081e2f84fe6b410031f8eb5235c : SUCCESS in 8m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d06284d907594eac9aead62f3066b873 : SUCCESS in 10m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2cd5530725894bafad79edf7e5eeb549 : SUCCESS in 11m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8eb0b0fcc1734545a077def20df0f44a : SUCCESS in 12m 05s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/57a104eff1914e92a82f9a1701b5c664 : SUCCESS in 14m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1a122d26bfc462dbcaa98c1689a509b : SUCCESS in 23m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7f0c957a80e841d5b0d8180f6e76b310 : SUCCESS in 40m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/020d56f44d6544a8889a4fd21f054cc9 : SUCCESS in 41m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d5599d54328c46a791de351eeb9eb62a : SUCCESS in 28m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e983aa138faa4359a0dfa6f92ea7fe74 : SUCCESS in 31m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8006e34380c43b885397bb2c7161d2f : SUCCESS in 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2cf63f356b34c6fbfd4fd35a10ebf6e : FAILURE in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d6c22d68ed54478b5a2f6c9cfeec20a : SUCCESS in 16m 39s","accounts_in_message":[],"_revision_number":4},{"id":"2cacaabf8f4f431f1b1d7e4892f0e3e89fc89830","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-13 04:50:09.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"c363afe0ee25a8e4b9457e633a7d3cde2e543649","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 10:23:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7789148176bf49bcb4963b7f7d4c986b : SUCCESS in 6m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/108c73a71dc54dd2b7ba32b07f131863 : SUCCESS in 1h 00m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cbf0b413bf704346a17d117f20db769c : SUCCESS in 53m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d705067a52fa4513934727a53d60209f : SUCCESS in 9m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d8d1079275249669e3776b390776b6e : SUCCESS in 12m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab3f586eb6c847deaa78fa79dcaa336c : SUCCESS in 13m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c697649471948ec9bdf781ac3038892 : SUCCESS in 12m 18s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/185dda0de359496e92821a12c8cd9e51 : SUCCESS in 10m 42s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/468917aa17e046c6baf8d15d24529212 : SUCCESS in 10m 34s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/c9696f03f47546e4b28ccf9065c43d7c : SUCCESS in 9m 55s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/f49b6bb064d84ad4a8f083a07284de30 : SUCCESS in 11m 18s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/2ab906b042194085a9a2fb5b651e9e01 : SUCCESS in 13m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e5baa635fe5c4a55908367ba14df9d62 : SUCCESS in 10m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/67e1dc550400488faf8132ccaeef26b5 : SUCCESS in 10m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/df5e7ac0b54a45c2866b03cf25ac29af : SUCCESS in 10m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/221afdf0dec548dda3dd971bf0239c89 : SUCCESS in 12m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4a08ad7f58fb40739ec6ddb810f66bed : SUCCESS in 11m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ac54b57b87841d3912ebf12ee8328a5 : SUCCESS in 13m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/34b147b16c944fbdb4b948689b0ce3de : SUCCESS in 31m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa44d7e5e2f04c6496f7f756b4a9cf2a : SUCCESS in 34m 09s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ad5617385c204789a976401faa3b4532 : SUCCESS in 34m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/47864851ebef452e90aab25d00921743 : SUCCESS in 31m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af0aac92be814609afbca8615f34c19c : SUCCESS in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcef979db8cd43f08def5e60ca558987 : SUCCESS in 5m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0ad2b8fc71d40ea8788f62e4f486000 : SUCCESS in 15m 20s","accounts_in_message":[],"_revision_number":5},{"id":"2aba5a15daff711c5ba8cf5d49e4688fe1013c9e","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-13 11:33:40.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0310089d5d951825555963e35a59bc260687b547","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-13 20:59:00.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nThanks for looking at this issue! Unfortunately, I think it\u0027s going to get a little complicated, but I\u0027m happy to help you through it ;-)\n\nOur recommended pipeline placement for versioned_writes and slo (namely, that versioned_writes should be right of slo -- closer to the proxy-server app that\u0027s going to be making backend requests) means that the ?multipart-manifest\u003dput query param will be ignored -- versioned_writes won\u0027t call into slo, and nothing else downstream of it knows what to do with that param. This is actually a good thing:\n\n* versioned_writes doesn\u0027t really want to re-validate that the segments when moving the manifest; it\u0027s enough to know that *at user-upload time* those segments were available\n* slo won\u0027t parse it\u0027s own on-disk format when coming from a client [1] (the wisdom of which is debatable IMO :-/)\n\nSince we can\u0027t punt to slo to fix the listing, I guess we\u0027ll have to do it ourselves. :-) This is where things start to get fairly weird. To get the total size into the container listing, slo shoves it into the Content-Type [2], relying on the container-server [3] to unpack it. This has some downsides (like needing to carry it forward on POSTs [4], and not having the *manifest* size available at all), but that\u0027s what we did first, so it\u0027s stuck. If we were implementing it as a new feature today, it\u0027d almost certainly work differently.\n\nThe next question, of course, is how to get the *value* that we should be dropping in! Fortunately, recent SLOs record that in sysmeta [5] which will only be set by slo -- so it can double as our flag to see whether the object we\u0027re copying is an slo manifest! I suppose that won\u0027t work for SLOs written before Swift 2.12 or so, but OTOH,\n\n* this bug has been around a while, so anyone using versioned_writes and slo together like this must either already know to not trust their listings or just not care,\n* I don\u0027t really think that versioned_writes should be in the business of parsing SLO manifests to determine the large object\u0027s total size, and besides\n* we\u0027re certainly no *worse off* than we were before.\n\nWhat do you think about adding some assertions to a functional test to verify the new behavior? FWIW, we\u0027ve already got a test that verifies SLOs can be versioned and restored [6] -- I think it\u0027d mostly amount to making some assertions on sizes returned from\n\n self.versions_container.files(parms\u003d{\u0027format\u0027: \u0027json\u0027})\n\nand\n\n self.container.files(parms\u003d{\u0027format\u0027: \u0027json\u0027})\n\n[1] part of why the copy middleware does this: https://github.com/openstack/swift/blob/2.22.0/swift/common/middleware/copy.py#L292-L296\n[2] https://github.com/openstack/swift/blob/2.22.0/swift/common/middleware/slo.py#L1336-L1337\n[3] https://github.com/openstack/swift/blob/2.22.0/swift/container/server.py#L561-L564\n[4] https://github.com/openstack/swift/blob/2.22.0/swift/obj/server.py#L688-L701\n[5] https://github.com/openstack/swift/blob/2.22.0/swift/common/middleware/slo.py#L1317\n[6] https://github.com/openstack/swift/blob/2.22.0/test/functional/test_versioned_writes.py#L1026-L1050","accounts_in_message":[],"_revision_number":5},{"id":"9e470fe1b80b9a3803ecd1fc713876865dd2737d","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-14 11:25:02.000000000","message":"Patch Set 5: Code-Review-1 Workflow-1\n\nThank you so much for the quick reply. To be honest I don\u0027t actually know Python programming and failed to set up an SAIO environment with the Docker file from swift repo hence I couldn\u0027t personally test it. This bug caused performance issue when using Swift(the extra call to get correct size) for me and I thought the bug was much simpler so I gave it a shot.\n\nAll that aside, I want to clarify these points with you to see if they were correctly understood:\n- versioned_writes is recommended to be after slo in pipe so the code I changed won\u0027t have any effect.\n- a potential fix is to retrieve size of sysmeta and put them in versioned_writes. However this has drawbacks: Swift versions compatibility and cross responsibility between middlewares.\n\nI can look into it and implement that fix and also add tests for it. However, my biggest challenge right now is setting up the environment, after running the Dockerfile file I didn\u0027t know what else to do, runnning tox failed as well. Much of what I understand were through reading code and just guessing. So any documents/help on that would be much appreciated. It would speed up the learning process a lot.","accounts_in_message":[],"_revision_number":5},{"id":"2d4d5c9ead86c47d8510a68657bed77c8eee2de5","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-14 11:54:17.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"569df5b410a9355a5edd5163e03da61463f1d5ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 12:57:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b3abd24c5d4110a5c7ec75a8c2cc3c : SUCCESS in 8m 16s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/666006b476b34f88810b2eb498022005 : SUCCESS in 1h 01m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a8a66996dbd44e678ce8948a2d87b21f : SUCCESS in 58m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ce2ffba9dcd24b2582d080ec81c2e646 : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/440930ebad684a8e8c75a3783ec95763 : SUCCESS in 33m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ff7155270864c8aaa3c86b7790e5589 : SUCCESS in 18m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1b517329b85446f9ef0984169aa18b5 : SUCCESS in 26m 58s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/38c4bce0cf214dec89900dfe32011ddd : SUCCESS in 20m 47s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/f66c7772b2b94cb190a379f35b6cd3fb : SUCCESS in 35m 10s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/f6374ee2f4ef4102a60bd517020e3fd3 : SUCCESS in 27m 58s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/97356d47923946a3ae101d64b18def36 : SUCCESS in 11m 24s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/05ab543cc5754227ad86f67c27de1055 : SUCCESS in 12m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2ff89adc5740403486419d476cc99e7f : SUCCESS in 11m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/534d1c440d1e41008c72ad7440d14de3 : SUCCESS in 10m 41s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/2b5c22b95c0f47ff966616152a6522fe : SUCCESS in 12m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e75bab9468b84eb1928a389f7d39212d : SUCCESS in 13m 05s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/be32de437b08497394e7ba89db388e43 : SUCCESS in 15m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f767003881694f07b79fcd0984ed5fc0 : SUCCESS in 13m 41s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/990f5a5da9b741a7a5b879c53225804e : SUCCESS in 31m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/31c78ffa21d54336b6824172623787d6 : SUCCESS in 30m 31s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/eebca25616d94600aea0e5b74290af2a : SUCCESS in 30m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f2672790b324ac3ac08eda4ad8ae541 : SUCCESS in 32m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4b9a5c87a204b82821196bd1f864ab8 : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/771942cea970436b90b10454954cf1fb : FAILURE in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/348195132711407981d8d9eea7d94aa0 : SUCCESS in 18m 19s","accounts_in_message":[],"_revision_number":6},{"id":"52a80f09813f43050638cf93dd12570297515c19","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-15 10:03:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0ab4ad74d6a86aacbbb27546dbed14402f38bc07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 11:11:55.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36a8cba913f84ad094d6ed11a43a9d26 : SUCCESS in 5m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a97e2c954e3a449c8b65afad9ba6e28b : SUCCESS in 59m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4c8a29155f064c959d858dbdb3cf07bb : SUCCESS in 47m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f600a8f46a9f4948b558823eff681d92 : SUCCESS in 8m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e9e2660b5cf043b4b6729512fd49536d : SUCCESS in 12m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2b97f69741704dec9fa153de23aa0b0d : SUCCESS in 11m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2be8bb7c42514ae38e8b97fb913fc1c7 : SUCCESS in 12m 00s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/0b12c634b8584dd1b53b5ece24b17694 : FAILURE in 8m 56s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/324535c7be8b488ea37cae9b4e737938 : FAILURE in 10m 06s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/ae266910e5a5478590f463e55c9c3962 : FAILURE in 9m 35s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/48cc78af27ae4b7f8b2b0036665a4f04 : FAILURE in 12m 04s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/8165b9dbaf3644a9aa9738ee43482ce3 : FAILURE in 13m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2b90d3562b684d0983704a5ee1b57b1a : FAILURE in 9m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a6ce9aa7a6934264916ad2594158c259 : FAILURE in 9m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/eb819717fe2d468e99946072a578a3a8 : FAILURE in 10m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f994cdfd412943d998852f6a72caf57b : FAILURE in 10m 44s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cedf9cae001944b2a831875f4de2d63c : FAILURE in 15m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8d97d6f460f442f588c2125d7e771597 : SUCCESS in 12m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/881325b4f3104d338d8993fb7dd24690 : FAILURE in 32m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e15d44934d1842d39ac5bece0c16cebd : FAILURE in 22m 55s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1dcbf0c9d26d440dba5344169fec5970 : FAILURE in 24m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ca440ba59a0f4fa192fa21523ff2f737 : FAILURE in 20m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24584263867344128ad23a48e5c23b09 : SUCCESS in 12m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a45e762f83e14dde8a2b9411e837b26f : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7af2a15cff14487c83808d19c9afe4e0 : SUCCESS in 19m 06s","accounts_in_message":[],"_revision_number":7},{"id":"14785a36c1dbe76690a870476aa1b9bdb4e445de","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-15 11:43:08.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"30c83e67c20df5dfd0d76e35cb52acd2f10f2970","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-15 11:47:31.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1410f5bfa089222bfea00b9de1174704b23761c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 12:55:47.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6fa3519ef8b436a98e8b041181e3d31 : SUCCESS in 6m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8e5cf9854d2f44188ae2483b24bf1a76 : SUCCESS in 1h 02m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7abac7030b694fe1841b60c126babc34 : SUCCESS in 59m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/631ba37ca9f24072ae537247505ede9a : SUCCESS in 8m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40de05939829487b9f8b826df2d01280 : SUCCESS in 12m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/472ba5b6130e422dbb0280aedb150da5 : SUCCESS in 23m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b58b9bb865fa4f75ae539e6523a79983 : SUCCESS in 11m 52s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/60db2c6549b54dcbababe9e57e3c43c5 : FAILURE in 9m 31s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/a6c79509d1e3488b816ea762819ffa3e : FAILURE in 9m 50s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/43bd62605e914bd0b507f6abb3e80bc6 : FAILURE in 12m 16s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/e2f630458c574a56a8da951537c07f8d : FAILURE in 21m 22s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/19af969ab6cb42cc911ceb69e7ffa056 : FAILURE in 13m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c84a8ba7b7bb4b958f15ec673f691a9c : FAILURE in 10m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1181eae80a5b48b0aec3faffd181e2ab : FAILURE in 10m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/fd22efb2eab44409af7e1d813aa15c4c : FAILURE in 21m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5223fa4d070c43b294c4904742b24d6e : FAILURE in 9m 55s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/01cb2fad7b104131b5008da0433018d6 : FAILURE in 13m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e9633c5c890c4c2d9594b7b4664433d4 : SUCCESS in 14m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/32796cc29316471ea160f6a1728c2158 : SUCCESS in 40m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d205b6cfb5a24c8497837d63c5807190 : FAILURE in 26m 03s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fdca4e92699748eaa177eb8498ffa1f5 : FAILURE in 22m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e9f936375e1e464ebd88eb7c4dd2dac9 : FAILURE in 24m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e47e2e9ba5141708027966a76ddc6d0 : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e32c45ec8b0842788c38135406e8fa99 : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6d72c241c17b41cebc245f68909e4ec2 : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":9},{"id":"3aa8b3be624957297826abe5dd656faa426a1d5e","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-15 13:25:33.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7a68a2982e87a6e325357e0521c0da743bd7bbaa","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-15 13:31:22.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4d694d286a7af691a2310b5f3193972137479828","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 14:36:59.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8bd8b2af4d84eaabd18df5a22f7b2ef : SUCCESS in 4m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/42f4a3e92b954a568bd2017b4b46bffd : SUCCESS in 59m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/681bda3c912f483e804635d8cd738e6c : SUCCESS in 52m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/60cec798881a4b03a1c10576cdd0e8aa : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2a55755418684bc39b6485ca8b7350bb : SUCCESS in 10m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1e019a6cccde4296a8aadb4d92c56cf9 : SUCCESS in 11m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3a2209b7153346c2b5cc68dabb5fbc7a : SUCCESS in 10m 37s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/026436decd034ed298ccfd829646795d : SUCCESS in 9m 00s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/18ea04336cba4f03a914d46cb4470333 : SUCCESS in 9m 18s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/e07124f5081b4d199e7730007058e14c : SUCCESS in 9m 18s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/f3e6b91ec3174d76bf009b4b01ce8a31 : SUCCESS in 10m 20s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/20a918d5b6e64546901a9bbe1a24274c : SUCCESS in 11m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/aea3fdd14b7141bbb5671b10cf2153b4 : SUCCESS in 8m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ef06ecb1840e462ba6f29f7baf32b9f7 : SUCCESS in 9m 32s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/07f79ee0f0a54007b0196c1603d66897 : SUCCESS in 9m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3335a4f1ab58413e82a9eaa1e5932eeb : SUCCESS in 9m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6b398df8f789439e8f2871afcb125856 : SUCCESS in 12m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c4ca2ea50ba54394b100237e0a513736 : SUCCESS in 12m 48s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c16fda57654d423cbb901e29d45150cc : SUCCESS in 30m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b5f5201b9e364be09b2e224a2148da7d : SUCCESS in 28m 58s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/8ca09b3afe8a49c5a3a19e6236da0292 : SUCCESS in 23m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dca111dacc4d49bdb86f9be5ccab8bcf : SUCCESS in 28m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c701c2e528b40f3a445ed6a956de54a : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ef4079073b4bc0a3e23c2fbe5e6154 : SUCCESS in 5m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f2988afe38f4c7c8da8ed612f9fb039 : SUCCESS in 16m 22s","accounts_in_message":[],"_revision_number":11},{"id":"5b4274c1852e5c40fc8e5281530eca1a14964a6d","author":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"date":"2019-09-15 16:14:24.000000000","message":"Patch Set 11: Code-Review+1\n\n(3 comments)\n\nAdded suggested fix and functional test","accounts_in_message":[],"_revision_number":11},{"id":"f80c0d021ffda5de97cdc2e809da20f32b21b514","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-18 18:00:26.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nThanks for tackling this! Change looks sane, tests pass, and if I back out the change in versioned_writes, the new test fails beautifully:\n\nFAIL: test_slo_manifest_version_size (test.functional.test_versioned_writes.TestSloWithVersioning)\n----------------------------------------------------------------------\nTraceback (most recent call last):\n  File \".../test/functional/test_versioned_writes.py\", line 1094, in test_slo_manifest_version_size\n    self.assertEqual(version_file_size, org_size)\nAssertionError: 202 !\u003d 1048576\n    \u0027202 !\u003d 1048576\u0027 \u003d \u0027%s !\u003d %s\u0027 % _common_shorten_repr(202, 1048576)\n    \u0027202 !\u003d 1048576\u0027 \u003d self._formatMessage(\u0027202 !\u003d 1048576\u0027, \u0027202 !\u003d 1048576\u0027)\n\nIf you\u0027re still having trouble getting an SAIO environment, I\u0027d recommend https://github.com/swiftstack/vagrant-swift-all-in-one -- it\u0027s what I usually use for dev work. Alternatively, keep working through https://docs.openstack.org/swift/latest/development_saio.html -- it\u0027s likely a bit out of date (the references to \"Ubuntu 14.04, 16.04 LTS\" are a little disconcerting!) but it\u0027d be great to have that exercised and fixed up if you\u0027ve got a bit of time. If you need any help, IRC will likely get you better turn-around times; we\u0027re in #openstack-swift on freenode.","accounts_in_message":[],"_revision_number":11},{"id":"7b4e601ea91e38cdde383ba06c0ec1080184c8c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:00:47.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"b2f0b69bff1cec6fef4ff37879e14562a235d00d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:10:54.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0476f7f8d964a27b82331432dcd18d1 : SUCCESS in 5m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fab344dd79114fa187f4f3dc4739cbd0 : SUCCESS in 1h 07m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39a6aca32fea44db9a9c64cf452c7fa8 : SUCCESS in 57m 37s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/aaa20a74859f424196cbac42a2c99fcd : SUCCESS in 9m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e53d34cabae7401d87ea921478510e95 : SUCCESS in 12m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dcfc20ad17c34df6be37f55d25c6b5a1 : SUCCESS in 11m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a32cde9bb3424064b846b73d6b84386c : SUCCESS in 11m 51s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/1d5c10cacfec43918790d93a7f4c4ff0 : SUCCESS in 10m 35s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/b13a937528f24886aa7607f140107926 : SUCCESS in 9m 36s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/ef46f0adf9774517b33b8b1adb92023d : SUCCESS in 10m 20s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/7400f6881ed84acab13425bb11f6d229 : SUCCESS in 10m 09s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/1ee814f434ae49d7af4ed8212fe2662b : SUCCESS in 13m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4dd9444716e24d87be56905852d58cc6 : SUCCESS in 12m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a05c45ee80174f8784480b6df3e26b83 : SUCCESS in 11m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/59378f5eac0d468ca124f27f41147c57 : SUCCESS in 10m 59s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/356e97c5babd4db19bfb4aa61a30db0a : SUCCESS in 15m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a411e8ed64c745e8a73812eb7e558f38 : SUCCESS in 31m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f64118dd1494e4bb83c3e0638eaff68 : SUCCESS in 32m 12s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7da4eeef97bb4c1a88f9d0278fbe1320 : SUCCESS in 31m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5df75a1b3f7d4769a8a544fb0e91a7f0 : SUCCESS in 36m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4245f01102ae4448ab5eb03d7258846b : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18986a41f90467095326cd9a29ab2a9 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":11},{"id":"4b605ac03a1b37d4ad606f2d1d6e2a48839130d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:10:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"015147f64a56383e16ce71247117f1db0d97e191","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:12:41.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaece55533c84df596250c20be29b93f : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/52c2b9b95ef44c8d9c5b781027ef4f3c : SUCCESS in 53s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"b4288b4aa6e6be2222f5f0e9ca8360c07040d5c0","revisions":{"dff27abaf70662d78bf56891d46ad32569bdb113":{"kind":"REWORK","_number":1,"created":"2019-09-12 07:39:56.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/1"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 07:39:04.000000000","tz":420},"subject":"Versioned Writes checks for SLO object before copy","message":"Versioned Writes checks for SLO object before copy\n\nversioned_writes middleware will now check for SLO header\nto add query \u0027multipart-manifest\u003dput\u0027 if necessary\nbefore performing copy from current to versioned container\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dff27abaf70662d78bf56891d46ad32569bdb113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dff27abaf70662d78bf56891d46ad32569bdb113"}]},"branch":"refs/heads/master"},"48980adfae3929ecc51f2fbbeabaca22a096fb29":{"kind":"REWORK","_number":2,"created":"2019-09-12 15:36:37.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/2"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 15:34:05.000000000","tz":0},"subject":"Versioned Writes checks for SLO object before copy","message":"Versioned Writes checks for SLO object before copy\n\nversioned_writes middleware will now check for SLO header\nto add query \u0027multipart-manifest\u003dput\u0027 if necessary\nbefore performing copy from current to versioned container\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48980adfae3929ecc51f2fbbeabaca22a096fb29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48980adfae3929ecc51f2fbbeabaca22a096fb29"}]},"branch":"refs/heads/master"},"9f24392d154ad8c0820f382cb45748b42e8cf45e":{"kind":"REWORK","_number":3,"created":"2019-09-12 15:42:28.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/3"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 15:42:23.000000000","tz":0},"subject":"Versioned Writes checks for SLO object before copy","message":"Versioned Writes checks for SLO object before copy\n\nversioned_writes middleware will now check for SLO header\nto add query \u0027multipart-manifest\u003dput\u0027 if necessary\nbefore performing copy from current to versioned container\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f24392d154ad8c0820f382cb45748b42e8cf45e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f24392d154ad8c0820f382cb45748b42e8cf45e"}]},"branch":"refs/heads/master"},"ad29f34e212d9039cba5f39177f869afee0e82c3":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-09-12 15:45:15.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/4"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 15:45:11.000000000","tz":0},"subject":"versioned_writes checks for SLO object before copy","message":"versioned_writes checks for SLO object before copy\n\nPreviously, versioned_writes middleware copy an already existing\nobject using PUT. However, SLO requires the additional query\nto properly update the object size when listing.\n\nProposed fix: versioned_writes middleware will now check for SLO header\nto add query \u0027multipart-manifest\u003dput\u0027 if necessary\nbefore performing copy from current to versioned container\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad29f34e212d9039cba5f39177f869afee0e82c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad29f34e212d9039cba5f39177f869afee0e82c3"}]},"branch":"refs/heads/master"},"46cf7a15b1906e523cba5ccc4e0c591835d44896":{"kind":"REWORK","_number":5,"created":"2019-09-13 04:50:09.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/5"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-13 04:49:52.000000000","tz":0},"subject":"versioned_writes checks for SLO object before copy","message":"versioned_writes checks for SLO object before copy\n\nPreviously, versioned_writes middleware copy an already existing\nobject using PUT. However, SLO requires the additional query\nto properly update the object size when listing.\n\nProposed fix: versioned_writes middleware will now check for SLO header\nto add query \u0027multipart-manifest\u003dput\u0027 if necessary\nbefore performing copy from current to versioned container\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46cf7a15b1906e523cba5ccc4e0c591835d44896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46cf7a15b1906e523cba5ccc4e0c591835d44896"}]},"branch":"refs/heads/master"},"a6b51b4059faa1b237f2a971aa5cfb4f74877281":{"kind":"REWORK","_number":6,"created":"2019-09-14 11:54:17.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/6"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-14 11:53:51.000000000","tz":0},"subject":"versioned_writes checks for SLO object before copy","message":"versioned_writes checks for SLO object before copy\n\nPreviously, versioned_writes middleware copy an already existing\nobject using PUT. However, SLO requires the additional query\nto properly update the object size when listing.\n\nProposed fix: versioned_writes middleware will now check for SLO header\nto add query \u0027multipart-manifest\u003dput\u0027 if necessary\nbefore performing copy from current to versioned container\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6b51b4059faa1b237f2a971aa5cfb4f74877281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6b51b4059faa1b237f2a971aa5cfb4f74877281"}]},"branch":"refs/heads/master"},"d2532614937e9343b2c59c08758b47c64ae7e4ac":{"kind":"REWORK","_number":7,"created":"2019-09-15 10:03:01.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/7"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-15 09:56:32.000000000","tz":420},"subject":"versioned_writes: checks for SLO object before copy","message":"versioned_writes: checks for SLO object before copy\n\nPreviously, versioned_writes middleware copy an already existing\nobject using PUT. However, SLO requires the additional query\nto properly update the object size when listing.\n\nPropose fix: In _put_versioned_obj - which is called when on\ncreating version obj and also on restoring obj,\nif \u0027X-Object-Sysmeta-Slo-Size\u0027 header is present it will\nadd needed headers for container to update obj size\n\nAdded a new functional test case with size assertion for slo\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2532614937e9343b2c59c08758b47c64ae7e4ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2532614937e9343b2c59c08758b47c64ae7e4ac"}]},"branch":"refs/heads/master"},"a264da4b3607d92931919cbb5ce071b21bd79dcb":{"kind":"REWORK","_number":8,"created":"2019-09-15 11:43:08.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/8"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-15 11:42:41.000000000","tz":420},"subject":"versioned_writes: checks for SLO object before copy","message":"versioned_writes: checks for SLO object before copy\n\nPreviously, versioned_writes middleware copy an already existing\nobject using PUT. However, SLO requires the additional query\nto properly update the object size when listing.\n\nPropose fix: In _put_versioned_obj - which is called when on\ncreating version obj and also on restoring obj,\nif \u0027X-Object-Sysmeta-Slo-Size\u0027 header is present it will\nadd needed headers for container to update obj size\n\nAdded a new functional test case with size assertion for slo\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a264da4b3607d92931919cbb5ce071b21bd79dcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a264da4b3607d92931919cbb5ce071b21bd79dcb"}]},"branch":"refs/heads/master"},"8f1ed1a4ecf04f6eaf5b63f69d87623f1d54e811":{"kind":"REWORK","_number":9,"created":"2019-09-15 11:47:31.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/9"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-15 11:47:14.000000000","tz":420},"subject":"versioned_writes: checks for SLO object before copy","message":"versioned_writes: checks for SLO object before copy\n\nPreviously, versioned_writes middleware copy an already existing\nobject using PUT. However, SLO requires the additional query\nto properly update the object size when listing.\n\nPropose fix: In _put_versioned_obj - which is called when on\ncreating version obj and also on restoring obj,\nif \u0027X-Object-Sysmeta-Slo-Size\u0027 header is present it will\nadd needed headers for container to update obj size\n\nAdded a new functional test case with size assertion for slo\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f1ed1a4ecf04f6eaf5b63f69d87623f1d54e811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f1ed1a4ecf04f6eaf5b63f69d87623f1d54e811"}]},"branch":"refs/heads/master"},"439fa7a03854eaf8a581a164f350ceb0992711ec":{"kind":"REWORK","_number":10,"created":"2019-09-15 13:25:33.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/10"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-15 13:25:07.000000000","tz":420},"subject":"versioned_writes: checks for SLO object before copy","message":"versioned_writes: checks for SLO object before copy\n\nPreviously, versioned_writes middleware copy an already existing\nobject using PUT. However, SLO requires the additional query\nto properly update the object size when listing.\n\nPropose fix: In _put_versioned_obj - which is called when on\ncreating version obj and also on restoring obj,\nif \u0027X-Object-Sysmeta-Slo-Size\u0027 header is present it will\nadd needed headers for container to update obj size\n\nAdded a new functional test case with size assertion for slo\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/439fa7a03854eaf8a581a164f350ceb0992711ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/439fa7a03854eaf8a581a164f350ceb0992711ec"}]},"branch":"refs/heads/master"},"b4288b4aa6e6be2222f5f0e9ca8360c07040d5c0":{"kind":"REWORK","_number":11,"created":"2019-09-15 13:31:22.000000000","uploader":{"_account_id":30970,"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","username":"bluv"},"ref":"refs/changes/56/681656/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/56/681656/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/56/681656/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/56/681656/11"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-12 04:31:42.000000000","tz":420},"committer":{"name":"Nguyen Quoc Viet","email":"nguyenqviet98@gmail.com","date":"2019-09-15 13:30:56.000000000","tz":420},"subject":"versioned_writes: checks for SLO object before copy","message":"versioned_writes: checks for SLO object before copy\n\nPreviously, versioned_writes middleware copy an already existing\nobject using PUT. However, SLO requires the additional query\nto properly update the object size when listing.\n\nPropose fix: In _put_versioned_obj - which is called when on\ncreating version obj and also on restoring obj,\nif \u0027X-Object-Sysmeta-Slo-Size\u0027 header is present it will\nadd needed headers for container to update obj size\n\nAdded a new functional test case with size assertion for slo\n\nChange-Id: I47e0663e67daea8f1cf4eaf3c47e7c8429fd81bc\nCloses-Bug: #1840322\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4288b4aa6e6be2222f5f0e9ca8360c07040d5c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4288b4aa6e6be2222f5f0e9ca8360c07040d5c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
