)]}'
{"id":"openstack%2Fswift~681970","triplet_id":"openstack%2Fswift~master~I1e502c328be16fca5f1cca2186b27a0545fecc16","project":"openstack/swift","branch":"master","topic":"bug/1843313","hashtags":[],"change_id":"I1e502c328be16fca5f1cca2186b27a0545fecc16","subject":"Sharding: Clean up old CleaveConext\u0027s during audit","status":"MERGED","created":"2019-09-13 06:24:02.000000000","updated":"2019-09-20 16:16:06.000000000","submitted":"2019-09-20 16:14:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681970-1568996060051-6712634d","meta_rev_id":"e5659b31cf1ce266f58b4158826534a324425d08","_number":681970,"virtual_id_number":681970,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-19 05:28:34.000000000","_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-09-20 16:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2019-09-20 13:43:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-19 23:58:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2019-09-20 13:43:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-19 23:58:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-09-20 13:43:00.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-09-20 16:14:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebf1dc6aa9899a16a810b5fe23ab4e0783c46a0f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-09-13 06:24:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1f5580160ac63245948efcd2076ed4a32bb922f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 12:08:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98653159e3574ef9b95936a795576213 : SUCCESS in 5m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bfbce3f1fd2b4182b6e13c37c1a6d3b8 : SUCCESS in 58m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3f80e37476d747bda7fe2e4dcb9c8b9c : SUCCESS in 54m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a16a7912465248b18f5754fac0142ddf : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9555bdba5bf643b9bbc4275d9c9a3755 : FAILURE in 10m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/639a3a96094840da9ab29c49044eae50 : FAILURE in 10m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/47af6d0becee441fa82cd4ca03116e09 : FAILURE in 13m 11s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/001ed45929f54318a64a822c443f741d : SUCCESS in 9m 34s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/73138d1b946249e08ed5cdac5acc70f0 : SUCCESS in 10m 23s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/41a9ba0dedd6496789704a19efd833b0 : SUCCESS in 10m 41s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/55a4987b69ee4bcf854c089e1eef15f3 : SUCCESS in 10m 08s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/99f6ad085fac4340848c665a9b966e02 : SUCCESS in 12m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ddd084969a2347b5af050794d309b3a8 : SUCCESS in 8m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fd203e3b71ea41f3b5a2a8b52233042d : SUCCESS in 8m 33s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/36b69cd3c0194a48ae3783597659e322 : SUCCESS in 9m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d0640a6db02c40449fbb694262a557fa : SUCCESS in 10m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f286e9fd5a634ece82d42372ee28bb8d : SUCCESS in 13m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/121d9189d2e14f30baffaf19197a8372 : SUCCESS in 12m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4de1ff6a97084bc181fbfeb9d99a5c0a : FAILURE in 28m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae101b20b8e64e4e90e7512b2647594a : SUCCESS in 38m 14s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/eea565a13b79481ca82762e8d2c9f633 : SUCCESS in 30m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0864f2bcdda74b6fbcf292989e02ecd1 : SUCCESS in 27m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fad4973213246c0836fbd1ec8604bef : FAILURE in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73bb7ffbd1474c13baf730dd7d5efb90 : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d9da36cfbad2482f82dba26a53e1fe21 : SUCCESS in 19m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e2acb785bd90150ff84c54af339b22c6d0bc89c6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-09-18 07:11:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bcb3171b092174d47966df50d31ec55de49af5e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 12:02:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c0fef122dc42bbbb024002238b5760 : SUCCESS in 6m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7c8fc7583f534feea107ac0760f5caf1 : SUCCESS in 1h 04m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5c3f11e6b35b483f8566e6e29612365d : SUCCESS in 1h 03m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/95aec2d5b0364ef695b0c10b4206cc29 : SUCCESS in 10m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a88a8a7e9124a49baefc2945bc2a94b : SUCCESS in 11m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/17f6f370f1374a04a5ecf68d1f30ce8f : SUCCESS in 12m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c5fcfe575ec941ae895582be4028f017 : SUCCESS in 13m 49s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/d107593acb104cc5b730d25b534b536d : SUCCESS in 10m 01s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/963cfe26acdd4a18888c72f914e9b584 : SUCCESS in 11m 23s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/5ae24ba954ab4a0e85c3b8480ec20df7 : SUCCESS in 9m 15s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/0a1af036dae44dd4b9d5adf417219f88 : SUCCESS in 9m 13s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/f2a597042c274f6992ce9e73f2f02f7e : SUCCESS in 12m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2f136fc40a2448dba3e8f706c7a84a8a : SUCCESS in 9m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6f71fd33adc54ed0a182b191bcb0e1a0 : SUCCESS in 9m 15s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/780a070306e9464ea5b0a6a1af2db7c5 : SUCCESS in 11m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/192a2501aa674845aa3d3ee0c94434e1 : SUCCESS in 9m 54s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/35ad1ab81a7f4bb2bf3b24ed153482f1 : SUCCESS in 13m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61c9afb672cc481d8378b79c46e0beac : SUCCESS in 14m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/78389c33b27943d6aa133ac4899fab3a : SUCCESS in 41m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f5b1837c0a6c40b795b62a8899438b60 : SUCCESS in 31m 08s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/6e79394a69294c638d0cf496a30684d2 : SUCCESS in 35m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/65b41e3db19942e0ae58e9efa7f917d6 : SUCCESS in 31m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13a9be9a14c74d64a6d060d0cccbf2ff : SUCCESS in 12m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c3e6378ab464188832e59eee216e90e : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5486111bd337484a8be364f2a550777e : SUCCESS in 17m 49s","accounts_in_message":[],"_revision_number":2},{"id":"21e351771ea7f5119d65493cef566daed4c80076","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-09-18 21:37:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(5 comments)\n\nI\u0027m definitely behind the learning curve here","accounts_in_message":[],"_revision_number":2},{"id":"fdf5fdd0f41448d5accde93ce66a1043a4f4dff1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2019-09-19 05:28:34.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nthanks for the review Clay, I hope my rambles make sense. And please correct me if I\u0027m wrong!","accounts_in_message":[],"_revision_number":2},{"id":"adebfa2ba0fedfa6274d679a5fed10c77bdbfdef","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-19 06:22:47.000000000","message":"Patch Set 2: Code-Review+1\n\n(7 comments)\n\nI\u0027ll take a closer look soon -- but this seems to make sense to me...","accounts_in_message":[],"_revision_number":2},{"id":"eb6ef2995c4c1ddca4aba3b249b2556db876b476","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-19 16:34:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ad429cdfc251f9bccdfc74f4734d93ade37b78c3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-19 23:58:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)\n\nLGTM!\n\nUpdated some probe tests in https://review.opendev.org/#/c/683257/\n\nOnly remaining question in my mind is whether we want to add last_modified or just use the sysmeta\u0027s timestamp. I think I want this for Train, though, so a mergeable patch that fixes the issue sure looks appealing...","accounts_in_message":[],"_revision_number":2},{"id":"89ed585033c8fbeb147da1888b72075ac50a7f14","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-09-20 13:43:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nthanks for all the explanation guys - these seems solid\n\nI agree with tim; there\u0027s a convincing argument to be made for using the explicit last_modified - i don\u0027t see a bit of extra sysmeta for clarity as prohibitively \"expensive\" compared to the value of the fix!","accounts_in_message":[],"_revision_number":2},{"id":"f75a48f74e13a3059702d3e16ae50c389e925370","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 13:43:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"565948d6d2cc9bca43faba8cb66bebcf4d84015f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:14:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4457d5d6a19744908febe4f265b96428 : SUCCESS in 4m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/735c3ec0f18f4204b33d22836bb3eb1d : SUCCESS in 1h 13m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/850e1c51655f49bfb3440b2f71fc6533 : SUCCESS in 1h 12m 48s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/571c8b5172d84eb2a693051c30dacfaf : SUCCESS in 8m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f022b90261a84872b7fcc5737e4f6c98 : SUCCESS in 12m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/16e3360f50354bfaa1283da4bfe8fa13 : SUCCESS in 11m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/de4126355be24a4b87e0030fe395fbc0 : SUCCESS in 11m 23s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/fabf2cca70194008879220f0762cb175 : SUCCESS in 12m 22s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/15f93de686bf4e4db6ef767322a2d481 : SUCCESS in 10m 52s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/308618fad58546ec88474a3485cbb90f : SUCCESS in 12m 49s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/a73db5d5e11244498962e92dba299e76 : SUCCESS in 11m 40s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/b31387bc940f4997af497e01b5c00378 : SUCCESS in 16m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6d90ad64300542fabbd565279d59d93a : SUCCESS in 9m 12s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/bcb98fd212a644cabc845373f71220aa : SUCCESS in 11m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2dd9673c8034438ba0e5dfd46db6f046 : SUCCESS in 12m 19s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/28a888c49422490699ddc0bdc41736e9 : SUCCESS in 15m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0e09850bc0924505be2278221cda1826 : SUCCESS in 32m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/49453d2728174d34b76d1a8a268a1856 : SUCCESS in 34m 40s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9ccaf73dabb34a7991a115c2bbec14fc : SUCCESS in 30m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/694a8ac5013943fe963a40a45742c119 : SUCCESS in 30m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3d2698556454ca0a71ada7e597673a6 : SUCCESS in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12494f0a74f14b3db4f81ba15c647586 : SUCCESS in 5m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac2cb8afbec345188adee8a4f7a2a1f6 : SUCCESS in 1h 12m 06s","accounts_in_message":[],"_revision_number":2},{"id":"0a1ded21571e14f83f0acf0cfdb62e3ba83c2c48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:14:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"89dfd205fad1eb83ef697b8d27e21e2f4352b0d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:16:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91e228be185746708ffb5b3ab2d17c66 : SUCCESS in 1m 16s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f04722720a124c959eed5f7c9272c3fd : SUCCESS in 1m 00s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81a41da5420313f9cdb9c759bbb0f46c0d20c5af","revisions":{"ac1582ff3ec2ad4c34aab9e7411a6d20b243126e":{"kind":"REWORK","_number":1,"created":"2019-09-13 06:24:02.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/70/681970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/681970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/681970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/681970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/681970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/681970/1"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-09-13 06:16:06.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-09-13 06:16:06.000000000","tz":600},"subject":"Sharding: Clean up old CleaveConext\u0027s during audit","message":"Sharding: Clean up old CleaveConext\u0027s during audit\n\nThere is a sharding edge case where more CleaveContext are generated and\nstored in the sharding container DB. If this number get\u0027s high enough,\nlike in the linked bug. If enough CleaveContects build up in the DB then\nthis can lead to the 503\u0027s when attempting to list the container due to\nall the `X-Container-Sysmeta-Shard-Context-*` headers.\n\nThis patch resolves this by tracking the a CleaveContext\u0027s last\nmodified. And during the sharding audit, any context\u0027s that hasn\u0027t been\ntouched after reclaim_age are deleted.\n\nThis plus the skip empty ranges patches should improve these handoff\nshards.\n\nChange-Id: I1e502c328be16fca5f1cca2186b27a0545fecc16\nCloses-Bug: #1843313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac1582ff3ec2ad4c34aab9e7411a6d20b243126e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac1582ff3ec2ad4c34aab9e7411a6d20b243126e"}]},"branch":"refs/heads/master"},"81a41da5420313f9cdb9c759bbb0f46c0d20c5af":{"kind":"REWORK","_number":2,"created":"2019-09-18 07:11:00.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/70/681970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/681970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/681970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/681970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/681970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/681970/2"}}},"commit":{"parents":[{"commit":"5cb53838a63791109697c91f742079b6dfb25263","subject":"Merge \"py3: (mostly) port probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cb53838a63791109697c91f742079b6dfb25263"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-09-13 06:16:06.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2019-09-18 07:10:36.000000000","tz":600},"subject":"Sharding: Clean up old CleaveConext\u0027s during audit","message":"Sharding: Clean up old CleaveConext\u0027s during audit\n\nThere is a sharding edge case where more CleaveContext are generated and\nstored in the sharding container DB. If this number get\u0027s high enough,\nlike in the linked bug. If enough CleaveContects build up in the DB then\nthis can lead to the 503\u0027s when attempting to list the container due to\nall the `X-Container-Sysmeta-Shard-Context-*` headers.\n\nThis patch resolves this by tracking the a CleaveContext\u0027s last\nmodified. And during the sharding audit, any context\u0027s that hasn\u0027t been\ntouched after reclaim_age are deleted.\n\nThis plus the skip empty ranges patches should improve these handoff\nshards.\n\nChange-Id: I1e502c328be16fca5f1cca2186b27a0545fecc16\nCloses-Bug: #1843313\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81a41da5420313f9cdb9c759bbb0f46c0d20c5af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81a41da5420313f9cdb9c759bbb0f46c0d20c5af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
