)]}'
{"id":"openstack%2Fswift~682112","triplet_id":"openstack%2Fswift~stable%2Fstein~I73f84b96f164e6fc5d3cb890355871c26ed271a6","project":"openstack/swift","branch":"stable/stein","topic":"bug/1843816","hashtags":[],"change_id":"I73f84b96f164e6fc5d3cb890355871c26ed271a6","subject":"bufferedhttp: ensure query params are properly quoted","status":"MERGED","created":"2019-09-13 16:46:59.000000000","updated":"2019-09-21 17:41:13.000000000","submitted":"2019-09-21 17:39:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682112-1569087593311-0f18a21d","meta_rev_id":"efc08097a6e8b1616b67b085606b4f75f4e7c1b1","_number":682112,"virtual_id_number":682112,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-09-21 17:39:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-21 04:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-21 04:32:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-21 17:39:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6dd98cf4d4a606b5976c72b4b0288070a012c74d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-13 16:46:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"035e58725ccaf9526e0d050fa111ba1c7751b87d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-13 16:49:21.000000000","message":"Patch Set 1: Workflow-1\n\nWill rebase this on a backport of https://review.opendev.org/#/c/642893/ shortly -- want to confirm that stein is currently busted...\n\nWe want to get this back to ocata/pike sometime soon anyway, so may as well use it to check the job statuses.","accounts_in_message":[],"_revision_number":1},{"id":"c6205a5bc32818ddfcd33a21e916affecf1065dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 01:32:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70eee5b8e6a2404fa440bd3a4cdf7070 : SUCCESS in 4m 39s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4792a8f69c014602af797c073ff95de5 : SUCCESS in 10m 48s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/38865d9914024ebfb2852bed41bddba6 : SUCCESS in 19m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/83f3e08749ef41c6b54501d101cdf72e : SUCCESS in 8m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f65b39d018b4ad9873ba35e33265ccc : SUCCESS in 8m 39s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/85099c4037fb438fbb1b62b4beed7ec2 : FAILURE in 9m 40s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/b08434bbf8274acf8a63b47a792c7865 : FAILURE in 8m 27s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/510a3bc6f4ee46b7ac4abb05cd8677dc : FAILURE in 8m 24s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/97f6f470828a4540b5fd70e1a6bef208 : FAILURE in 8m 39s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/491774a481dd469e9058f9d8190a3a15 : FAILURE in 11m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/51b82b1d0c70415fb199a6150c6a2090 : SUCCESS in 12m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2626c24b10cd450391159baf4ac3d29a : SUCCESS in 31m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/43e20001955c425c9d6de9d781964496 : FAILURE in 25m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d0fe1bc012b94026adab7c6b668b835b : FAILURE in 21m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10a0fce3f7434647a1ffab3edf93cea5 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbea2200ec0c4562a1b355ae4173441e : SUCCESS in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d3190d630a864e55817c69e6e4c32c37 : SUCCESS in 17m 07s","accounts_in_message":[],"_revision_number":1},{"id":"8e3e69277562ad7d66e5f09abc52a75743b159d2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-19 16:54:19.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"09bf84ab9af84c338a34a90da749864357ab136a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 03:56:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b761859ebf047c98c773bbe7ab9494f : SUCCESS in 6m 05s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c0b12e4c43cb48849499b64766550c95 : SUCCESS in 10m 18s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/fd8429c87f3b40b2a978b5eed21123c0 : SUCCESS in 9m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eaa4cc2811464367892f55b780f3610f : FAILURE in 8m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8c1300054f5b4e4db7f60f29b5b9c0a5 : SUCCESS in 9m 39s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/77b37b9c2be04792883b5903250ad566 : SUCCESS in 10m 06s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/b79d8504041f4c5bbc8521cbcacffa4a : SUCCESS in 8m 52s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/e45f8e7f325d49219e5709e265ef830b : SUCCESS in 8m 54s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/0a035e4b1035488bbfbc07c879a00473 : SUCCESS in 10m 41s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/2bf748ffda3e45d1be2a44f1f8d07fbd : SUCCESS in 12m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/10e58d0885244d2c8ea092edb3e3192d : SUCCESS in 15m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f30e3bcfd79c420c9262f66fb9eec5ba : FAILURE in 27m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/355f76a1d5e44c98b1838e0b011840e6 : SUCCESS in 34m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/93bc524e6ce5444d809887081d15b2a0 : SUCCESS in 26m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b989e05ae55b4e54957f80633faff80a : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf97621f4324455caa23a4478108c718 : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59e67b05fd694597aa57b34b0e8a64b3 : FAILURE in 23m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a27022994cea9c95926aa66a9669917873ae1b58","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-20 04:27:01.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b5738b2416cf88d68b35df7fb1f60007c7470844","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:09:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c82e7b7e97f489695adf39e2c1f2110 : SUCCESS in 7m 18s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/094dbc06af0847859e55fb45104d1f04 : FAILURE in 17m 36s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/2a9c11bc8c9b450f87a9202d155cda86 : SUCCESS in 14m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0162c9a2f7e34d139cc455840e1adcf4 : SUCCESS in 13m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/585808e2b4064e56aa1b58eb6bda8827 : SUCCESS in 13m 22s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/e9a6c7fcb1fe493bad7dba3aebbf70de : SUCCESS in 8m 45s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/f83a6373201c47cbbe52e92da3f36a42 : SUCCESS in 9m 45s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/301089cd7f714150bec94e423fce698a : SUCCESS in 8m 25s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/dedc2b0b36c146d49b160e23001f6170 : SUCCESS in 9m 35s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/dc010bb157e14a40a2a22719e2f0eaa5 : SUCCESS in 12m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6867dadce9464d4390ad86cfdd270f21 : SUCCESS in 14m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dd7542fc93324a6fa2c8d8da31fa4155 : SUCCESS in 30m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/faca66bfc1f94ec1b88856dcda6582bc : SUCCESS in 32m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/863b19801f4f44adb2d0efcdb2d7a6fa : SUCCESS in 30m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ba34ca77fec54ff78c173980f97366fd : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b329ef92bb084374b83e26622f6f9172 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2388e112196b415a8973afec4d3bbebc : FAILURE in 16m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d04ef0314fffb7e18efc19d85573fd92f813ac6d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-20 15:56:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3a10092d7245d304e54e330ba649abda9c2b568a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:26:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f34366216ed48a588af9a0571d8ecae : SUCCESS in 4m 52s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f3c8ba5ee1ab43daad1ec298cb771e72 : SUCCESS in 10m 50s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/8c806ce480f74a079b67f11e873f71e2 : SUCCESS in 9m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c866c3aaf6ae4313af3c8fb611606137 : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3dc9fe3d6a1e4ccab8e18aed4381250e : SUCCESS in 10m 11s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/21a4d998987f42e7960fc4f2fd72bed8 : SUCCESS in 8m 04s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/1ff2fa98019d4fecbeb95242e4cc4917 : SUCCESS in 9m 59s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/cb51be4087fd45bab9f0e4028fc0e9a9 : SUCCESS in 10m 38s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/bedf42eb655942709d8793ca0ea0a920 : SUCCESS in 9m 26s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/1d867cd3d514431ebe44efbcbbfbf916 : SUCCESS in 9m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ece5762407684179a9c91369afa39e02 : SUCCESS in 16m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d1611504ece9457ab4492a62e194cc24 : SUCCESS in 33m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fd0e1c60df745ab88fcf7a0bed51b8c : SUCCESS in 31m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f8c048cb3d6484ba1935458708938d1 : SUCCESS in 29m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1b49d6df2ae46748b14ad2a14340eba : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2adebcfba05643d2a2975605db89f19a : SUCCESS in 5m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ea1ebefbfa94b8e86b253bbcbaaa016 : FAILURE in 21m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f573882f41d906f98302b443db626f9917282b31","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-21 04:32:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"15fdadd4975ec8965327ac1e55c178ea10b0cb96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 04:32:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6535ea5a01263dc89e13d23fb8366bedfe7388b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 05:54:38.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97e07ac35afa490a977bcc64cd5ab56f : SUCCESS in 5m 26s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/185ef7482d5a4b74a396ff8d2889925c : SUCCESS in 12m 45s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/8be3da67189f40f99bdcf7bac5dc182c : SUCCESS in 9m 11s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/cbb95a50509f43419274a910d4cddbfc : SUCCESS in 9m 06s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/24901a4f791445a48a5e961350f65a02 : SUCCESS in 9m 11s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/899ef924f5fc47488030db9c1ed9f21e : SUCCESS in 9m 24s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/cc150071a101493e858980d6b0aed009 : SUCCESS in 9m 44s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/922592045404440b834b1901a01e8a76 : SUCCESS in 10m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3fd04547b9c24c81a7a44d2de7d1892a : FAILURE in 43m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6b43d3d9dc354114acde428c6f3ecae9 : SUCCESS in 30m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70c1a21dc98d461e99f9d5e13d59a517 : SUCCESS in 31m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0180d3df41de439894a64df2263ef498 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8352c914fa7649b0a6a5eb8e6ad8bcc7 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":3},{"id":"4188e75804d00daa5dc50044b2b82a21ce389ae9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-21 16:14:10.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"239fe83f3f4a6f3780198a10a657b734bef74ecd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 17:04:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6e49681f6548dba3a68b7b83fa5993 : SUCCESS in 4m 37s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ae17b15aa074bf0aafbdf0157e9433b : SUCCESS in 10m 45s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/e3b478ca307c405fbff67695cdf47702 : SUCCESS in 14m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9cff2baafe634c469cbc9d8938159f0a : SUCCESS in 10m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c2b1a149f927409c91511b859ef80f85 : SUCCESS in 7m 55s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/57f53e7560bf4198a668363c9f1fea52 : SUCCESS in 13m 58s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/f1772e6314de4eb3b9745fcca9d286b3 : SUCCESS in 9m 30s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/542122f7349c4ea3814d9a96c4230ec0 : SUCCESS in 19m 01s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/06a9f67407ce4c79aafc3f9dfa612743 : SUCCESS in 9m 14s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/75ee97a54642487ba0ea597e8b8e3946 : SUCCESS in 23m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1d1dbf2303bf4689b382299c43476e99 : SUCCESS in 14m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d09fc052dccb4680b467bbaaa9825411 : SUCCESS in 29m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f12c44d6689644de93b7ae89fe894642 : SUCCESS in 41m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72ab9fa2fae541f6b61bdb12cd77e76d : SUCCESS in 25m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ee63d7ed68f4e70b0a886d2428ce1bf : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256c5a01fb3143f4ac7d96ab3fd2ee80 : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e262781444c47a981fa67f97373b56e : FAILURE in 15m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c30761d81976928fa7db66687a4eed72d8581619","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 17:04:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"70aa81ea7a7a7911825d8adc0572dba6b71cbf74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 17:39:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ed8ccb7d145c65327eca3fa138a0125a7f33039c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 17:39:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/608dbddd28e54764988b1808a568422b : SUCCESS in 5m 26s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e75987c2c130417798ed734305493176 : SUCCESS in 9m 57s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/c05c9383372e41feb26b4b43338b846f : SUCCESS in 9m 49s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/d5cd762ebad54a688ed945be7d4d755d : SUCCESS in 8m 07s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/7eb1628df0d4440fafcd54b73e89c984 : SUCCESS in 9m 21s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/6b1a5f9fc755412fbc129adf671688ee : SUCCESS in 8m 19s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/f7a090a0cf584bfb8d6a6cb8c1c37647 : SUCCESS in 8m 58s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/7d429e7a5f754b6e827c4a421246e846 : SUCCESS in 9m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8b18dd450b71418e9d5f0313424ed5de : SUCCESS in 30m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17407d33b35d4afab69b5b647068ed75 : SUCCESS in 26m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/83b47fa7c5d94cb5b6051d251814f7cd : SUCCESS in 25m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f050dc3d15cb4fa2b0465e9c6e11691e : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcce52e89e574e4985bd3afd6e3247f6 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":3},{"id":"25240e97b90ecf2d9224cd4717305ebfe27b9f63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 17:41:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4fb939fd4e404d89f5d15925cc84ea : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9cc6d4138946034516fdf579ac084cb954ea6b06","revisions":{"b8a53b939a2bda75487a9db60b6e8f29d35dcc28":{"kind":"REWORK","_number":1,"created":"2019-09-13 16:46:59.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/682112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/682112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/682112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/682112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/682112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/682112/1"}}},"commit":{"parents":[{"commit":"31a5bc909cd08ac5121137803138766659ec76a8","subject":"Merge \"Remove invalid dict entries from hashes.pkl\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31a5bc909cd08ac5121137803138766659ec76a8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-12 17:59:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-12 18:51:09.000000000","tz":-420},"subject":"bufferedhttp: ensure query params are properly quoted","message":"bufferedhttp: ensure query params are properly quoted\n\nRecent versions of py27 [1] have begun raising InvalidURL if you try to\ninclude non-ASCII characters in the request path. This was observed\nrecently in the periodic checks of stable/ocata and stable/pike. In\nparticular, we would spin up some in-process servers in\ntest.unit.proxy.test_server.TestSocketObjectVersions and do a container\nlisting with a prefix param that included raw (unquoted) UTF-8. This\nquery string would pass unmolested through the proxy, tripping the\nInvalidURL error when bufferedhttp called putrequest.\n\nMore recent versions of Swift would not exhibit this particular failure,\nas the listing_formats middleware would force a decoding/re-encoding of\nthe query string for account and container requests. However, object\nrequests with errant query strings would likely be able to trip the same\nerror.\n\nSwift on py3 should not exhibit this behavior, as we so\nthoroughly re-write the request line to avoid hitting\nhttps://bugs.python.org/issue33973.\n\nNow, always parse and re-encode the query string in bufferedhttp. This\nprevents any errors on object requests and cleans up any callers that\nmight use bufferedhttp directly.\n\n[1] Anything after https://github.com/python/cpython/commit/bb8071a;\n    see https://bugs.python.org/issue30458\n\nCloses-Bug: 1843816\nChange-Id: I73f84b96f164e6fc5d3cb890355871c26ed271a6\nRelated-Change: Id3ce37aa0402e2d8dd5784ce329d7cb4fbaf700d\nRelated-Change: Ie648f5c04d4415f3b620fb196fa567ce7575d522\n(cherry picked from commit 49f62f6ab7fd1b833e9b5bfbcaafa4b45b592d34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8a53b939a2bda75487a9db60b6e8f29d35dcc28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8a53b939a2bda75487a9db60b6e8f29d35dcc28"}]},"branch":"refs/heads/stable/stein"},"107f1c9bd5b51a04c11e02275d1640f504741ce6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-19 16:54:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/682112/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/682112/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/682112/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/682112/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/682112/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/682112/2"}}},"commit":{"parents":[{"commit":"50b24e3715e59bc94790f97330dd1ccdc8633ae6","subject":"py2/3: Stop using stdlib\u0027s putrequest(); it only does ASCII","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50b24e3715e59bc94790f97330dd1ccdc8633ae6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-12 17:59:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-19 16:54:10.000000000","tz":-420},"subject":"bufferedhttp: ensure query params are properly quoted","message":"bufferedhttp: ensure query params are properly quoted\n\nRecent versions of py27 [1] have begun raising InvalidURL if you try to\ninclude non-ASCII characters in the request path. This was observed\nrecently in the periodic checks of stable/ocata and stable/pike. In\nparticular, we would spin up some in-process servers in\ntest.unit.proxy.test_server.TestSocketObjectVersions and do a container\nlisting with a prefix param that included raw (unquoted) UTF-8. This\nquery string would pass unmolested through the proxy, tripping the\nInvalidURL error when bufferedhttp called putrequest.\n\nMore recent versions of Swift would not exhibit this particular failure,\nas the listing_formats middleware would force a decoding/re-encoding of\nthe query string for account and container requests. However, object\nrequests with errant query strings would likely be able to trip the same\nerror.\n\nSwift on py3 should not exhibit this behavior, as we so\nthoroughly re-write the request line to avoid hitting\nhttps://bugs.python.org/issue33973.\n\nNow, always parse and re-encode the query string in bufferedhttp. This\nprevents any errors on object requests and cleans up any callers that\nmight use bufferedhttp directly.\n\n[1] Anything after https://github.com/python/cpython/commit/bb8071a;\n    see https://bugs.python.org/issue30458\n\nCloses-Bug: 1843816\nChange-Id: I73f84b96f164e6fc5d3cb890355871c26ed271a6\nRelated-Change: Id3ce37aa0402e2d8dd5784ce329d7cb4fbaf700d\nRelated-Change: Ie648f5c04d4415f3b620fb196fa567ce7575d522\n(cherry picked from commit 49f62f6ab7fd1b833e9b5bfbcaafa4b45b592d34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/107f1c9bd5b51a04c11e02275d1640f504741ce6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/107f1c9bd5b51a04c11e02275d1640f504741ce6"}]},"branch":"refs/heads/stable/stein"},"9cc6d4138946034516fdf579ac084cb954ea6b06":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-09-20 04:27:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/12/682112/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/682112/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/682112/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/682112/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/682112/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/682112/3"}}},"commit":{"parents":[{"commit":"2b4d58952cae8b174fb60529d5284c1d328e9287","subject":"py2/3: Stop using stdlib\u0027s putrequest(); it only does ASCII","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b4d58952cae8b174fb60529d5284c1d328e9287"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-12 17:59:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-20 04:26:52.000000000","tz":-420},"subject":"bufferedhttp: ensure query params are properly quoted","message":"bufferedhttp: ensure query params are properly quoted\n\nRecent versions of py27 [1] have begun raising InvalidURL if you try to\ninclude non-ASCII characters in the request path. This was observed\nrecently in the periodic checks of stable/ocata and stable/pike. In\nparticular, we would spin up some in-process servers in\ntest.unit.proxy.test_server.TestSocketObjectVersions and do a container\nlisting with a prefix param that included raw (unquoted) UTF-8. This\nquery string would pass unmolested through the proxy, tripping the\nInvalidURL error when bufferedhttp called putrequest.\n\nMore recent versions of Swift would not exhibit this particular failure,\nas the listing_formats middleware would force a decoding/re-encoding of\nthe query string for account and container requests. However, object\nrequests with errant query strings would likely be able to trip the same\nerror.\n\nSwift on py3 should not exhibit this behavior, as we so\nthoroughly re-write the request line to avoid hitting\nhttps://bugs.python.org/issue33973.\n\nNow, always parse and re-encode the query string in bufferedhttp. This\nprevents any errors on object requests and cleans up any callers that\nmight use bufferedhttp directly.\n\n[1] Anything after https://github.com/python/cpython/commit/bb8071a;\n    see https://bugs.python.org/issue30458\n\nCloses-Bug: 1843816\nChange-Id: I73f84b96f164e6fc5d3cb890355871c26ed271a6\nRelated-Change: Id3ce37aa0402e2d8dd5784ce329d7cb4fbaf700d\nRelated-Change: Ie648f5c04d4415f3b620fb196fa567ce7575d522\n(cherry picked from commit 49f62f6ab7fd1b833e9b5bfbcaafa4b45b592d34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9cc6d4138946034516fdf579ac084cb954ea6b06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9cc6d4138946034516fdf579ac084cb954ea6b06"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
