)]}'
{"id":"openstack%2Fswift~682444","triplet_id":"openstack%2Fswift~master~I4cad872ef97313ec19a8e4ef042696565a56d4c5","project":"openstack/swift","branch":"master","topic":"bug/1797305","hashtags":[],"change_id":"I4cad872ef97313ec19a8e4ef042696565a56d4c5","subject":"More multi-char-delimiter func test coverage","status":"MERGED","created":"2019-09-16 18:06:23.000000000","updated":"2020-01-22 22:52:41.000000000","submitted":"2020-01-22 22:50:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682444-1579733404018-8f9c0b7e","meta_rev_id":"5894e7ce61d9061171d506c6af05d51125769851","_number":682444,"virtual_id_number":682444,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 22:50:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-22 20:38:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-22 20:38:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 20:38:31.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2020-01-22 22:50:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0a8e6797c925f371ac4eee566b0373667e04233","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-09-16 18:06:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2abfeb784d1b36728d54677552d0460e758e8205","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 23:35:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6902a70519c94dc3aee4200dbc230f9e : SUCCESS in 5m 24s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/e704146e2e7d4963adc5fce047b94a91 : SUCCESS in 23m 34s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/3ad543f7259b43a9ac944c6554fb320a : SUCCESS in 9m 43s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/74cdd148498548ee9194010fc34d3528 : SUCCESS in 8m 58s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/b4d8507a01a8434789aaafb943b48302 : SUCCESS in 9m 49s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/5db709d36ed743ebaf6a01c33afb6f8d : SUCCESS in 13m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7effb2f14bcf4b5cb4505fd6d8829603 : SUCCESS in 11m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/55bcd459ed54433eb89bef22a1d73f3b : SUCCESS in 10m 26s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7fe723f510f3499f8b8ad4b2a414ad4c : SUCCESS in 9m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1cd855f8acb74bfa99bf5dd87176a141 : SUCCESS in 8m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/300fae03954542dd8d2a4ac744d877d6 : SUCCESS in 13m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f629f6c754584bee82a820af45523305 : SUCCESS in 16m 01s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/057b5dd5d85c40ba8370f3a7a948901f : SUCCESS in 38m 18s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/aada1307c8f14ac69e970eef21563b33 : SUCCESS in 34m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b7dc857fe3c47f790e003113265dcf0 : SUCCESS in 30m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93741c55988243df9c1344e892b9045c : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a82c4bcbdabc4332bed4c516034362ab : SUCCESS in 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"385b651fb9d027b6dcd7e3e8aa82bee533ce6e94","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-18 16:47:16.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d6888e985512ccabe647fe9d811611f6038abacc","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-09-18 17:02:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nwhen I re-spin this I\u0027ll see about adding some more tests too!\n\nfeel free to push over with more tests if you\u0027re looking at it anyway :+1:","accounts_in_message":[],"_revision_number":1},{"id":"784bb0cf39e7c59115517e4b9aa65862a7693602","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-22 20:37:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd57fe28898d095adb90a575117f2601c38062a9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-22 20:38:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"256ea92a8497ceaa0c786e23a7e9d27d60e7e9b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 21:12:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdfd84ad47cd445fb9077adbe2e1dc6e : SUCCESS in 7m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c10fc191943e4ec4a66d3dfcaf6bc41f : SUCCESS in 10m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/72b0d65c8863456c8c37cf3ab0936b36 : SUCCESS in 10m 42s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/4324a1c8533149be88805dacdf3ff6aa : SUCCESS in 11m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5a45c696054c4b1d989810224fa2305f : SUCCESS in 10m 28s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d743be8de21c4da99751f37e66a86963 : SUCCESS in 13m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/56f75af9657c48fa8aa83a94b983d743 : SUCCESS in 11m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dccedf2e303049908303a5b2bc38dd24 : SUCCESS in 10m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5b63e076b1a14071ae5256e7a7bc1847 : SUCCESS in 9m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5fc745bf0ecd4fe6be176c035c5a917d : SUCCESS in 10m 41s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d6c476a53a4042849680bef068b6d945 : SUCCESS in 12m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22c11ee0690940d68922dd4ddc852f83 : SUCCESS in 16m 45s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/227d059081a74e2bbd45488f4ff44aca : SUCCESS in 30m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b0011e55de31445fb402bd0cec4ef85a : SUCCESS in 30m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5fa85b3cd0c445cb94a9d90ae190c177 : SUCCESS in 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82b486f22f96467aaa26a68e98f258fa : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/81065f134bc945f09d00e23e656eccef : SUCCESS in 17m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4f1bf8ec892847c74ce9b75b9d10390b5177641b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 21:13:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9568eeb7fcc372a215d2e3b8e6690364f570e8b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 22:50:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e077817b637b4d8e85aac72ffc372b7c : SUCCESS in 8m 21s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d93696dd7a1545068e5b9677f21a9dcd : SUCCESS in 10m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/025c73fa8f3e44459c776f9dd13488fd : SUCCESS in 12m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aa35b78504664e8f84859c81fcb94e70 : SUCCESS in 12m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5b4ed782d5de46dc9bd7868f27d411f0 : SUCCESS in 12m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bd0e2275e98a4db8a7818b4bd3d57cc1 : SUCCESS in 14m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a95ca4e5218445c8897b325775f8ee51 : SUCCESS in 11m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/462c2db0ea0545fcbe4df4bb49227d3b : SUCCESS in 12m 24s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/da2c9b126c0d453cbab8e0ff4571b549 : SUCCESS in 10m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0cd91ad5a867455bbf26b7e70e250853 : SUCCESS in 11m 14s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/28e16174eec04d7c82fe19b62f96ccc5 : SUCCESS in 15m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/47f184b5d69f4a75af9a628b3f6743c0 : SUCCESS in 11m 51s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/846347d01bbd40e5bd5813a88183c63d : SUCCESS in 12m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bfc1064a7d7a4232bcbfbd18dc950b0b : SUCCESS in 12m 09s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/6f316912194b44149fbf9c591d4c7bc9 : SUCCESS in 15m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ffb8a03ac3db4478a747867010f9f72c : SUCCESS in 29m 04s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/220b83179d754a368ce9dc5441dc08d8 : SUCCESS in 32m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d6816fe68be04009b3ee2effbf2d9d04 : SUCCESS in 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59442aad9ade40abaec22046104ec4e5 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"01a708af64098d249ec463b12f3153ddd07d1127","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 22:50:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"acb8ad7660f7b4c8515931a436bc3af81437f69a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 22:52:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fa733affe4d4824b6659364dcc1c16e : SUCCESS in 2m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/455a08b938a04e518accc939c8b3a5d3 : SUCCESS in 1m 06s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/fd844d9f361a4b569ff6367160e5793e : SUCCESS in 1m 18s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2be4716bc9ebbdb1da69231d4618489dff8af85f","revisions":{"50fa7a13d4b3c376bad7b657ddc31eea1d35116b":{"kind":"REWORK","_number":1,"created":"2019-09-16 18:06:23.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/44/682444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/682444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/682444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/682444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/682444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/682444/1"}}},"commit":{"parents":[{"commit":"1ded0d6c8793ca3eca573c098cef78b5ae41f080","subject":"Allow arbitrary UTF-8 strings as delimiters in listings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ded0d6c8793ca3eca573c098cef78b5ae41f080"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-09-16 18:06:15.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-09-16 18:06:15.000000000","tz":-300},"subject":"More multi-char-delimiter func test coverage","message":"More multi-char-delimiter func test coverage\n\nChange-Id: I4cad872ef97313ec19a8e4ef042696565a56d4c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50fa7a13d4b3c376bad7b657ddc31eea1d35116b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50fa7a13d4b3c376bad7b657ddc31eea1d35116b"}]},"branch":"refs/heads/master"},"2be4716bc9ebbdb1da69231d4618489dff8af85f":{"kind":"REWORK","_number":2,"created":"2020-01-22 20:37:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/44/682444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/44/682444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/682444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/682444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/44/682444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/44/682444/2"}}},"commit":{"parents":[{"commit":"1ded0d6c8793ca3eca573c098cef78b5ae41f080","subject":"Allow arbitrary UTF-8 strings as delimiters in listings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1ded0d6c8793ca3eca573c098cef78b5ae41f080"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-09-16 18:06:15.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-22 20:37:21.000000000","tz":-480},"subject":"More multi-char-delimiter func test coverage","message":"More multi-char-delimiter func test coverage\n\nChange-Id: I4cad872ef97313ec19a8e4ef042696565a56d4c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2be4716bc9ebbdb1da69231d4618489dff8af85f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2be4716bc9ebbdb1da69231d4618489dff8af85f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
