)]}'
{"id":"openstack%2Fswift~682668","triplet_id":"openstack%2Fswift~master~I671a6e4a3d1011dbbc2267b44134cfaf3380fb22","project":"openstack/swift","branch":"master","topic":"skip_versioning","hashtags":[],"change_id":"I671a6e4a3d1011dbbc2267b44134cfaf3380fb22","subject":"Skip test when object versioning is not enabled","status":"MERGED","created":"2019-09-17 14:53:19.000000000","updated":"2019-09-19 00:25:56.000000000","submitted":"2019-09-19 00:23:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682668-1568852629867-1246a8dd","meta_rev_id":"c65a15072f3a1cdc8d7b39857716e8efa6bfe6d1","_number":682668,"virtual_id_number":682668,"owner":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":2,"date":"2019-09-19 00:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2019-09-18 19:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2019-09-18 19:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-18 19:45:44.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-09-19 00:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53ae7d1ca12a27c6db2eb5b5250aebc58c6ad3ab","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-09-17 14:53:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"952f391a8e731f381189393c1dfa880ea5b42f1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 00:02:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90aa53a0720142c186afaadf78232449 : SUCCESS in 4m 41s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/0662ac47778c4443b8c4a7acb4efa6c1 : SUCCESS in 8m 32s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/13e2d819fbb548a4a06a9970d94a11f2 : SUCCESS in 9m 05s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/4ef35848a591499497899debe9d56387 : SUCCESS in 11m 51s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/7a81a6460c6c454f8f9be209357bb960 : SUCCESS in 9m 42s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/fa6b571072d145a8af86b4007312b774 : SUCCESS in 12m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/318578dc9e9e4515ab52768542f3516f : SUCCESS in 9m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/975b439c909f42b29a5c4020ed2cda44 : SUCCESS in 10m 52s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/e24c3169962a49ac986c11a0598b30de : SUCCESS in 36m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ea824457603041928b0413c40a0f5a09 : SUCCESS in 12m 03s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/aeb8592d52e04127ac1670c4700f4de1 : SUCCESS in 14m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/57312fe809c2492b8d3d11654a24803a : SUCCESS in 16m 18s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bfe754a1c15d4110bb8ed8e48ccded62 : SUCCESS in 25m 39s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d8e524bc30104ef1b5badd28e07bb1ef : SUCCESS in 31m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c06ada09d34d425db7b4c5af9bbcb21d : SUCCESS in 29m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a07cba4a0e634be0b9b99657d831f486 : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88906697384948dd914bb15e68264e99 : SUCCESS in 15m 40s","accounts_in_message":[],"_revision_number":1},{"id":"93abe2ee75d8d3140543244dff06bc89cf807bd9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-09-18 19:45:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nsolid fix, master sux:\n\n\t\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\tERROR: test_slo_manifest_version (test.functional.test_versioned_writes.TestSloWithVersioning)\n\t----------------------------------------------------------------------\n\tTraceback (most recent call last):\n\t  File \"/home/vagrant/swift/test/functional/test_versioned_writes.py\", line 985, in setUp\n\t    raise ResponseError(self.conn.response)\n\tResponseError: 412: \u0027Precondition Failed\u0027 (None None) txid\u003dtx837942d5df6c465f90196-005d8271d4","accounts_in_message":[],"_revision_number":1},{"id":"dfd7761b24ad7a729b37a7450da674aad5fcef65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 19:46:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e789b6c6e22074d182201babf6699f0c26cd7a71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 21:50:59.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f17c22a9ee2c2e5ceeddc2714f41c4249e28d8bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:23:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"af0e0c922dfcd6fd067809c553a88ab7b6c6e248","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:23:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08533b3ed8e446f692cacfd4a09436cd : SUCCESS in 6m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1c055622e2d94d3ca367d75ad573718d : SUCCESS in 9m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a91d0e9c23c544aa880526fbae5a8e09 : SUCCESS in 19m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8b4c50fed87a467b8d0d3cf74c7619d0 : SUCCESS in 13m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2439f54acfc94f4ba1315b3d017fd6c9 : SUCCESS in 13m 59s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/c64697c83c1c469994f5a256fd496d52 : SUCCESS in 17m 28s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/46fa83f46daf4b11a45d0c401f08187a : SUCCESS in 17m 40s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/c45678942cdc49b3a2097a640e5b7a78 : SUCCESS in 11m 23s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/e71a284a171b4cc48ef346eecec7a1cf : SUCCESS in 11m 39s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e0055b5cb46d407e8b7a622507e5b31b : SUCCESS in 21m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/806be7f33e1f4ed386dfd1ea98bbb50d : SUCCESS in 24m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3a3a53a9d82e4638a214dceb5961ef91 : SUCCESS in 23m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d1ca4d24e4a344d99ba078a4784d792d : SUCCESS in 14m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c45fae1c147444b7bb6ea5773b11ae3c : SUCCESS in 14m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e538d77c550438d913be2827f4db1f1 : SUCCESS in 23m 31s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/380a297c74314691a13ae042e9c3d547 : SUCCESS in 27m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/227667ea59f84d9b8fb97cfab56763b4 : SUCCESS in 29m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e501bdebf547404daabd8ec51c7e2645 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7a734ba3f0207e559485c571a17bd5d60669c2c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 00:25:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a2bc4096ca48749e8d0413fd3d8936 : SUCCESS in 1m 31s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7359c08e6f2741328e06165a3dfe40db : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9698b1bb957c1f646ac30fb64ec3528627fcee1c","revisions":{"9698b1bb957c1f646ac30fb64ec3528627fcee1c":{"kind":"REWORK","_number":1,"created":"2019-09-17 14:53:19.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/68/682668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/682668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/682668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/682668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/682668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/682668/1"}}},"commit":{"parents":[{"commit":"6339bd8f6c2755c0bfd5e1769d2a0fafee04c176","subject":"Merge \"Allow arbitrary UTF-8 strings as delimiters in listings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6339bd8f6c2755c0bfd5e1769d2a0fafee04c176"}]}],"author":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-09-17 14:52:55.000000000","tz":120},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-09-17 14:52:55.000000000","tz":120},"subject":"Skip test when object versioning is not enabled","message":"Skip test when object versioning is not enabled\n\nChange-Id: I671a6e4a3d1011dbbc2267b44134cfaf3380fb22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9698b1bb957c1f646ac30fb64ec3528627fcee1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9698b1bb957c1f646ac30fb64ec3528627fcee1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
