)]}'
{"id":"openstack%2Fswift~683204","triplet_id":"openstack%2Fswift~stable%2Fstein~I1359c475fbe93db2f0fcc4b450be6dd5073f922e","project":"openstack/swift","branch":"stable/stein","topic":"bug/1843816","hashtags":[],"change_id":"I1359c475fbe93db2f0fcc4b450be6dd5073f922e","subject":"py2/3: Stop using stdlib\u0027s putrequest(); it only does ASCII","status":"MERGED","created":"2019-09-19 16:54:19.000000000","updated":"2019-09-21 05:56:56.000000000","submitted":"2019-09-21 05:54:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"683204-1569045277132-3f74366e","meta_rev_id":"d5d65ceaf8f282eefd7da183b73fa48b58bdd7bd","_number":683204,"virtual_id_number":683204,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-09-21 05:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-21 04:31:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-21 04:31:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-21 05:54:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9f7540b6ad4df999e373305ea2e3a64447e39f0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-19 16:54:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cd284dbdaed0153ad3638aaa8d4f602bad70359","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 03:32:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfdb8952447b4c61ba1062bac216c16c : SUCCESS in 5m 13s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/bfd64922afa842e5a53a82bb5792a9ee : SUCCESS in 8m 52s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/85608f0f3fcd450187a0b657aee804d6 : SUCCESS in 8m 35s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/47d652293d8f4a9194906a975ad22622 : SUCCESS in 8m 25s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/3e797ad60b3f407e9685f10b045d6c2c : SUCCESS in 9m 52s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/18fa4adb9e4b4ab9954cc8f6190e29fd : SUCCESS in 9m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65358edf3e0b446a9d21d10069c197b6 : SUCCESS in 14m 18s (non-voting)\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1b4a7981d7b45c2bc301fae0d6d2ea8 : SUCCESS in 28m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae71cea06de8433995cd58c7001f8454 : SUCCESS in 27m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3bf1aab3b1043f1ae2c7a19348b0f18 : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1acbce4977604e428dce10a17a3c4c53 : FAILURE in 17m 11s","accounts_in_message":[],"_revision_number":1},{"id":"39a3fc87e650b6a315999b1cd72e915b507c282a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-20 04:27:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b42f737798a45390d02d4689ae694f128d5262ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 12:10:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ffe460d6e0340c49d37ea2b9bc2e61d : SUCCESS in 5m 45s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28ccca7022504d66a263651a76875db2 : SUCCESS in 11m 06s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/130e7eb54af949369c36ac7aa7b2a7a1 : SUCCESS in 8m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ae90c393df4542f0942926eba14445a1 : SUCCESS in 12m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1d96717e11804622ba61ffbb565843f3 : FAILURE in 16m 22s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/3b3ae0b8a0124d65a02b77d64987f72a : FAILURE in 16m 14s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/83d76dc877194c41aaa6b69e8dd501e3 : SUCCESS in 9m 36s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/1b2c839bb2ed46d8aaee2d5ff8fcd191 : SUCCESS in 11m 25s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/1b9c2bdfcf834f80b50b8608ca6d1c2f : SUCCESS in 8m 53s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/130a768632ab428a95c0ba5c1589ea85 : SUCCESS in 18m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/55565c1cb3244b8b9ae8af00077066bc : SUCCESS in 14m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/431094c707404a75ae75c06424c18e30 : SUCCESS in 31m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e0299dba850414f8e2b98d0f1436178 : SUCCESS in 32m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d01c3cb76f8d42f9bd3b03b21b9b99fc : SUCCESS in 35m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a54c803868034e359c7f7451bb3280b2 : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25566f7e7d0d43f28d2e43426ca6b43d : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24a678141d124c75987a1226cb141992 : FAILURE in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"008773a5f018277b6c1bd974891a3198e1805898","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-20 15:55:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1a6f59514893195654ea16a979f570eaafa7e8a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 01:17:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd335a4c48b4f8587ec0b0fb60ae203 : SUCCESS in 5m 16s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8ac89b80a55444eaa4cb5de8d9d9dabf : SUCCESS in 10m 37s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/5ca23d862d8f437d9a9d06cb6c7ced59 : SUCCESS in 7m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d5e649232eb44732b3768caff9855d4d : SUCCESS in 9m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2a491f7760e34311bdbfc1c5b07a048e : SUCCESS in 8m 41s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/08763db357534b779285ae6e3d18de8c : SUCCESS in 12m 43s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/8bcc239831864cf8a6f240d77a7be92d : SUCCESS in 9m 08s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/3a00d992badb4886b7a33d1e56200441 : SUCCESS in 7m 33s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/a9cb790893464ca1a2f64c556b9b8c6f : SUCCESS in 8m 01s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/88f2c3c8f0c946f0bcd4ec128b786b56 : SUCCESS in 12m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cbe86bd1a21c4f748dc5315e05b7002f : SUCCESS in 15m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f53ebf0a4fde42919dd7bada96f34622 : SUCCESS in 32m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/580dc215a130491f96442d80b83bbb6e : SUCCESS in 26m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3c0662e361d34a31882a645c6ab47195 : SUCCESS in 27m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/941d9f12643c4e3dacd395ab13fa63ca : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/088d53ae0b2d42ab9e69bfb139f220e3 : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d4dc47993f8747ce886edffb0dd4a421 : FAILURE in 18m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1042c3d62fcc8f59b380773ec779befac8dce9c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-21 04:31:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3a0104bacc536c6b7fc4af66ed75d4f73dd24261","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 04:32:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"63699c36c17ac285db8fdcad115a5e65338aba33","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 05:54:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d4893d89f534f788aab3a082b0b49c5 : SUCCESS in 5m 33s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6eda83c96be945d0b4e557c6cf73bf53 : SUCCESS in 11m 08s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/15c4952ded634b9fb05b5d02847a3a5e : SUCCESS in 8m 48s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/3c66684af56d437da9535c22d35ad4c4 : SUCCESS in 8m 45s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/3b01919ab29c4429b2e4e0fd45b27c5d : SUCCESS in 8m 52s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/e96b4a52eca64cfc8dfbc4930939fbd2 : SUCCESS in 8m 30s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/a0ab6aa6199344938aa46cb69f88107f : SUCCESS in 9m 26s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/7922839501c6420aaa9b9807a84beabc : SUCCESS in 10m 38s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/37a3ea6ecd554f05925c299bf60b85f8 : SUCCESS in 31m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ab57dbfd8b64e7f996c6ca8069cbb64 : SUCCESS in 28m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/47e95aafb08d4ac7bd8f425a447bac42 : SUCCESS in 28m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f67828aea7e4cc680068388f92a9d2f : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be9bdc10d2b9433898f778ef547ec851 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":2},{"id":"2ee83030a68cbc3b2ac809d6ac76249ce13a0f2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 05:54:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"617bcea25499a7e0c5485189789f7280a5086bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-21 05:56:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de939a6c01034377b0d7f02d99e34c5b : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2b4d58952cae8b174fb60529d5284c1d328e9287","revisions":{"50b24e3715e59bc94790f97330dd1ccdc8633ae6":{"kind":"REWORK","_number":1,"created":"2019-09-19 16:54:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/683204/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/683204/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/683204/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/683204/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/683204/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/683204/1"}}},"commit":{"parents":[{"commit":"31a5bc909cd08ac5121137803138766659ec76a8","subject":"Merge \"Remove invalid dict entries from hashes.pkl\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31a5bc909cd08ac5121137803138766659ec76a8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-03-12 20:36:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-19 16:50:01.000000000","tz":-420},"subject":"py2/3: Stop using stdlib\u0027s putrequest(); it only does ASCII","message":"py2/3: Stop using stdlib\u0027s putrequest(); it only does ASCII\n\nNote that this only affects the functest client.\n\nSee also: https://bugs.python.org/issue36274 and\nhttps://bugs.python.org/issue38216\n\nThis was previously done just for py3 compatibility, but following\nhttps://github.com/python/cpython/commit/bb8071a our stable gates are\nall broken -- apparently, they\u0027re running a 2.7 pre-release?\n\nChange-Id: I1359c475fbe93db2f0fcc4b450be6dd5073f922e\n(cherry picked from commit c0ae48ba9aafb0b91869ea3bae8da07a32088777)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50b24e3715e59bc94790f97330dd1ccdc8633ae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50b24e3715e59bc94790f97330dd1ccdc8633ae6"}]},"branch":"refs/heads/stable/stein"},"2b4d58952cae8b174fb60529d5284c1d328e9287":{"kind":"REWORK","_number":2,"created":"2019-09-20 04:27:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/04/683204/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/04/683204/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/683204/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/683204/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/04/683204/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/04/683204/2"}}},"commit":{"parents":[{"commit":"31a5bc909cd08ac5121137803138766659ec76a8","subject":"Merge \"Remove invalid dict entries from hashes.pkl\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31a5bc909cd08ac5121137803138766659ec76a8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-03-12 20:36:21.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-20 04:26:17.000000000","tz":-420},"subject":"py2/3: Stop using stdlib\u0027s putrequest(); it only does ASCII","message":"py2/3: Stop using stdlib\u0027s putrequest(); it only does ASCII\n\nNote that this only affects the functest client.\n\nSee also: https://bugs.python.org/issue36274 and\nhttps://bugs.python.org/issue38216\n\nThis was previously done just for py3 compatibility, but following\nhttps://github.com/python/cpython/commit/bb8071a our stable gates are\nall broken -- apparently, they\u0027re running a 2.7 pre-release?\n\nNote that we have to make the rolling-upgrade jobs non-voting, as we\ncan no longer run the prior tag\u0027s functional tests.\n\nChange-Id: I1359c475fbe93db2f0fcc4b450be6dd5073f922e\n(cherry picked from commit c0ae48ba9aafb0b91869ea3bae8da07a32088777)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b4d58952cae8b174fb60529d5284c1d328e9287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b4d58952cae8b174fb60529d5284c1d328e9287"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
