)]}'
{"id":"openstack%2Fswift~685505","triplet_id":"openstack%2Fswift~master~Ie733bc50466c66d6e6eb5c6bd42e42a05ef88798","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ie733bc50466c66d6e6eb5c6bd42e42a05ef88798","subject":"Add experimental job to test upgrades from stein","status":"MERGED","created":"2019-09-27 23:19:32.000000000","updated":"2019-09-28 18:18:58.000000000","submitted":"2019-09-28 18:17:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"685505-1569694640372-01c14357","meta_rev_id":"157a030ab74b83ffe447735dd437748ed2af2eb3","_number":685505,"virtual_id_number":685505,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-28 18:17:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-28 15:53:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-28 15:53:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-28 18:17:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c68fe18cc3307695acb09963605439cd85aa3e03","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-27 23:19:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9e9c43ce5b23d362ecdbc0c5f02baf84dd04355","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-27 23:19:44.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"c81778e3e68c2dadd8205f468045c3e4eeeb198e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 23:57:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/1347a1b624474ddd9497dc8df83bd240 : SUCCESS in 13m 41s\n- swift-tox-func-centos-7 https://zuul.opendev.org/t/openstack/build/c349b6e3bc0346d1ab7518728d38dd19 : SUCCESS in 11m 53s\n- swift-tox-func-encryption-centos-7 https://zuul.opendev.org/t/openstack/build/550ebedc7d2e432cade54961abfa17d7 : SUCCESS in 14m 34s\n- swift-tox-func-ec-centos-7 https://zuul.opendev.org/t/openstack/build/5915922db9ea401d9d986c3a80b57a67 : SUCCESS in 12m 06s\n- swift-multinode-rolling-upgrade-queens https://zuul.opendev.org/t/openstack/build/b9dd4aa897e74a5d8dac148965dad393 : SUCCESS in 15m 29s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/8adc32123b834dc48b3bdcc6ab7a38dc : SUCCESS in 16m 54s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/471ada9b55ec494c8a07ddc0c80ccd9b : SUCCESS in 15m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b52223f0a14ca740b6563807bd532c25f3e524d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 00:31:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c0453d128b475d908695a0baf9b7aa : SUCCESS in 4m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/20fbe2f046aa485b8938dbe4a5c5fc22 : SUCCESS in 1h 05m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9dc36967d39425c954065b50f638ac3 : SUCCESS in 50m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f89a68cf0c834f16bcecba432a3fe164 : SUCCESS in 8m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f37f61cf7b2d4af4bdb5b67a10bc4f86 : SUCCESS in 10m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b9b7932c517f4ad385b192a84c4f21af : SUCCESS in 11m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/58eb04a7611b47029ed02b57150d8b90 : SUCCESS in 12m 42s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/b0fc21067ab641e68b259c02e4ddffdf : SUCCESS in 9m 58s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/9f6591fc151d4f818a5853d4757c14e9 : SUCCESS in 9m 15s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/b91d490dd4be4217bd8cfaaefb8b9bb6 : SUCCESS in 10m 07s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/2ee1a9cdd2a749c3ae44527cc142261a : SUCCESS in 9m 37s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/18f8e6eef1ca4d5d9135c71000822bd8 : SUCCESS in 12m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/723854d304ea488f830261359d890ba6 : SUCCESS in 9m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/87d81249a0ed4d1598f9ea34e332f62c : SUCCESS in 9m 10s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1880e2a524df44d0b073c49a84326cd2 : SUCCESS in 10m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b405939e04444a6ba56c7c9581213aad : SUCCESS in 10m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/90451a1d292b4079b89523116c8f604e : SUCCESS in 12m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3679038a49c7469baf5f98c3ddd63547 : SUCCESS in 13m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/577d3d7bc70945419ad47daf6a7d67b0 : SUCCESS in 33m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/921b37bc130d4be1a936f8a96d9f8bc2 : SUCCESS in 27m 00s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/cb45af047e50481f87ace4db6f73701d : SUCCESS in 27m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c24d2e641364fecb6320f00fb607691 : SUCCESS in 27m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b678d35939f4d1f88af0f10e154d57e : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e644cea1fcae42e3bd86ed6be968f865 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c54a1270e46e486691b49c859aaa5eb1 : SUCCESS in 15m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf522e6f281c43dfa4767869f9512195 : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e834c12be6de47593168f3eb223748a607d6d56b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 03:36:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"afe8ac12dd0fead615c98734770d759bdd0ed1b2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 03:36:30.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"6e1067d6b581d2cc9b027c5f0e200ee852d279c1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 04:09:54.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"756212bf09f4ce814b82daee472a35639564f917","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 04:11:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/1e7bb4d60c0648618bc60a79becb74b0 : SUCCESS in 14m 30s\n- swift-tox-func-centos-7 https://zuul.opendev.org/t/openstack/build/d3dddd5cad91445ab64788b02673ad13 : SUCCESS in 12m 59s\n- swift-tox-func-encryption-centos-7 https://zuul.opendev.org/t/openstack/build/fb1712ab3f75424692d67410aefff3ce : SUCCESS in 13m 09s\n- swift-tox-func-ec-centos-7 https://zuul.opendev.org/t/openstack/build/f9ddf5877b414865b0ee4db77c6ae731 : SUCCESS in 13m 11s\n- swift-multinode-rolling-upgrade-queens https://zuul.opendev.org/t/openstack/build/a627c781c288452a91525b9ecdddcff1 : SUCCESS in 18m 57s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/6a3e297522be41beab140c0e815280a1 : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":1},{"id":"86eaa4da74834b71de96f898ac213f0daeb5e722","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 04:29:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"23b17a6e78e9b114a1620900a8486388b4c5ffdf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 04:30:33.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"819914b3b240829d0d6a207689bb4e227a2ae267","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 05:05:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/c4f63353c070431799b00a4916df2982 : SUCCESS in 14m 07s\n- swift-tox-func-centos-7 https://zuul.opendev.org/t/openstack/build/0768f9422fe840c28f10b755d793b00d : SUCCESS in 12m 20s\n- swift-tox-func-encryption-centos-7 https://zuul.opendev.org/t/openstack/build/d13b8106347e48518ba3249c19eed8bb : SUCCESS in 12m 23s\n- swift-tox-func-ec-centos-7 https://zuul.opendev.org/t/openstack/build/0df154d0a97a4370a9b5cc6e973bb3a1 : SUCCESS in 13m 01s\n- swift-multinode-rolling-upgrade-queens https://zuul.opendev.org/t/openstack/build/cbc1dcc9b4244d6c924a1914b29364f4 : SUCCESS in 16m 33s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/6f71e37139104bab80caf796d1bf3166 : SUCCESS in 17m 06s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/2a0cb197c8e94a4e9548b4e5533ee920 : SUCCESS in 17m 19s","accounts_in_message":[],"_revision_number":3},{"id":"01842ddb8fecd4e7d8ba7afbed5d12ce388b1c25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 05:40:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adbb76a620624490b0d3f9eba4255765 : SUCCESS in 5m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ff2e221ee1fa4f7680bf0c4671663af2 : SUCCESS in 59m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/af8c6249f53644e982395feb166e00e9 : SUCCESS in 59m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/11083320ceb94c77bdfc87cd5cffebb7 : SUCCESS in 8m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/07e3d294c050422a85e9b56dca63efe5 : SUCCESS in 11m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ff0a4832b5464f1797cbf092f093028e : SUCCESS in 11m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4800c7dc87b94932a9bc1339221d4f61 : SUCCESS in 14m 14s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/4b7faf792d9148c18b15574f51465e16 : SUCCESS in 10m 22s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/952a1fecb73f4e86a56e3b04c06c8d61 : SUCCESS in 10m 08s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/d5f8eac389704561b4b8fcb6ca24f087 : SUCCESS in 9m 55s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/44f601460f8e42dab2fa8b058422b51e : SUCCESS in 10m 19s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/19e31ac9be0b4a9998b25f0f1ac87f2b : SUCCESS in 12m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/208382d47b45423e9f899e1877b3ea4b : SUCCESS in 9m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/05a1ec52d06647b29bbcfcd676c37992 : SUCCESS in 9m 36s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/39e99095b2e24f988e4180511e4760f6 : SUCCESS in 10m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6c515882d1644c69a0051abc6196f934 : SUCCESS in 9m 52s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5574c70ac924486391f85135a395ab9e : SUCCESS in 14m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/244cc6641ac24782a6ba7c67553a660a : SUCCESS in 11m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0302ffc1652449199eb169760cbc28a9 : SUCCESS in 32m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6da1bb9065274ce9a86adc1af36f9b09 : SUCCESS in 27m 30s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/10a36417ded1430f9ab428914409b440 : SUCCESS in 27m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b85bdde4563f41ac884aaf1c800adf68 : SUCCESS in 28m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/983e9ebb402c45d08f947e02a3ea9c0f : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df311a96ebe849039d8f7f670a5f87c6 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47ba15eb27b2407ea9ae67dd12f6af28 : SUCCESS in 15m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84a181e985204f4381be714d178527d9 : SUCCESS in 1h 05m 06s","accounts_in_message":[],"_revision_number":3},{"id":"c83316603c95880df1e5967e5ee1542f8433a6cc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 15:53:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e71f1d89cc2af21fcab4fa31d929072ce8f5a3b0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-09-28 15:53:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e05a5a8de0dbea63f8c41bd1f2068cd06210cde3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 17:04:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d46660f2be64d1bab5859d6e2b0999e : SUCCESS in 4m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4c4327804b2d4c81be6da5b5580a4b45 : SUCCESS in 1h 04m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f853b874199146e6879335826e9b165a : SUCCESS in 50m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/88c59782011b4436b9a6d3e8a093afca : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c12d6c390fc543fa9fbce2311b1b6579 : SUCCESS in 11m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/080721a514ee4fd98dceb19051782298 : SUCCESS in 11m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8b89b4df77bc42b4b4312d5651e14c06 : SUCCESS in 11m 02s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/89b2918bbbfd4a4b8cd31d643f5ae958 : SUCCESS in 8m 50s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/87c2deca4f714eccb325d9ac3334ef2a : SUCCESS in 9m 17s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/3d938ddb83b14480841db52b654afc0b : SUCCESS in 11m 37s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/71959a7b00d14237b987154aeb857b48 : SUCCESS in 13m 55s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/8e9eeedf79244f19bac541de60c2d52d : SUCCESS in 12m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cf1f760cec0940358f1b76d1f6937c75 : SUCCESS in 9m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3a4e4653593a4c98a868e713a90f3044 : SUCCESS in 11m 03s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b9601559fc244bdc861d50b7d3ff30e6 : SUCCESS in 8m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/505a699c2f454f659c0a3380af8a34fa : SUCCESS in 9m 14s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a2af684cb72248f1805a2aa3748e1bc9 : SUCCESS in 16m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f019c600f1d40b58cfa042bd85117f1 : SUCCESS in 14m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7c33ec678a254d9fab21886f7a60ac7e : SUCCESS in 32m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e641db1a86a7489eb7df8ab1615eb8fe : SUCCESS in 27m 18s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/1e325a12fbfc4929aa360682ef414dbb : SUCCESS in 31m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd5386e805824b3ab8a3b1ffb45a7c25 : SUCCESS in 26m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/11a90883511c43249d585fdfcd961445 : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e54c91966c9a4b029b54f7d4b53cec95 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3a9ebb093844424a95a4353175523018 : SUCCESS in 17m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bc2e5ec7a57430f9bdb14351a7b0280 : SUCCESS in 54m 58s","accounts_in_message":[],"_revision_number":4},{"id":"7a04616db6ead04de05fed0fe430525eadc7de52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 17:04:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9db486618c6d45170d0059de8bad3f801bc0b5b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 18:17:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9a0c16623fb593c739345c24443d9cb4f52cac63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 18:17:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f9a9c997d84658bc4eb3dff4b30b8a : SUCCESS in 6m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/17df4b2fdd624e0fa78ede5aff38e7ee : SUCCESS in 1h 02m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3013a1ccf6f45a7a71163df8001e450 : SUCCESS in 58m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/323a643d33304ae39f6fe147f5da3c92 : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cb93baab126449fd92b27c69d9c089c9 : SUCCESS in 10m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1bbe44bd368b4d28b8c1e41620872bbf : SUCCESS in 13m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2e8f54f9d6f54348afb3156e694e4238 : SUCCESS in 22m 56s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/02687873266340738de4bf72f846c3b1 : SUCCESS in 11m 14s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/17bd616594dc4602b1006f3a19eb670d : SUCCESS in 13m 15s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/11d8bb00e61947098a8397ffcde0fe0a : SUCCESS in 10m 59s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/ed935bbde5024b5aa6c864f43e687c08 : SUCCESS in 10m 27s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/d10bef0592434a3a90dbf836dbae92d6 : SUCCESS in 14m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c60e95039c8c4bb29db9d3c024347746 : SUCCESS in 13m 35s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/382c70eed74244bd8e69bf1e3aac6560 : SUCCESS in 11m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/efa23e4bb1a646fcbdeb429e77b843c4 : SUCCESS in 15m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/7ea65c3c5f6040c99999daaed2f681c8 : SUCCESS in 15m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9061d35fdf35453fb84243f221650b43 : SUCCESS in 36m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d9ab6dd53654a2989c2b3c4b1e7f6e3 : SUCCESS in 30m 21s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/9359d450966b44afb0f550d25219bda0 : SUCCESS in 34m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3a49de4b7d3b46ecad57c8d6bd2caece : SUCCESS in 30m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8140120be234b21a8b8338fdfe7a4a1 : SUCCESS in 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/418f384e86504f5481efee86f101f363 : SUCCESS in 6m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca9a834dff6e44dd9137a1367b4d4d37 : SUCCESS in 1h 08m 23s","accounts_in_message":[],"_revision_number":4},{"id":"157a030ab74b83ffe447735dd437748ed2af2eb3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-28 18:18:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed7c9cf92704abda8e09e38f1cbec76 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cb243b3814f04c82a8a12f72632dba05 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"be41721888913320bd448b8aaa4539f3ac6d4e7c","revisions":{"376ffa72b12483dae7a8dafcd409be89462a1ef0":{"kind":"REWORK","_number":1,"created":"2019-09-27 23:19:32.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/685505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/685505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/685505/1"}}},"commit":{"parents":[{"commit":"2d87ad6333c3b0b9f7403d2ceb0b282cc1628c6b","subject":"Merge \"sharder: Keep cleaving on empty shard ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d87ad6333c3b0b9f7403d2ceb0b282cc1628c6b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-27 23:18:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-27 23:19:19.000000000","tz":-420},"subject":"Add experimental job to test upgrades from stein","message":"Add experimental job to test upgrades from stein\n\nChange-Id: Ie733bc50466c66d6e6eb5c6bd42e42a05ef88798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/376ffa72b12483dae7a8dafcd409be89462a1ef0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/376ffa72b12483dae7a8dafcd409be89462a1ef0"}]},"branch":"refs/heads/master"},"3da7c648c1857f4e7e6fa85a3c9b723ebdf49dc1":{"kind":"REWORK","_number":2,"created":"2019-09-28 03:36:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/685505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/685505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/685505/2"}}},"commit":{"parents":[{"commit":"2d87ad6333c3b0b9f7403d2ceb0b282cc1628c6b","subject":"Merge \"sharder: Keep cleaving on empty shard ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d87ad6333c3b0b9f7403d2ceb0b282cc1628c6b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-27 23:18:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-28 03:35:38.000000000","tz":-420},"subject":"Add experimental job to test upgrades from stein","message":"Add experimental job to test upgrades from stein\n\nChange-Id: Ie733bc50466c66d6e6eb5c6bd42e42a05ef88798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3da7c648c1857f4e7e6fa85a3c9b723ebdf49dc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3da7c648c1857f4e7e6fa85a3c9b723ebdf49dc1"}]},"branch":"refs/heads/master"},"69b22e07870f298ffb13fa7bf25da52eea93ef51":{"kind":"REWORK","_number":3,"created":"2019-09-28 04:29:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/685505/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/685505/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/685505/3"}}},"commit":{"parents":[{"commit":"2d87ad6333c3b0b9f7403d2ceb0b282cc1628c6b","subject":"Merge \"sharder: Keep cleaving on empty shard ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d87ad6333c3b0b9f7403d2ceb0b282cc1628c6b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-27 23:18:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-28 04:28:23.000000000","tz":-420},"subject":"Add experimental job to test upgrades from stein","message":"Add experimental job to test upgrades from stein\n\nAlso, correct the version that we check out when upgrading from stable\nbranches.\n\nChange-Id: Ie733bc50466c66d6e6eb5c6bd42e42a05ef88798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69b22e07870f298ffb13fa7bf25da52eea93ef51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69b22e07870f298ffb13fa7bf25da52eea93ef51"}]},"branch":"refs/heads/master"},"be41721888913320bd448b8aaa4539f3ac6d4e7c":{"kind":"REWORK","_number":4,"created":"2019-09-28 15:53:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/05/685505/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/685505/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/685505/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/685505/4"}}},"commit":{"parents":[{"commit":"2d87ad6333c3b0b9f7403d2ceb0b282cc1628c6b","subject":"Merge \"sharder: Keep cleaving on empty shard ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2d87ad6333c3b0b9f7403d2ceb0b282cc1628c6b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-27 23:18:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-09-28 15:53:07.000000000","tz":-420},"subject":"Add experimental job to test upgrades from stein","message":"Add experimental job to test upgrades from stein\n\nAlso, correct the version that we check out when upgrading from stable\nbranches.\n\nChange-Id: Ie733bc50466c66d6e6eb5c6bd42e42a05ef88798\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be41721888913320bd448b8aaa4539f3ac6d4e7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be41721888913320bd448b8aaa4539f3ac6d4e7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
