)]}'
{"id":"openstack%2Fswift~686250","triplet_id":"openstack%2Fswift~master~Ibcbacef4e8ab38e6443c3442484f394f8d0b3796","project":"openstack/swift","branch":"master","topic":"bug/1840507","hashtags":[],"change_id":"Ibcbacef4e8ab38e6443c3442484f394f8d0b3796","subject":"DNM: Frustrated rant about OOO under a shifting py2/3 base","status":"NEW","created":"2019-10-02 19:03:26.000000000","updated":"2019-10-02 21:16:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"291e35479eb2960a76ce318ffd9b587551d6d041","_number":686250,"virtual_id_number":686250,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-02 21:16:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 21:16:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ec61246006e1386ec8c1c182c9953a80c57cb431","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-10-02 19:03:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca2feacf96112d58a781706d303169cb8536c6d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 21:16:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246bbc8f74ff4284a2a874d2f049c9ba : FAILURE in 4m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0502f48dbb2c4fa7a373c819793b0fa0 : SUCCESS in 1h 50m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04a9d1003afb420190b87fdf260b8182 : SUCCESS in 57m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/671798e6dc1c44a8be32c75e964b860b : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cdc69243d30e41cf87ff703c467a6704 : SUCCESS in 11m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0aa3e61cab6f47b893b69f468e4512a4 : SUCCESS in 11m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5bd9ae6c25b44956b96357538149ff30 : SUCCESS in 24m 52s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/f4288a166f19480db296e8bff7e18975 : SUCCESS in 11m 58s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/4f7f7a9be63f4583be317c82942ccb1f : SUCCESS in 22m 23s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/aec6f29ef58e4bc2924e35096eecefc4 : SUCCESS in 10m 06s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/80458fb96d2b4c38bc96077d4833931f : SUCCESS in 18m 33s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/5ffc4ce9f9e54765bca98c746646f512 : FAILURE in 11m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0a385576050247f2a60d9353ccdae64d : SUCCESS in 12m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4352e0261d9c447aac27ccd646213310 : SUCCESS in 13m 13s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/196747aa380f4259bc45317de4dce4c4 : SUCCESS in 19m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f780c93903c948cb822703babf32e8aa : SUCCESS in 11m 30s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/875cdb7654a04a1bbbd8b52e618eb52d : SUCCESS in 13m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/527188357d3f441ab036562ff6ae7727 : FAILURE in 14m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7ce21e629ed543b88e5a42162d93a307 : SUCCESS in 39m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e7cfe4cd070444194e93297b65bab8f : SUCCESS in 29m 09s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b0f44a5563474a61a959cde3bf39b621 : SUCCESS in 33m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/241dd590d5314301a304c07cc252ea76 : SUCCESS in 48m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d313d89fb2b4ef1b1887053d2187a8c : SUCCESS in 23m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f911d2ee48b4e12b734f4f52dfafb92 : SUCCESS in 9m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39a99398c3ce4dd3969bc0530360cd2f : SUCCESS in 17m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5102fa35d65a4aea806edbea1fc73db8 : SUCCESS in 1h 13m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d05840fc8b4c3f58f830ef5be2859a1f3b50a1d","revisions":{"3d05840fc8b4c3f58f830ef5be2859a1f3b50a1d":{"kind":"REWORK","_number":1,"created":"2019-10-02 19:03:26.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/50/686250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/686250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/686250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/686250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/686250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/686250/1"}}},"commit":{"parents":[{"commit":"bf9346d88de2aeb06da3b2cde62ffa6200936367","subject":"Fix some request-smuggling vectors on py3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf9346d88de2aeb06da3b2cde62ffa6200936367"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-10-02 19:02:47.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2019-10-02 19:02:47.000000000","tz":-300},"subject":"DNM: Frustrated rant about OOO under a shifting py2/3 base","message":"DNM: Frustrated rant about OOO under a shifting py2/3 base\n\nChange-Id: Ibcbacef4e8ab38e6443c3442484f394f8d0b3796\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d05840fc8b4c3f58f830ef5be2859a1f3b50a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d05840fc8b4c3f58f830ef5be2859a1f3b50a1d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
