)]}'
{"id":"openstack%2Fswift~686438","triplet_id":"openstack%2Fswift~stable%2Ftrain~If576e03ebe8e3bd392bae706e1d7b9af86f2fc17","project":"openstack/swift","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"If576e03ebe8e3bd392bae706e1d7b9af86f2fc17","subject":"Update .gitreview for stable/train","status":"MERGED","created":"2019-10-03 16:35:30.000000000","updated":"2019-10-03 19:46:17.000000000","submitted":"2019-10-03 19:44:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686438-1570131887714-39d839bb","meta_rev_id":"b6078861537954eccfd2c209c7f06d5758af179e","_number":686438,"virtual_id_number":686438,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-03 19:44:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-03 18:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-03 18:17:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-03 18:17:42.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-10-03 19:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a8c77eeb382af907411802952da83c2e12adfb4","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-10-03 16:35:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5c0e7e35830d2cc288260f649f5435228521c7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 17:48:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1b559128cc54833a8b67ff1315b4912 : SUCCESS in 4m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1fc45c2d46db472db3d7f2bbb90cb1a3 : SUCCESS in 11m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cd51ab3f0781403b9ed9b62872cf4f94 : SUCCESS in 25m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/562048ec802940069ab70a63ee746c74 : SUCCESS in 13m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d09447038107419c88c3661b72b40ec9 : SUCCESS in 14m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61887e1fe6cd43d0b10057ea41430aad : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a7da5977c2478384a8287a69f0d460 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"efdccef1cf3756af126d4023b7dd1aedde8434a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-03 18:17:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8efbba65aeea80d2b84df6be742df4b8eeb39d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 18:18:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4a9d724d3cbac13cdbc26696c948a5274e7d446","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:44:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef838d5e5ea6470abee7973a57793b1a : SUCCESS in 6m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c209525b5a7244f5b3bce41596c914dd : SUCCESS in 9m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e25743b2b4a24107af11852f2806a782 : SUCCESS in 14m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d27822b6fd6741219ba3f61ed6254858 : SUCCESS in 14m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1ec8bc8ccd40482ba8531710177c1bb0 : SUCCESS in 12m 44s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/6d4e76a4ed1a4d1c963d5e6be1de0aef : SUCCESS in 9m 40s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/0bf78b214d5c4d7bbfcd0d640866f889 : SUCCESS in 11m 32s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/316f46ffe7bb498d833c56ae64f16c89 : SUCCESS in 11m 14s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/bb29c80bc29d424baa05d7ec19310b50 : SUCCESS in 12m 13s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e15e56953df0435f932f252b8668ca43 : SUCCESS in 14m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ab683132faab43f4864fe3547b45bbcd : SUCCESS in 11m 48s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/debf5a192e7649ad898b0bc59606614d : SUCCESS in 11m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d38c4a2229ce44efa6bfad8acf2c7049 : SUCCESS in 11m 03s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a99615a27fda4534b45d97895e9f052a : SUCCESS in 14m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/436552c4ad964c4ea39471048739e602 : SUCCESS in 19m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f61f97a3025412693f7192010127e4b : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3a311938342a0caa5dba3a1956e907db8f18d20a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:44:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1fe27f65c1d201c14cea51236a7528132f18e56b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:46:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14674420635f436f8de0b510a2e5490b : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/eddb63f41bfd4ccba0a493388b2a4236 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff1f99a8636b8d75f3377566a756b98a2e7149e3","revisions":{"ff1f99a8636b8d75f3377566a756b98a2e7149e3":{"kind":"REWORK","_number":1,"created":"2019-10-03 16:35:30.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/38/686438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/686438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/686438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/686438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/686438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/686438/1"}}},"commit":{"parents":[{"commit":"8a4becb12fbe3d4988ddee73536673d6f55682dd","subject":"Authors/changelog for 2.23.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a4becb12fbe3d4988ddee73536673d6f55682dd"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-03 16:35:24.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-03 16:35:24.000000000","tz":0},"subject":"Update .gitreview for stable/train","message":"Update .gitreview for stable/train\n\nChange-Id: If576e03ebe8e3bd392bae706e1d7b9af86f2fc17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff1f99a8636b8d75f3377566a756b98a2e7149e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff1f99a8636b8d75f3377566a756b98a2e7149e3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
