)]}'
{"id":"openstack%2Fswift~686527","triplet_id":"openstack%2Fswift~stable%2Ftrain~I719c9a8cba92ca2625425c04675502712ac109cd","project":"openstack/swift","branch":"stable/train","topic":"zanata/translations","hashtags":[],"change_id":"I719c9a8cba92ca2625425c04675502712ac109cd","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-10-04 07:04:40.000000000","updated":"2019-10-05 17:42:49.000000000","submitted":"2019-10-05 17:41:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686527-1570297284410-0f5a3271","meta_rev_id":"734e964a470aea0376d97e96a4d458fb7a45b3e6","_number":686527,"virtual_id_number":686527,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-05 14:52:39.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-05 17:41:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-04 15:33:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-04 15:33:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 15:33:14.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-10-05 14:52:39.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-10-05 17:41:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8deee1ec9526d6d3d7e75ddd5119a6052ebb2d15","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-04 07:04:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ed095f9bb2bec4bf98d996247f2cee8694d5d08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 07:46:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f52bb5e74b3e4073b871d9cc15d36aba : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06619b3ace3841e5ba2f9861b529f004 : SUCCESS in 8m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dc8ad0ac642a4acb8961a70fccde7ffb : SUCCESS in 9m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/08a635ca1a394235bf347c912f16486c : SUCCESS in 11m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9727117b46964b4799657f69195f57a2 : TIMED_OUT in 40m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/71be8d545c784fe7889f1654c138f9d1 : SUCCESS in 11m 13s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/df12384815fb4d95b163a2245a872bf2 : SUCCESS in 12m 02s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/e0b01e5aaa364a4fb719c79e24977e46 : SUCCESS in 19m 30s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/1a18fbbf2cc24b65abf4a7016ad76ade : SUCCESS in 9m 38s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/5fdf94bc9ca0428d96d47b8b16542d2c : SUCCESS in 19m 33s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/71b544caadd248bc9b905026da9a8d9d : SUCCESS in 12m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7a94e0fc270a47e1ab7fac9ca1effba3 : SUCCESS in 27m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8db1801b2f2444fcbe2165ba7aaf535b : SUCCESS in 10m 50s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d3de54d998fb4ac1afea0b7aac2d36c5 : SUCCESS in 22m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bb1c5d3e926e44b8937970fda6d70b5b : SUCCESS in 10m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/70df669c54c34821bd7403e143fe83b2 : SUCCESS in 13m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ca8bb4f7b1b4d68aa42452de96675c8 : SUCCESS in 16m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a2a21163a87645beb1e523ffbe24a362 : SUCCESS in 33m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58e88ce13a8c4b6d9aed880372bc4e8e : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3bb8e655af6476d8ebd131d3c32ab9e : SUCCESS in 15m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/57599dd648ca4d1c9a5d21e5172fad5b : SUCCESS in 18m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e05c2c1e08d6b4889305634daf224feab384a3ab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-04 07:57:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a82d7dece3ad5e2843f46e14163f8a9bbf2ba086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 08:39:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b80832a5d62f4e4d91c69a777b800a59 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b459932eee1041f3a29d501518abf5e7 : SUCCESS in 9m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e80844c645f4469b99b7603c229afb53 : SUCCESS in 9m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d461e273e2a94e4ca08f9f29bc5c64f0 : SUCCESS in 11m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03c065224e2840198a460da32ade1f48 : SUCCESS in 23m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b59e0aeccf7a4f2db56828779eab3718 : SUCCESS in 24m 10s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/3e92a38772af43efb45ba6d5c464098f : SUCCESS in 9m 48s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/453f2dc297f94de99f5fb0c2d9c1bbdb : SUCCESS in 22m 17s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/11c17ec7d9d047ea8255f1ea1b0b1f39 : SUCCESS in 13m 56s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/13f48498889a4d88a436de94ab651dd9 : SUCCESS in 23m 18s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/594746d206174177b6b12da177510b3f : SUCCESS in 14m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f215733805c54643b20b3e816df22e50 : SUCCESS in 23m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2c7d82198d1f4a53a372972b0a9f9ae4 : SUCCESS in 10m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/15436be8362f415dbc3a4f2f3ca5ee8b : SUCCESS in 19m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/92903cd7bebb4b14b585d87a154010db : SUCCESS in 24m 02s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/79f4948e01bd4ab8a0d9814b33417d53 : SUCCESS in 12m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f69bde4f1417492da830d611f5b9706e : SUCCESS in 16m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1b68c5c994044e6ba0f8eb6e5c346167 : FAILURE in 40m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6da9d178ded345d4840e3d1eca661ec5 : SUCCESS in 19m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7b6cdef61194bcb84b43594fb1b2b34 : SUCCESS in 7m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a73c88b70bfd496698f645240702a52c : SUCCESS in 24m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7a4abe920ad6fa83d497696e402d99668f2e4e2a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-04 09:42:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f2150a718bab7301aa4e34c7d67c138b9bf6c11f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 10:16:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b513a09e41b4ed18b76dcc91cfc9892 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de136bc6c2414a42bb639f5e9164d6b2 : SUCCESS in 9m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ae222c30b5a4cb194d8b65432352eb0 : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/079c8e7e6a024217a905ba4886e3507c : SUCCESS in 16m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4580cf8517a9429ab01a0662bd0443a4 : SUCCESS in 11m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d2861d1e3556415ba3b00e22c90d9e74 : SUCCESS in 12m 30s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/4671dd1243994ffd9bb36b5826f6e449 : SUCCESS in 10m 20s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/46505640d88f4656b3954a4c92791b1f : SUCCESS in 19m 16s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/7a33539c6f1b4617b15b561c2d8edf6f : SUCCESS in 20m 20s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/a46e2ff02a464f1081950391e7807747 : SUCCESS in 11m 39s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/7e4bb7027fae4f5eaec15809d1e989f8 : SUCCESS in 12m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dac4e2c8e6204db9a82a81e1831b2baf : SUCCESS in 19m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/562b0cf025db4c9cbaf12c70bdb77e72 : SUCCESS in 10m 46s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b3accb01808a4e7e9192467e833f362c : SUCCESS in 20m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4ae08c36e2bf4cd395f966e5c5d5b417 : SUCCESS in 11m 20s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/a061521442e942fdb855faf3a785f30a : SUCCESS in 13m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b62036daca3f4056b6a04f1c511d1b1e : SUCCESS in 23m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6d65c6cbeb094f34ac4e1232ae3e2d7b : SUCCESS in 30m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b39053116bc40b4bd9c95f236f5eb7a : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0a4c2614ca54fbda176efadcc6c3460 : SUCCESS in 7m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/90cc8f1bd17443c49ccb38f5bf7344ba : SUCCESS in 20m 12s","accounts_in_message":[],"_revision_number":1},{"id":"fbde2b1d62e7afa2efc155545c1d9663ea8c759c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-04 15:33:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a7631b80916211a76d558d38fd39155bfccd0c31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 15:33:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f2962d985ba3e8f22bf4edb7376e843c71c6a87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 19:53:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a8c43ff961e43fd98f9fa61dee6d235 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/148a50e5fdb44df9b8d152fb0b678589 : SUCCESS in 9m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/94ed59c506204be897d7e1ffe41a98f8 : SUCCESS in 9m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/573b243ed7934950b7cab1cbadbadd6c : SUCCESS in 12m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/26881b78596248bd8e4f5236e4cfa598 : SUCCESS in 15m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a98f6627b72149b7875b7c75b401b98d : SUCCESS in 12m 31s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/7883fb96efbd473f8d8be6e0dbc9b4de : SUCCESS in 9m 39s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/bf43a216af564efe85b39bf9fec73246 : SUCCESS in 10m 32s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/9233461fb7774cbb9404dcd146a0bd48 : SUCCESS in 12m 23s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/9354f52835ff44ea93ef5d5ba19ee446 : SUCCESS in 10m 42s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/53cd84b6b6bb4e9890549dd350fc1533 : SUCCESS in 14m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a1e663b2703640178c618f1b1f5dc71d : SUCCESS in 9m 45s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/494f100e50c746c69ecb98aaa6fc1b56 : SUCCESS in 12m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/09b27cd98e684ead9f024a90551736b0 : SUCCESS in 11m 16s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ca146bb21abd47eeb2f9b9ff88bf8f37 : SUCCESS in 16m 39s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ae5bf16dab94442ca3be428a589a5562 : FAILURE in 38m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/529f3c8562c94f07838e547cdecb67c5 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a876308c9e3470ea45f950740dd3985 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"e442255b276bd473a5ef0c7c2c3f3bf7f717afaa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-04 21:28:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c61c17f932c3180d86fa2b633bbe1db61a83d99c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 23:34:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5887b4bdab4f1ab18e609d52374bef : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60eeb001e5904f78ac9634ddd986136a : SUCCESS in 8m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3f8e28865a24c918415cf8d01afa050 : SUCCESS in 9m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5cde7c21ec6f422a990ec3c86939058d : SUCCESS in 12m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a493132e64c643fa90c9876093a473f3 : SUCCESS in 11m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9d55d74ec1e24a0db104efae2e29c7de : SUCCESS in 18m 27s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/322921e0e0c84ab086b6a9ab77afa090 : SUCCESS in 9m 45s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/4a49039afad0413d8c0b43983613228e : SUCCESS in 10m 35s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/d21f9a0dc1f64c189b828fd73f09e59c : SUCCESS in 12m 10s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/176cf590760247b7862192a624a82e57 : SUCCESS in 9m 09s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/56ab20fa8220418a975965449e65e9a0 : SUCCESS in 12m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/38de9b9ced0546048bd7d973a00c4081 : SUCCESS in 9m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c447047e940b4adbb4c390db54afc995 : SUCCESS in 9m 44s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/30db97f543604f0b8e2dbda87bea3737 : SUCCESS in 9m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f3a33a855eff404daf0867cba045c6c0 : SUCCESS in 10m 25s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/4e149c4afaa440ab984c864398546697 : SUCCESS in 15m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/33bfadd539984456bfa53eff9fea425b : SUCCESS in 14m 38s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4b3943c448c2447ab005c45d98656899 : FAILURE in 36m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f65d59c985a4099a07a22d8e77204c3 : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9c7d71f1fa04a99a8118ae56fcbdbcd : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f1a8f6e3e9e46c890728669c6ee6e24 : SUCCESS in 16m 57s","accounts_in_message":[],"_revision_number":1},{"id":"6c82dc2365378c0b8a7a7065f821519ae50a96a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-05 02:25:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f37edb26cda8480ad0d8e2581b150e64cde86ae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 02:59:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a20f8a11cc46449b6cfc0ca6a75ddb : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/065347798fb7410a8053cc3fe1f2fff7 : SUCCESS in 6m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ece24e06956a4ba7ae043123543e6267 : SUCCESS in 7m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7d5d7a5eb54d4d47be2ff095d5c7713a : SUCCESS in 14m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/191b39c773e642eb93d8b0b041f3f5d3 : SUCCESS in 17m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7a2a64af8b7e4987a0ae0caeac838ee0 : SUCCESS in 13m 02s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/1415eecee6d8401d8af6b266ea21f5cc : SUCCESS in 10m 50s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/202165ceee8f420b95bb93ca116724cd : SUCCESS in 12m 17s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/84405baf395c46bdb0df8154ce779c83 : SUCCESS in 8m 46s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/60004b2ba3f84816b69c5efb842be6e2 : SUCCESS in 15m 51s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/fc65b09a22ee4d2589b7c577d6e9322e : SUCCESS in 14m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e9a34e8b061447ce87c5905eb44ff9c1 : SUCCESS in 9m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0d0ef1d62c784061ae9093a7c07bc49b : SUCCESS in 16m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/842fb5aa79094cb8b6a1fc0896aab0c9 : SUCCESS in 11m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/34ecb0784e9c4a48ac49c28455cdc217 : SUCCESS in 16m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/2aa4f83ca3694f7e832c36982ec33513 : SUCCESS in 14m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df48b79874fd462eb1b2c7febfc63d2b : SUCCESS in 19m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c2492f3e577f4358a1249723d8b0637f : FAILURE in 31m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a5c43f0ebed499bae780192849e37ae : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a952c8f21a4d44188c179211484c9556 : SUCCESS in 6m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9cb92f76a55542aa8d900039d37fbaa2 : SUCCESS in 22m 33s","accounts_in_message":[],"_revision_number":1},{"id":"0b8f3a88ecba349a7fb1050050970f935b301638","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-05 04:02:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67d31ab45d14be6931dc9234e40b7b30322d7741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 04:35:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739c76e17d58455aaf5ef0f9924c71d6 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1e9eebf3a734db1afe87637153f33d1 : SUCCESS in 7m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae14e8e2a3f54ddd85213da3a52d9c85 : SUCCESS in 9m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a45342f951546028ef4e84ef5d25594 : SUCCESS in 16m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe6d6a31d58f4b15b5f482ac6aadbb33 : SUCCESS in 12m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f07a768436544bdeb401c8d73544b379 : SUCCESS in 12m 20s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/8f61a61bc31949f0a1056e328c7e15f1 : SUCCESS in 16m 16s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/e44156c546e941799a34681f2b764e92 : SUCCESS in 14m 42s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/1c3a45bdd55f4907b1759ebf2dd3bbc2 : SUCCESS in 16m 59s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/a487e8393474485c8c66dbd751e1d7ea : SUCCESS in 15m 15s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/1c4fcd56331c4b3682e55e4283e22b1c : SUCCESS in 12m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/20fb60fd9dd84aabb2e93866db230818 : SUCCESS in 9m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/68fb6938378f4d1eacff984827832925 : SUCCESS in 17m 02s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1cf35ddb73cc4563bac07d6bb45462f5 : SUCCESS in 10m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9c2a522e40e14bb5b7673ed5ec85675c : SUCCESS in 10m 21s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/9006b1893dca477ca7e90d53fc9c7883 : SUCCESS in 18m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fc806ab133cf40359acc910501b39254 : SUCCESS in 13m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aa7ed32578aa4a69a9795685440ff57d : SUCCESS in 28m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5cbb45399f084145954190dc78cea165 : SUCCESS in 18m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c73e63a10946aebe64b98f0fe1461c : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f1f57d08496144af847e39a9e806b249 : SUCCESS in 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"4bf03c68368810c3ed4d3f5f330407f511d15afa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 04:35:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e0a1e0f1114b3167ad22d43afaf520781c1db7ff","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-05 04:55:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cc247bf856cfc480284ddbdb8148ebe3d882a876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 05:30:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32750ff46a384012aff27392a3d545e0 : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32d40e5ed06a4cd0b07b37d916202bff : SUCCESS in 7m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1af1ef035a5540eaa0b0f8d336498097 : SUCCESS in 9m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/54e476cfc4d34b27a2aab8a2c83aadce : SUCCESS in 13m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/04641e69640e4cde93e3632a0b7cb55c : SUCCESS in 21m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b3fe82adc7a84b988089e66018fa91df : SUCCESS in 19m 41s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/8855439c2fb940458c66629c84dff586 : SUCCESS in 9m 03s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/f7b21666e1f144cebf392f8645754523 : SUCCESS in 11m 53s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/da750c81024149bdb8a5526a768cd774 : SUCCESS in 9m 51s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/fdaf98c8f7e44130a7ee9853be0ddeb8 : SUCCESS in 21m 10s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/946b80e0f43144a3961c3b63de72078e : SUCCESS in 21m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2a58ac7665d54291b51e9b46980c013b : SUCCESS in 12m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e47788c4290d4658a40074588a42795c : SUCCESS in 9m 39s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/bdb5e5861de14a6e831327aaf5bb202a : SUCCESS in 11m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/278dd86ea8f4491687109845fdf9cd64 : SUCCESS in 21m 29s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/45e35ae67201477db20d0db08383c288 : SUCCESS in 12m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fefb40eb55834d6193d2dbc35bd2f919 : SUCCESS in 15m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1b7a74c7762c4910aa6376279122d00e : FAILURE in 32m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30790721f2a84914b72983b9664245fe : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3d1daa33134d37abe92b5a1dfc09e9 : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/73bc24290e984bdda1cb895280fe5228 : SUCCESS in 29m 42s","accounts_in_message":[],"_revision_number":1},{"id":"795ad3727cc4cbcb05eea01d34c6344893b639e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 05:33:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c6f34462a0466fa12f7690d4add9d2 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4522b72eb9e94492a1aa48aff30e250b : SUCCESS in 6m 30s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/83fe748fc75240009f5ad044b99c524a : SUCCESS in 8m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/874aa91dfb004b9b9796454c1ee29110 : FAILURE in 14m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3b622ae42f4c400090d268e8db1e20d9 : SUCCESS in 14m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ec47399774944bde93ec0878c2323aed : SUCCESS in 14m 14s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/4a8652be4ffd418b9dff3aad7786c26b : SUCCESS in 8m 38s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/d2730e09566b433b90bebc21b782d451 : SUCCESS in 9m 38s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/aeb447b060c741e4862e8858858f5a90 : SUCCESS in 9m 36s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/86d1ef047b1342d99923e41cf7e35096 : SUCCESS in 9m 14s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/06d3228cb91948e08c16f92e84dea531 : SUCCESS in 13m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e7e15a5e641044fb8f19f874ce0884cb : SUCCESS in 8m 53s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/ae64285b5c1a4da783086c4e08ba3715 : SUCCESS in 8m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/523f8f705d1b4279950a3282f9dad794 : SUCCESS in 9m 53s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/75edb38b03c442d9966b947644c4cd07 : SUCCESS in 12m 37s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f3db558f9a2e4eb1a9588176bed2ddd0 : SUCCESS in 28m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f67b9feebf374ef5b43f086ac953ebe8 : SUCCESS in 12m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89447cb15e20458f81817cfe64a09301 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"e4fbff86b2c3bc07810a887e5c9b2e74428c101d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-05 07:06:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"49eea1ae6455d7f78e93f36e1ed80ce0efcf4c8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 07:44:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dd38a8b19384bf59301208d594a3238 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88076cf65fcd407d86074b2ed9b543a6 : SUCCESS in 7m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cc7ddf02a19747d18a5f46cad357700b : SUCCESS in 9m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/71455e8b8e054d4fb4cfcb74a7aa2902 : SUCCESS in 10m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ad237362879d47aaa31c7aa0f7322c0f : SUCCESS in 11m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4c8106765ca14d75bd81c2ca012ecbab : SUCCESS in 11m 34s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/6ae7f53f3e7e4ba9ab1350c0c2262fde : SUCCESS in 8m 47s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/bb0b21b65ada405c832efb196708d4a0 : SUCCESS in 9m 07s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/932ddeafeae64e49a293124d44cbc33b : SUCCESS in 8m 59s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/30b7498910f84698858a7a6b797376ef : SUCCESS in 9m 45s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/0f21285c18c64c658a6906f713891143 : SUCCESS in 12m 16s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/54e8c83e33544cc99341b26327c61317 : SUCCESS in 9m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f3b1166c6c9b465e9675738765b36f71 : SUCCESS in 9m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/867b8d36b9cb42a584dfe71a3f165875 : SUCCESS in 9m 21s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bbc671c69697432dabf1f353b1de5791 : SUCCESS in 10m 10s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/42479cb51300462bb1f955d05f8d3a64 : SUCCESS in 12m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/624f3f00a6794eecac4c15ccddcdd32d : SUCCESS in 13m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/536fd4d7d431497fb2459cba8e702f4c : SUCCESS in 29m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/008c85393d68481bb12c5819d36cd1c5 : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7876df57843446883f5806b5daaa0f9 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/07479d7baf3a4b2ca16e44c5c7b3e045 : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b19b211106e215e1ee18eab2a8f91045caeef1cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 07:44:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"101595d32c52fc83b5dc6f83430cbefbc3fc0797","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 08:18:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a9868a992c1427284e6cb7a45c649a0 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb9b98246e5940e9abc4c7c1854a2233 : SUCCESS in 7m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/41e6816e93864b26b4842a25c91b6eb5 : SUCCESS in 9m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f233400a685441d5bffce45f8c3efc6e : SUCCESS in 10m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/18a5e6ce8517404d8b9ef8c10e7cfe9e : SUCCESS in 17m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9b929f65209947a1b6576d2df3140e59 : SUCCESS in 11m 28s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/a8274cf246ab418dadff0550cd33e9b5 : SUCCESS in 12m 53s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/db85da57f21043dfb5374c157d32e0cd : SUCCESS in 16m 43s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/5466933d7f5e4a4ebefdf47decc1e080 : SUCCESS in 11m 58s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/a4bb616030284d84b05e0ab68f376492 : SUCCESS in 9m 55s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e370da925c0c4e32a99b57625a894673 : SUCCESS in 11m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fe5a9f7ff8944294863798db2f0ea7b9 : SUCCESS in 13m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/78cc46a3ef464fa880d8e7b4143fdeaa : SUCCESS in 16m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cde60abf4f344fbe8de7a7a480dfcb44 : SUCCESS in 11m 30s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/86c7c6d926a94cdcb7ddb6385df8543d : SUCCESS in 12m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4dece7fdf9e547f4a583e0d97d6aba66 : FAILURE in 31m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3925bb4344c44e78a925dd9ecf8fbec : SUCCESS in 18m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d48041ca546443e8902ff6981cc235b3 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"95f80e76ee604a67c4848b2d0ee3a98d77fd8210","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-05 10:35:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b030516bc62d43046d2380c239c99a112c93e77b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 11:19:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2429332c660b4823a4223eb1c51284dc : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a07be1733cd4d919e4233e51ec69fd8 : SUCCESS in 7m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4b256b8540946cda2fa0d9b62b256e2 : SUCCESS in 10m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f9c4fe0f23764214830697a8c8ce1404 : SUCCESS in 13m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c98b0a7a58fc4f87b617b9cf1affb8ac : SUCCESS in 11m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/841d89a9186f45a1a1416f47bd9762a9 : SUCCESS in 22m 42s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/5c3a45a92d384984b6402c3e74fc7036 : SUCCESS in 11m 13s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/923fd49ff17441e6981560d24fa6fa72 : SUCCESS in 21m 31s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/1bc58ff86aeb43b0886d660097c0bdcf : SUCCESS in 21m 15s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/b0304adc46d14097a03a63c18691ab62 : SUCCESS in 22m 58s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/eaf9bc032bb440c0abac58992440836d : SUCCESS in 12m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bb987e6018dc449d90cd8bc0b077e3ff : FAILURE in 13m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/484a5e03dbc243cb911c33ed975c4627 : SUCCESS in 21m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9fa553bf9f1543f2b0603c241044ba03 : SUCCESS in 9m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1995072b08df49b8abe3121573be3aa4 : SUCCESS in 10m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/0dcc4daf93e44f9eb70d1b51269dfac6 : SUCCESS in 12m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1031bc58e284224bce4dbd6c0a8685c : SUCCESS in 17m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/88a3d2ca71f444fda3ad8f6d74c2bba4 : FAILURE in 31m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea57dc810ff54fe9b91539c327a5971d : SUCCESS in 23m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7a499d32514ae989db9477363a26cc : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9420df297b024563b2eb122138f42ba5 : SUCCESS in 23m 24s","accounts_in_message":[],"_revision_number":1},{"id":"53154e6db1e12abba3f49fdf533175a1daf63370","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-05 12:51:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e2cd8e71ac4caa9a3c4ae3c6eece59e7799818f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 13:37:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59ed0e811efc4d03905484f5b654f24d : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da087f7d4ad243ffa5249d4c0edca960 : SUCCESS in 11m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/384db63cb62646709d92e8c8859aa1ee : SUCCESS in 9m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ef9999f723c24783928624889f981ff6 : SUCCESS in 14m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e1ef81f5c0341a5bb054863804575fa : SUCCESS in 22m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b8da057cd04d4b438b70e012a2786df8 : SUCCESS in 13m 22s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/3be420a367fb4e368f5e26a5e02c2a8a : SUCCESS in 20m 59s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/0ebe710d3e7743e9a5ce169dc5f89e37 : SUCCESS in 16m 03s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/3c9ecc9734f94b949b91faecf0a21905 : FAILURE in 14m 46s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/9d340ef05691406cb3e043bfc3eb680c : SUCCESS in 22m 22s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/68bca27485be4a8c807c66d4738458fd : SUCCESS in 20m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/01bf968125664699864fc81922d8d5cb : SUCCESS in 21m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ba3a626a6a50468b904834e7cf4669ac : SUCCESS in 16m 50s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/aa558c1876924214b62bffab2e57ac6b : SUCCESS in 23m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c774f11f9cb04875ab87ab6e655a460e : SUCCESS in 15m 37s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1722a1aa8ede47b18a4ee7f396768556 : SUCCESS in 19m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e99103b2457e4939ad7a46d29732525b : SUCCESS in 21m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c27510db9fb441f183a61ff3b02007a6 : SUCCESS in 33m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8244e52ccc0c4d94954f70fef86d46fa : FAILURE in 22m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90999edbd5be45eab151e6879456f8e8 : SUCCESS in 7m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e1db1d68fbe74fe0afb5a9dd0232d6dc : SUCCESS in 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d978f7c4cf761decd27ba47b372396003d73701e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-05 14:09:10.000000000","message":"Patch Set 1:\n\nrecheck - flaky swift-probetests-centos-7","accounts_in_message":[],"_revision_number":1},{"id":"7ef8e1c48aa7d3f7c28670305363e391cbe0ea48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 14:48:19.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4b9411e49949cda25fb2bc03395c97 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37489c5cc50f4d7db0b4928609220160 : SUCCESS in 8m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/13753818d28340898bb7a8c1d35413f7 : SUCCESS in 11m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f90baa5873e2473f94e2b7e466aa72dc : SUCCESS in 14m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9187eb44942942aeaa073bea2d98673e : SUCCESS in 18m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3092573b67a9460e9f498eddc56e1204 : SUCCESS in 13m 58s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/f41742bd3675400196d1eaa641aca810 : SUCCESS in 16m 17s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/91f6995149b1440bb4e8524d95340430 : FAILURE in 12m 00s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/4a8bdcb903114abab742c379ddc9615d : SUCCESS in 16m 49s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/96577c3d549a4de59d774845221437d8 : SUCCESS in 11m 34s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/36c9fb8351d04efabe70c0c6ffe0e779 : SUCCESS in 16m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/40538efcd208410b8c223a805e14ec19 : SUCCESS in 11m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b0b381e868a049429e90c0fad6bccd6c : SUCCESS in 13m 55s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d877ec9a8633487a95e9b6264873cfb5 : SUCCESS in 12m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5395c90d04654e27a9238e49c43376fa : SUCCESS in 18m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b01237f72a4047758dcd68274e49205e : SUCCESS in 15m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/616f28ffec744a5ab7832e9e0fc98e9b : SUCCESS in 13m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d9969a02be2f419b9a11fc9e761e7c43 : SUCCESS in 35m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e215ac48e8c747819c8fb74e08368bfa : SUCCESS in 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2353079e8e4f406792007db2f91ae273 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/78067b6f4a7e459f82a37b66e30ddde2 : SUCCESS in 27m 24s","accounts_in_message":[],"_revision_number":1},{"id":"af02f7ac89ecbd8158665ecce5ce3470b51badab","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-05 14:52:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"50e94cc1e02928dc3ef1b002e22b370bddef361a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 15:36:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e7572d65bd24450894243bf65a324db : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb3f4da3b82d48b98669f4c0e793c80d : SUCCESS in 7m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ca436f84c924ae5b5a32bf6276f357b : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/03d4e86c2b0449e786e6f1e3d3664830 : SUCCESS in 14m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c8d7ebe450a4befa6d2abd5c642a2d9 : SUCCESS in 14m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/04a648bd607c46928de12f8ffa683901 : SUCCESS in 14m 18s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/499cc8f9b72c423abe1f5dcf63441b4e : SUCCESS in 10m 37s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/bd8f1140bfc642ca975ef32a8d85ab04 : SUCCESS in 10m 49s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/57e13251f2ae4f169e3b32e351b3cafc : SUCCESS in 11m 53s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/fd23e3046c884e4fb2d6a715a7385e98 : SUCCESS in 12m 38s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/b4494b9c15a04b6596936557a55deaf9 : SUCCESS in 14m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/88807fa22a0344f2b922ce65eaea7909 : SUCCESS in 12m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9da41f59e7f54366a8728f9986ee4ba6 : SUCCESS in 14m 26s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/9147bb6c895f42b48907de1b4e18e9cd : SUCCESS in 12m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3368d278663d4d30bd34b7815fbe4451 : SUCCESS in 12m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/47ccf16487a848c889d1babb4c23bee6 : SUCCESS in 15m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c6bb509be5dd43ef8633dde0a6bd3ac6 : SUCCESS in 19m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8f121d94e7b14a009450a2051166d263 : SUCCESS in 36m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d47b2dc7cd094db8ba68132a871a12c6 : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e97f950d7de4ef595c01680c8c3fdd6 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b51f915109c442b8a96445490024481 : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2a4f553ce99b018556dfb6c08f34e9ce96e44681","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 15:36:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"198e337cc1e3a9ec39341f2963ba170cad0b24f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 17:41:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7aca641ed983ffd6b7962fbe7d12231fa134ac85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 17:41:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cb49034779d4b2bb672c800b9da56f0 : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2881ec70dff14808962514012182fc18 : SUCCESS in 7m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5f81fe3d6aaa4d68a8f8667717a4f49c : SUCCESS in 8m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa05621fade64945b7e60e19b7d742f9 : SUCCESS in 17m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4eae08ab0e7484e9802c1d8113525d1 : SUCCESS in 22m 49s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/28c80dd74ec7447f97df8665210397f2 : SUCCESS in 22m 57s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/f5ec4022ae98444885e74bd8ac1d895a : SUCCESS in 10m 11s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/deff6503f2be401d99946feeda831e9a : SUCCESS in 10m 44s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/b8d67b2560104d2993aa611a359bb00e : SUCCESS in 21m 58s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/1032b937bcca4b23a4de935fdbf5d426 : SUCCESS in 11m 00s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/f7e985d37a2f47aabc66e8422e0201ab : SUCCESS in 13m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bfb400bd80c4473d8d0a327f78e13ebd : SUCCESS in 16m 46s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f2d97473732b4e46bbc50eccbd2b7580 : SUCCESS in 10m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6098d9272b004863be7a8d0d08bff909 : SUCCESS in 22m 30s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/422e202a02a84bd2b4cf0a074b22278a : SUCCESS in 18m 06s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/51e7bfd69b9145d798f9953bc56b3411 : SUCCESS in 33m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d019563dd9849f1a7c6f41ebd74eba7 : SUCCESS in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b61930f16a647e0a1037cba2d728dd3 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"521379f7cb54ab9e1835baaf8504a91a3c19a13c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 17:42:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a4a9969559435ea00282c5df762d73 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4292496c58af4f7aa5394807b64a5d9d : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/56c85018908f4c748feedf1fd1f52898 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0e3cc830b4a15c03397c4bff44cb1ceac5578a3","revisions":{"a0e3cc830b4a15c03397c4bff44cb1ceac5578a3":{"kind":"REWORK","_number":1,"created":"2019-10-04 07:04:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/686527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/686527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/686527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/686527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/686527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/686527/1"}}},"commit":{"parents":[{"commit":"ff1f99a8636b8d75f3377566a756b98a2e7149e3","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff1f99a8636b8d75f3377566a756b98a2e7149e3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-04 07:04:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-04 07:04:39.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I719c9a8cba92ca2625425c04675502712ac109cd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0e3cc830b4a15c03397c4bff44cb1ceac5578a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0e3cc830b4a15c03397c4bff44cb1ceac5578a3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
