)]}'
{"id":"openstack%2Fswift~686528","triplet_id":"openstack%2Fswift~master~I40ce1d36f1c207a0d3e99a3a84a162b21b3c57cf","project":"openstack/swift","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I40ce1d36f1c207a0d3e99a3a84a162b21b3c57cf","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-10-04 07:05:08.000000000","updated":"2019-10-04 19:40:13.000000000","submitted":"2019-10-04 19:38:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686528-1570217893930-adebb769","meta_rev_id":"d97b4e17b9558e2c1eff24a1322a3a6136cef110","_number":686528,"virtual_id_number":686528,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-04 19:38:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-04 15:32:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-04 15:32:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-04 15:32:12.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-10-04 19:38:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e00d3f9fd75dbf95aff9612564bcb6e9d16f5f1","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-04 07:05:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"277293bf97424eabe06acabdd05b93e5162d9f61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 08:36:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/317e59c32357407191ce36bfab7d80fb : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cff2e45d341447fb576f71a54e49c28 : SUCCESS in 7m 36s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/095d9234bc9f4ea9b99fb5b7b66a8e33 : SUCCESS in 1h 09m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e64d87d2e69044cbb070a60c27a8ed7e : SUCCESS in 1h 16m 36s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/39ae4d95624b43d5906c5c382c5bb05e : SUCCESS in 9m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6dfac6c6965f4d6d85fac2c7637d881f : SUCCESS in 11m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7981868df87e4df1a2d71dc53dac5df1 : SUCCESS in 22m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/67b80563b68e4cf8a9ebdffc6c7374c9 : SUCCESS in 11m 33s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/f9e80b00c9a34d499e57537263fd8133 : SUCCESS in 21m 17s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/3e960a45ab684d0eacbd16cd03d51b4b : SUCCESS in 10m 10s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/5011ce8945f34f64b557bad7eb485697 : SUCCESS in 22m 32s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/584ebf42c02b49faac6d62ff3855604d : SUCCESS in 10m 41s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/bd6bfdbef27b4142afce7d916c8e21b6 : SUCCESS in 24m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4e9139d149e344be854d6a68fd761dd9 : SUCCESS in 9m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6c971f64880347ec9acddea4cccd9b5a : SUCCESS in 9m 12s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/533d01a195a448d1a6f4367566583b21 : SUCCESS in 13m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dab8323ad607425b9f5a5e89b50e3063 : SUCCESS in 27m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/cc9f511fe21c4741a7e990819db056f9 : SUCCESS in 12m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df41345059ed4a99aa67f566c18f02d5 : SUCCESS in 12m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/43bd26fcfe2247118451b0a33420a487 : SUCCESS in 36m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eafce0201e7b474e85c9f6996d651607 : SUCCESS in 47m 20s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ff814b7a34334ef18b4d66fee987ab45 : SUCCESS in 40m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9230e57d0117492faa4c22e36f62b41d : SUCCESS in 27m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d05cf45967ec4f32b66a64dc9b5b4763 : SUCCESS in 23m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db526626a0ba4e77b66c7ad85bfd3272 : SUCCESS in 10m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f4b5c20326c4b5a99d6e70a7a5bc020 : SUCCESS in 31m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1a26f5be9254a9abd9f51746c705cdf : SUCCESS in 1h 25m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9d7d6e69de88cf5aa2e507793ca857aa279883d5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-04 15:32:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"310863e31dab6880bc6db8191192b0739de15dd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 15:32:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"860b2ab8a79e78b0377591ab14d81c9e1eb5708e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 19:38:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8310718560e34ee98fa0429755e87bd7 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b50a990af4f47b1828a1b08baf6204a : SUCCESS in 9m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/12f86295cf1b42cfa3c6ca2241b36c57 : SUCCESS in 1h 02m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a600c5a1166d4c438419a013f7c5496d : SUCCESS in 56m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/26b3902faab046e292c5e4402ea2b4a3 : POST_FAILURE in 9m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1514c9bdff6b46029ec676f6eae807c7 : SUCCESS in 14m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5268c2aaccd9411690f0cea7fb4af514 : SUCCESS in 11m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3497a683326e40e5ac1ec588726318b8 : SUCCESS in 11m 39s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/2fe01792c6884d108aa341cafca24f02 : SUCCESS in 10m 40s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/7fac10aaf7914d47ba3efb3e00cb38b5 : SUCCESS in 12m 38s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/9ab151d56e1e41c8a33d554fece9fe5b : SUCCESS in 10m 43s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/8d71c2d97c624659bc2b3ff257f441c8 : SUCCESS in 11m 41s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/6443cc5c5b23470cba7f00757794ca1f : SUCCESS in 14m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/36d2c3506051414784146abfb5e4ce5b : SUCCESS in 10m 54s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/197dfb5ccf8444a391a461d67bf42048 : SUCCESS in 11m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b5909f26aa614c11a1587209998e5b5f : SUCCESS in 13m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f30b7388fdcd423ea9efdf66a249bf9d : SUCCESS in 14m 57s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d186c279eb3c4a0098a725f0c645603b : SUCCESS in 36m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e91e332ed0894290ad12f7c51089d7dc : SUCCESS in 33m 58s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/6ef3e3f3723d4bab90893c1b49a4af18 : SUCCESS in 32m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc882f87fccd41848a0ef9902b606928 : SUCCESS in 28m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9e0ad1b1edb461984bb38fefb4934c7 : SUCCESS in 15m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73072ab3e294319ba1642f9c7945cf4 : SUCCESS in 6m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d98f328fb874137a0d147b27669f289 : SUCCESS in 1h 18m 40s","accounts_in_message":[],"_revision_number":1},{"id":"24cc9c3a478e779a4a05ff91e96f86b56f17a645","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 19:38:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"30ddc016e74c20f1a3ae5d69fa68dce089b29853","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 19:40:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba694153acb45e18af79d0fd12e37a2 : SUCCESS in 1m 25s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7f39cb354724f3aa50baeeb1248c74f : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6c248055cb3042dea1acf8cd069f3677 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb76e00e90aea834c8f3dd8a6ca5131acd43663b","revisions":{"cb76e00e90aea834c8f3dd8a6ca5131acd43663b":{"kind":"REWORK","_number":1,"created":"2019-10-04 07:05:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/686528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/686528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/686528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/686528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/686528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/686528/1"}}},"commit":{"parents":[{"commit":"527a57ffcdefc03a5080b07d63f0ded319e08dfe","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/527a57ffcdefc03a5080b07d63f0ded319e08dfe"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-04 07:05:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-04 07:05:07.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I40ce1d36f1c207a0d3e99a3a84a162b21b3c57cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb76e00e90aea834c8f3dd8a6ca5131acd43663b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb76e00e90aea834c8f3dd8a6ca5131acd43663b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
