)]}'
{"id":"openstack%2Fswift~686864","triplet_id":"openstack%2Fswift~feature%2Flosf~I5eb60d51b880a8e8f0578f870f6f237b61a29857","project":"openstack/swift","branch":"feature/losf","topic":"merge-from-master","hashtags":[],"change_id":"I5eb60d51b880a8e8f0578f870f6f237b61a29857","subject":"Merge remote-tracking branch \u0027origin/master\u0027 into feature/losf","status":"MERGED","created":"2019-10-05 04:04:16.000000000","updated":"2019-10-05 05:39:38.000000000","submitted":"2019-10-05 05:38:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686864-1570253906663-6aa585c6","meta_rev_id":"1e349fa9fec1d146748fabf45fb69ba71b8ce89c","_number":686864,"virtual_id_number":686864,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-05 05:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-05 04:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-05 04:08:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-05 05:38:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94eb3a4515c3c24016e938321f37e1002dac3f67","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-05 04:04:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf2aa9f90556f80216450c0e5cbdc5ca72c1a842","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-05 04:08:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa1047ac55b70c9d1b0441aaff294201ee10eeb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 04:42:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b8fd2870904d1abaf36521a74cab0c : SUCCESS in 6m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec264f5ade4741a8800f878e9d7014df : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2703aec09f5f43fab557ad7cb5c2e045 : SUCCESS in 10m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a32a24258d9a483bac9855c3f9d89ac5 : SUCCESS in 11m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9967baa7502d40b0a948bfca8f79cf0c : SUCCESS in 11m 10s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/4ca3067428a942479c8259a0ec9fe8f3 : SUCCESS in 8m 55s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/6a29a131da9b482c84433221ba827c0f : SUCCESS in 9m 16s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/d47bfc0e67e045fda32b5605e36e8203 : SUCCESS in 9m 48s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/58534e34d6fb4381a86da5ad55c87f79 : SUCCESS in 12m 33s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/446fe3d66c5849f688134f6fc2c58371 : SUCCESS in 11m 46s\n- swift-tox-func-losf https://zuul.opendev.org/t/openstack/build/d1c77d4e60714ba8bf5c43d128f158fc : SUCCESS in 13m 35s (non-voting)\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a84976b4daa24aac95eb3c01b176d103 : SUCCESS in 11m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/61fa5df4260448559fd594afcec3219f : SUCCESS in 9m 58s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/18904de5541f4c9588c7c359c1cfc96a : SUCCESS in 11m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ecd2c6ab652b45ad92508f00da79182d : SUCCESS in 12m 43s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/1ff38150bf184ca98b0d55c92f361071 : SUCCESS in 16m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d2bec63e72e4d6793135f76e303626a : FAILURE in 5m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8b87ff62bdcc4d9e8e3bc0c8baa8b937 : SUCCESS in 35m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/93d1673e88c244b4bced248cdb2397e9 : SUCCESS in 29m 52s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0be9e7e09c5c403db75686577d85413a : SUCCESS in 29m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b089c7f2d5a340e29ab8b894e2cdffaf : SUCCESS in 30m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2335519f01c84299b9bebcc3eed22d49 : SUCCESS in 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c3efbc31e6e486a99ecc5c9b21eee79 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0dba2835cfc54c2a9302c77696f8993c : SUCCESS in 17m 24s","accounts_in_message":[],"_revision_number":1},{"id":"cb7d6858a7c72ea354c74926d3fef8c8d2dad0d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 04:42:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ceb3ae680c2183847d0c0f51279db9ce7413e733","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 05:38:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15722618ca284023b46ebeaf4d7d8726 : SUCCESS in 8m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/adcabcc30de84258ac7867c38a700e23 : SUCCESS in 8m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5f6ec4e5ab00448cae5e664aead7f70c : SUCCESS in 36m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/533b65c7eb0d4548ac017d7916f90519 : SUCCESS in 36m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cd01519f09994d61ae98fe2e2950f069 : SUCCESS in 14m 55s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/03d149b50e7f464fb56f7d2e11814760 : SUCCESS in 9m 22s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/d52df7b8ceb249b09808efefb8e38c79 : SUCCESS in 13m 33s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/feebc11f2b5044ec8c46aa04ca28c8c2 : SUCCESS in 9m 39s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/e80d178099564a97a7400d1692477cb6 : SUCCESS in 11m 44s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/c23d0ba2b19041dab16bb1eaff79f496 : SUCCESS in 30m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1066959d519a493693816f1dfacadc74 : SUCCESS in 9m 01s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1be85fb38b2f4f46bae0e7628656549c : SUCCESS in 23m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bbeac0a9edc642d3b1bf5144f41b322e : SUCCESS in 10m 31s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8b897afe732e4001a62fe08be32801bf : SUCCESS in 18m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/330975df617a47ceb3c62bca167748ed : SUCCESS in 34m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c59659c840524edb93f488cfce12f86b : SUCCESS in 43m 04s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a7343fc41b9e4275b954f91eaf67bdf3 : SUCCESS in 43m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40193ad424c64f31b9f9a10a22307b08 : SUCCESS in 28m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0936a98a4762465089a26b29fdf0d660 : SUCCESS in 23m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf96cd7f33af4a9cad1a21b68c5c7a3c : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7a8e37a5f7ec94db94d8b1470d8b8e9384b1f393","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 05:38:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a64116446237dfe92cf80a188346677582d81337","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 05:39:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696709fa72684f839ca2f308a6735750 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/530f20d889934cadace813fd4b2e1384 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bfa8e9feb51f2b10adfec3a741661a76fcf73216","revisions":{"bfa8e9feb51f2b10adfec3a741661a76fcf73216":{"kind":"REWORK","_number":1,"created":"2019-10-05 04:04:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/64/686864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/686864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/686864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/686864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/686864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/686864/1"}}},"commit":{"parents":[{"commit":"dd8b12e2a8afcb9fa08bed5646e4b1466a9c4170","subject":"Merge \"Add tests for vfile.py\" into feature/losf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd8b12e2a8afcb9fa08bed5646e4b1466a9c4170"}]},{"commit":"ff91df230251eaa038bbd9976bc0e4c20a850476","subject":"Merge \"ring: Track more properties of the ring\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff91df230251eaa038bbd9976bc0e4c20a850476"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-05 04:03:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-05 04:03:48.000000000","tz":-420},"subject":"Merge remote-tracking branch \u0027origin/master\u0027 into feature/losf","message":"Merge remote-tracking branch \u0027origin/master\u0027 into feature/losf\n\nChange-Id: I5eb60d51b880a8e8f0578f870f6f237b61a29857\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfa8e9feb51f2b10adfec3a741661a76fcf73216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfa8e9feb51f2b10adfec3a741661a76fcf73216"}]},"branch":"refs/heads/feature/losf"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
