)]}'
{"id":"openstack%2Fswift~688326","triplet_id":"openstack%2Fswift~master~Iae332bb58388b5521445e75beba6ee2e9f06bfa6","project":"openstack/swift","branch":"master","topic":"bug/1847955","hashtags":[],"change_id":"Iae332bb58388b5521445e75beba6ee2e9f06bfa6","subject":"py3: Fix swift-drive-audit","status":"MERGED","created":"2019-10-14 04:57:22.000000000","updated":"2019-10-15 17:45:23.000000000","submitted":"2019-10-14 20:57:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688326-1571086636253-d762e54f","meta_rev_id":"67a1b3a0343fa1e9414e0225b47af0cc4914fd98","_number":688326,"virtual_id_number":688326,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 20:57:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2019-10-14 18:18:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-14 16:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2019-10-14 18:18:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 18:18:16.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2019-10-14 20:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cadff88bb79365e279a04800aa626b63011c9f91","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-14 04:57:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce07596f8a1b0c2a58a489bf750ab5ea69cc9814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 06:15:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ead7fc8a3604f8ca83a1e92e721e3ef : SUCCESS in 7m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f4739c2e6db84d5780053770711ccabb : SUCCESS in 1h 01m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f2f472e09df9434a9bf03f3c1be220e4 : SUCCESS in 50m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04b34b3612524b0488c241de8e475b4c : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fff5f2b57dbe4bdc9ee251463e642ec5 : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c5767f9ba4f74afeb22139449cdc1fb7 : SUCCESS in 11m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/195c45f5cab640f9b456b721050ead88 : SUCCESS in 12m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b6b114e7a99449e79f0e98bf69ee7eff : SUCCESS in 10m 39s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/3a0c02ba12a245359f0b08d4155109e8 : SUCCESS in 8m 40s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/8bc2914eb9c24f929a105a6f75a39ca6 : SUCCESS in 9m 59s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/c88508f043834b068f02d31a7ade870c : SUCCESS in 8m 46s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/ab3230ecc91e426fb6b1f7ae8cc6fd03 : SUCCESS in 9m 25s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/5a514d23533f45f9aef712eb189df511 : SUCCESS in 11m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/78024dbbf2e74af5a8508a02860827c0 : SUCCESS in 9m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e1f77a88019a4e9784c92f30df71ca0c : SUCCESS in 10m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/20142ad7557c450a94d7f7b5871e6493 : SUCCESS in 9m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6d6fd6eebb6d492b805335941a9bca44 : SUCCESS in 10m 49s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/8875607b0f7941fc96559e023c6681cb : SUCCESS in 13m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8843833c2d4845ce942c4d7b484e0470 : SUCCESS in 12m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/63e6bf3e212745bdb4aa5b3957067a56 : SUCCESS in 31m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0303f8ed6f86430785f6905f3190a8f1 : SUCCESS in 26m 07s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c918c626c99b4c2b84199f869c60a296 : SUCCESS in 29m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ecc3e86f8bc4eaba4747aa9199f865a : SUCCESS in 31m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d3e41198d98452aad5d954652ce7297 : SUCCESS in 16m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae00ec0aa5e94199ac6f34c93d365b69 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3363979f16324f55a39b25c6c89078f5 : SUCCESS in 16m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9c93668fbdf4b85a44ec45319a9ea01 : FAILURE in 1h 08m 24s","accounts_in_message":[],"_revision_number":1},{"id":"bb0c57cbb1fd7ecc54f471b55b3b98af0473b500","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-14 16:46:09.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\nFrom IRC:\n\n \u003cbaffle\u003e timburke: Seems to work now, I had to set python3.6 as interepter at the top, but that is distribution specific. Ref. swift-drive-audit.","accounts_in_message":[],"_revision_number":1},{"id":"25bbc46664ed4f014aa58e4c764234b894397864","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-10-14 18:18:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nHa, I should have read the bug report first!\n\nAt first I was concerned that we should probably just makes \"backwards\" work in py3... but after I got a \"io.UnsupportedOperation: can\u0027t do nonzero cur-relative seeks\" I realized that\u0027s not gunna happen\n\nSo I guess we have to backwards in binary and do the decoding line by line\n\nI\u0027m fairly confident as best I can functionally test on my saio this still works as well as it always did under py2 \u0026 py3","accounts_in_message":[],"_revision_number":1},{"id":"1048d07d045bc4e9a20ce40a284434cb6f9998b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 19:40:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb579e2373541e4ab2ded44cad4680b : SUCCESS in 9m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ba931dab5fd946bca665b226e4dac936 : SUCCESS in 57m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7c92aa4a13c6424badfcb1d2420a847b : SUCCESS in 51m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd3ae4847bdd4e8c8608abe5d3a7053f : SUCCESS in 10m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d693c85935cf4c4f8c461b9ea33b4914 : SUCCESS in 11m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/697faa692fa1490b951248f1f308102d : SUCCESS in 12m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2d5898f717c34c3eb9a739a4a7e6a6fb : SUCCESS in 13m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c714bb78f45b40e0897840e70531b691 : SUCCESS in 12m 42s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/b9b9307fa13a4b529daa67f8dc1e0c96 : SUCCESS in 9m 52s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/f0d9be1bb41840bca7fefb42b57ae4ca : SUCCESS in 10m 23s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/5928ef08a7bf4b5dafdd16661bd969fc : SUCCESS in 9m 42s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/43ad653ecaa7432ba0de08b10384fe72 : SUCCESS in 10m 59s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/791a28a36b224b9f8da940785b3965d9 : SUCCESS in 13m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/606525d5617543b5a277954981ee6d30 : SUCCESS in 9m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/aba9a83abb2b422e88658eee6752e3b8 : SUCCESS in 10m 23s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/8f2b310f36ed46a98c25d47245ec89ca : SUCCESS in 11m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7d17b21c91ea494b87c1b129bd4f5cca : SUCCESS in 12m 04s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/bbb34231bfab4e18820575e12798cd40 : SUCCESS in 14m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3732f54fa4214e54a4dcfb820f78a205 : SUCCESS in 16m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9ae1e204ddc0410b9e220c2c398f3427 : SUCCESS in 36m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a46c5764a27d4ab9a7d131802fd5bd40 : SUCCESS in 30m 25s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/45f9e2877d424b3ba913ea236c01841b : SUCCESS in 33m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/37460d96fe8f4442bf9df4279e30f844 : SUCCESS in 32m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a1533633c8b412a871cd1deca7969a8 : SUCCESS in 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/883194dbe7734e269e42b154f0e44e77 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/edcc1579e9d849b3b9e9f03a6c51523c : SUCCESS in 19m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cbd105a7b0146b5856cec9abb0330f9 : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cd4cb2d9c7c3f477a770abbac88afb35be1c0e4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 19:42:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"210c874ccc40a8ca4c5941529a8bf0377a27bdcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 20:57:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed14083c5d5048b9ba70d9c089445b41 : SUCCESS in 8m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/13488138676846eca0fff86b798e6410 : SUCCESS in 55m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5439f48acf1242d1826f35e41c1d6042 : SUCCESS in 1h 01m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/81e0180a84bb4973a7cb65f068532cb8 : SUCCESS in 11m 13s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/08551d36f070407dabfa136ad2e6e682 : SUCCESS in 12m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4500367cd0f34a56aab0801d51e3b387 : SUCCESS in 14m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28a7165b13094247bc57ed0496226e29 : SUCCESS in 16m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d4a04900ed104af1838eac49e29e6757 : SUCCESS in 11m 49s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/d09a375575f74650820be973065051ae : SUCCESS in 9m 09s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/f049740ffc984b99a257b95764347607 : SUCCESS in 10m 26s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/1eec69a88add4d63bc87d56cbdc6c32e : SUCCESS in 11m 47s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/5cc90b37ce9f4802945c039ec63daf6d : SUCCESS in 12m 10s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/8a0b980e89d54b1f81413289bd233efc : SUCCESS in 14m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7bbd4a8b79c849ce8a8f6a4447a3309c : SUCCESS in 11m 25s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3709ee6a6beb4c77901ad2cc52a91a22 : SUCCESS in 13m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/38f0c88d83e3420b889d359cad5cf742 : SUCCESS in 14m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d190608c7e6140799c15eb9f45297074 : SUCCESS in 15m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2fcf2a47d60543178e1d361d129602ec : SUCCESS in 36m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0d4d88ec66d7410ea3415631dc90dbf9 : SUCCESS in 29m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0dd90b1f509c4dbfb1493f4d38f5ead3 : SUCCESS in 35m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77effb472614478b9ce416cdb6414ee8 : SUCCESS in 32m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6cea7d2949240df999d5b9dc6ebe72c : SUCCESS in 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b733c7b8a497449abd581a04653cee64 : SUCCESS in 7m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60043037a30a4876a884b3e7d4d113fe : SUCCESS in 1h 07m 21s","accounts_in_message":[],"_revision_number":1},{"id":"42d2b940a7dd1bc5a46800b265a13df3b53da21d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 20:57:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b3c9194de6840f5a48ee68e635ba46887e08e9b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:04:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd015bce354a4d288a35c17eb2c53f52 : SUCCESS in 1m 21s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/37841be7182f4d8a9697d41cf83942b0 : SUCCESS in 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/8b6ac16b29084faaac44ea5a29cf521d : FAILURE in 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62b4225e379b5f361304b972536ed32c188899be","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 17:45:23.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit bcb2016673434e720e92fe4e75969c3b518a92c8","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"405a2b2a55866b74f196ad81eaafedeee4615fa8","revisions":{"405a2b2a55866b74f196ad81eaafedeee4615fa8":{"kind":"REWORK","_number":1,"created":"2019-10-14 04:57:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/26/688326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/688326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/688326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/688326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/688326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/688326/1"}}},"commit":{"parents":[{"commit":"29d46ca9f624cde5f452f7f50ed4244be09eb97a","subject":"Merge \"Fix misleading error msg if swift.conf unreadable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/29d46ca9f624cde5f452f7f50ed4244be09eb97a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-14 04:55:54.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-14 04:55:58.000000000","tz":-420},"subject":"py3: Fix swift-drive-audit","message":"py3: Fix swift-drive-audit\n\nWalking through the kernel logs backwards requires that we open them\nin binary mode. Add a new option to allow users to specify which\nencoding should be used to interpret those logs; default to the same\nencoding that open() uses for its default.\n\nChange-Id: Iae332bb58388b5521445e75beba6ee2e9f06bfa6\nCloses-Bug: #1847955\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/405a2b2a55866b74f196ad81eaafedeee4615fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/405a2b2a55866b74f196ad81eaafedeee4615fa8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
