)]}'
{"id":"openstack%2Fswift~688571","triplet_id":"openstack%2Fswift~master~Ic41b37ac75b5596a8307c4962be86f2a4b0d9731","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic41b37ac75b5596a8307c4962be86f2a4b0d9731","subject":"Consistently use io.BytesIO","status":"MERGED","created":"2019-10-14 21:03:51.000000000","updated":"2019-10-15 16:42:18.000000000","submitted":"2019-10-15 16:39:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688571-1571157560166-9bbf562d","meta_rev_id":"ee2fb11ce219c3bffa8ada16eb4662e91b118f79","_number":688571,"virtual_id_number":688571,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-15 16:39:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-10-15 14:55:52.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2019-10-15 14:55:52.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"date":"2019-10-15 13:10:49.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 13:10:49.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2019-10-15 14:55:52.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-10-15 16:39:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c3dce8a1ec78b7a2bcf8adf588f5ff3bc33f382","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-14 21:03:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7ea0cfd7e85220176a764be8b6d4ab163c3719f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-10-14 22:43:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"365d4a1e4a133407c8e45403c71953cee1957bea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 22:47:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92861ad957624de59b94bde49d09600f : SUCCESS in 7m 06s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fb0cfda507904371a527787ed21bb7d0 : SUCCESS in 1h 07m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f4a393ae67743e0a1b5a974221a087f : SUCCESS in 57m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7550eb2adc4d4de7b6984ea4a5a437a7 : SUCCESS in 8m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0ccc2dd6e8c54992a8d72641e544dd72 : SUCCESS in 9m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f1f1cc1efc8f4ef09c1807995f181245 : SUCCESS in 12m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a4d609837d3141709e2dc8b52fc124a2 : SUCCESS in 13m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4b4afa1f044745eea46da935901cb3e8 : SUCCESS in 13m 02s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/2c79054c1eaa4d0d980193f1c0752f05 : SUCCESS in 14m 18s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/6e93d5176d01401b9639c02d08b3d834 : SUCCESS in 12m 02s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/cab333e9397d40c6bcf9535ed9fc8ea9 : SUCCESS in 9m 50s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/d2540c74c5f441f6ba77055335f3fc2e : SUCCESS in 10m 11s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/3d407bac054b4edeae93ef280a74699d : SUCCESS in 12m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a0912e6d8eae432a9cbde95a3394dec6 : SUCCESS in 9m 32s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c0eb11ae545d44ce9ad48e47bb5096fb : SUCCESS in 12m 18s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7ff18832c3214551bbf57568f5bf38ed : SUCCESS in 10m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e4ab332880754037a1ed38b806acb840 : SUCCESS in 12m 42s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/80ed9a2998e142ef8cfe55763b1dcdc7 : SUCCESS in 15m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04df9e0002a84189b91e5dbd292549fe : SUCCESS in 22m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ae0aab95fcd848a1a5bc1db0462445f7 : SUCCESS in 31m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c3c63729c72f4d03aef8dffa26b337ed : SUCCESS in 36m 57s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/7232703333834af3913fa7f98a214642 : SUCCESS in 37m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/885215673f024a839e298da0dbdb56c2 : SUCCESS in 31m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14cebcac6722424c82447e681bf8555a : SUCCESS in 13m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca9acc3821754fb19f2402e511fb7c88 : FAILURE in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c7e6c40222049b6ad68db960309965e : SUCCESS in 20m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd4577b1cb7349b89b8fa2e326a1ca38 : SUCCESS in 1h 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7a2212e7ec2e141007fcc03fc188cdb178d334c2","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2019-10-15 13:10:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50fc6e0951ef82062984df59a851287348405fc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 14:27:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11d3a48bbece4fabb383807644119778 : SUCCESS in 7m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a0718da5bb0444fabe38fcefcf53daa0 : SUCCESS in 1h 03m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/541b4cb1f5394e6e823db490122c0b89 : SUCCESS in 47m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/85d9074388764225a19f8488eda043c5 : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9bded7e7108b4220a7be20ba91ad8f0d : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/62b67f2b507b4ff8ba8a0105f4fb7fb6 : SUCCESS in 11m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c1c9dd0c70c34e82822dbc616a3eef9c : SUCCESS in 13m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d661224fa5894a3e845d5e295736a4a0 : SUCCESS in 14m 55s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/cad82687ba914881be86de4eb0e971fd : SUCCESS in 12m 07s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/1967429816724f54867a077e537a3b63 : SUCCESS in 10m 50s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/eba5ffdc3fba42ff8b3a7fa760fdb7bf : SUCCESS in 9m 13s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/f395fd193bc24239b032c7998a123340 : SUCCESS in 11m 57s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/d9d38868e8924c7887fa5a055323910c : SUCCESS in 16m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c730f751866543ceab0936d4fe317cca : SUCCESS in 13m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/79c33c92eea945bf83a8835b41b790b9 : SUCCESS in 14m 42s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3144c30492614b8a8df6620e3ab135d5 : SUCCESS in 12m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4403e85aea3b4d7eaccd0f6e8a236120 : SUCCESS in 12m 08s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/e056af811d284cfda31fb8e3ae53dc0d : SUCCESS in 15m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2713001fb48042a9b85e706d1a02f3a8 : SUCCESS in 12m 56s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f1d735c083c347258af8db0f926094e3 : SUCCESS in 30m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b6b563bde6354a49a1bc3b61bae26876 : SUCCESS in 31m 51s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e528fbac6358467d9525a38e43062e7c : SUCCESS in 36m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c6fc7f79de494f4ca46df667c4ba7f3d : SUCCESS in 31m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3d506fdb4fc4b9496ccb3cab3b96d53 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf390f78457b432dbe790940ebdcf91c : SUCCESS in 6m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10581bbfaf974a0ebaaee096e6db16e3 : SUCCESS in 27m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/990fe13f54dd4279bc1530c06650460b : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":2},{"id":"49b6eceac7edbb5868f411fe77a368be9ec80cd4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-10-15 14:55:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8f7ec3f005809d84df9f973d185d7c37f5705044","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 14:56:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a407534d6e739ea45218bccb0dd1f4b4aecd093","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:39:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bf6789d23794ab4b2bc80b4317fcb80 : SUCCESS in 8m 27s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f9be0861281443089fa7b0aba6e50ce7 : SUCCESS in 1h 04m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/baa0ab15578f4bfead9ddff8dbfdf66e : SUCCESS in 1h 01m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8dd4462eabef4016a6be2e03e1cbd8fe : SUCCESS in 9m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2523ba5ed2b94ff28cbf51a3ad06a289 : SUCCESS in 16m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5f52202b517f4b4c898b8668c7aea972 : SUCCESS in 11m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/023bd81cf8b540bb94fb94d6370c92d4 : SUCCESS in 11m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/06e9cee236a249488ffe2dbdc5fb4639 : SUCCESS in 11m 36s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/6e9716305b654d16b52bd9cc84f7c2eb : SUCCESS in 10m 15s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/66986e65cc4e474b862851d82f360e80 : SUCCESS in 10m 28s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/77baf06c6a664d9eba05b531b90a8736 : SUCCESS in 9m 27s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/abd426f5fc8040969cb97ba7d46d219e : SUCCESS in 9m 52s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/337c9bd97b054ac984c27a7a81f27302 : SUCCESS in 13m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0bc4bf29f15e409ba4bbdf4d6f045a92 : SUCCESS in 11m 01s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/b88ab96dc3924d2883d25a1de2eca577 : SUCCESS in 12m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c342034ef4b446918aab7ece6d074eb5 : SUCCESS in 12m 42s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/122db970e8b147c6ade990c35cb8f23c : SUCCESS in 16m 44s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0487662322f44097b48f7d2aa68e78f7 : SUCCESS in 30m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/259df5f3bac34d5aa53440b72ac77356 : SUCCESS in 29m 46s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/511472155bd64290b93322faca39876b : SUCCESS in 25m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d61d218dea849e28e912ae00aadbee9 : SUCCESS in 34m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d170015f1b16497bbc008c9bb328a0a9 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7302ba331ca54839b3477d842ffb2b74 : SUCCESS in 6m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d507b1c0b8e34cfda4d17b28576def19 : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ce3e1380dc40af638cbb538d430756ccffc30acb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:39:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ee2fb11ce219c3bffa8ada16eb4662e91b118f79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:42:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375835bd6ffd4ed898b808850f0a51d0 : SUCCESS in 1m 15s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4166bd9f16554df58fb3ebf2fddcd5aa : FAILURE in 1m 01s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/71bc6d33bb4940fc81a45d9bea505d5d : SUCCESS in 1m 20s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d270596b6766bf59ea8eaa58a7d48dd2b368cf37","revisions":{"96fbb7258d2115b886c1051011a04127450d0a96":{"kind":"REWORK","_number":1,"created":"2019-10-14 21:03:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/688571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/688571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/688571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/688571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/688571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/688571/1"}}},"commit":{"parents":[{"commit":"1c0661e9203373a92b4eec19571f40309263376a","subject":"Merge \"Start building py3 docker images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c0661e9203373a92b4eec19571f40309263376a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-14 21:03:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-14 21:03:16.000000000","tz":-420},"subject":"Consistently use io.BytesIO","message":"Consistently use io.BytesIO\n\nChange-Id: Ic41b37ac75b5596a8307c4962be86f2a4b0d9731\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96fbb7258d2115b886c1051011a04127450d0a96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96fbb7258d2115b886c1051011a04127450d0a96"}]},"branch":"refs/heads/master"},"d270596b6766bf59ea8eaa58a7d48dd2b368cf37":{"kind":"REWORK","_number":2,"created":"2019-10-15 13:10:49.000000000","uploader":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"ref":"refs/changes/71/688571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/688571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/688571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/688571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/688571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/688571/2"}}},"commit":{"parents":[{"commit":"1c0661e9203373a92b4eec19571f40309263376a","subject":"Merge \"Start building py3 docker images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c0661e9203373a92b4eec19571f40309263376a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-14 21:03:16.000000000","tz":-420},"committer":{"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","date":"2019-10-15 13:09:46.000000000","tz":120},"subject":"Consistently use io.BytesIO","message":"Consistently use io.BytesIO\n\nChange-Id: Ic41b37ac75b5596a8307c4962be86f2a4b0d9731\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d270596b6766bf59ea8eaa58a7d48dd2b368cf37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d270596b6766bf59ea8eaa58a7d48dd2b368cf37"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
