)]}'
{"id":"openstack%2Fswift~688589","triplet_id":"openstack%2Fswift~master~I799e288f66230a7847a13b7f303e4cd8e63eea59","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I799e288f66230a7847a13b7f303e4cd8e63eea59","subject":"Use `\u003d\u003d` to compare against the empty string, not `is`","status":"MERGED","created":"2019-10-15 00:49:53.000000000","updated":"2019-10-15 22:58:28.000000000","submitted":"2019-10-15 22:56:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688589-1571180204465-25c6a380","meta_rev_id":"48d0674ad7142e41c9bda23daa58373f1423705a","_number":688589,"virtual_id_number":688589,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-15 22:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-15 18:54:27.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2019-10-15 01:54:11.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2019-10-15 01:54:11.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 01:54:11.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2019-10-15 22:56:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c3a62fe249490dadb734daf0dbbae39853e549d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 00:49:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c1b5ae3f3273b10fb81da78ac461d9009130e00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 01:52:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0709042b21b84906a8a5822c9a8cfc59 : SUCCESS in 6m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/707e94875d4041b399485505a080a401 : SUCCESS in 56m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0aa162f45dd6419c918619657a7a01b2 : SUCCESS in 53m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1ae23a71b1a541b1a96b84cc5f3e2268 : SUCCESS in 7m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d7a6a126659b43028585eef72caeb474 : SUCCESS in 8m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e96740dfeae413e8fffd96c66f90c16 : SUCCESS in 10m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f241d4fbfd154076aaade984e7e6741b : SUCCESS in 12m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bc14a502ebca402187b7fbf777c1294a : SUCCESS in 10m 33s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/b37b089ad9f641b4849450bb5f97ec2b : SUCCESS in 9m 11s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/a11d5b865a06452eade319a1cbbf245c : SUCCESS in 8m 38s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/534699f5ed0a4769b962409ecfbaf8d2 : POST_FAILURE in 15m 51s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/33a384b7f8204c1f94d45a69d0fe1df7 : SUCCESS in 9m 18s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/822572fee0b844e9ae730a5209c61c0b : SUCCESS in 12m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/da7ffe04ca104649be2aa1eb4e8cf1de : SUCCESS in 11m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7453f5dc085a49c2b1af56d3c78087e3 : SUCCESS in 12m 06s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a976780a541f4d59b69e9e3640a8d03f : SUCCESS in 11m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/cb4a3cf7a65b495f863352254d261f9d : SUCCESS in 12m 26s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/be8d4471d78049bba2c43a3066aab9df : SUCCESS in 15m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da6282bbef6e4a1c859a1de72181e324 : SUCCESS in 16m 31s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bba1ca2ac4d84537a98fa6bc2b7bdf96 : SUCCESS in 34m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/274806e66acb43fea1e8884a0f6da6d7 : SUCCESS in 30m 33s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a772cbb8a397498f9b088c407a98cc75 : SUCCESS in 28m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/279202f256b24d7184c9557a5c41f711 : SUCCESS in 29m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03018371d35f42e483dbba9c56749e43 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e430dd556b9d464cafe83f3962448295 : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13ff62ec084c47dbb5ad8ec44568de4e : SUCCESS in 20m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/405acc4abd7a47f1aaff63548bb3ea50 : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c47e2f218b2f953fc4f9e54e794514032ce886be","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2019-10-15 01:54:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ed0445a6bd33fccb1e4089e3a651293966ded16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 03:06:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e31b9b0ed54c9295fa164ab1918043 : SUCCESS in 7m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/09fdfdf4eb564ac8bb4908da355ac977 : SUCCESS in 53m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b33c71f5bdf4ada9b79beffda9a9ba7 : SUCCESS in 47m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/14406eae169c40a7bc562ad04941de6e : SUCCESS in 7m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/43024278cc0c4ea6b9279554b7f659d0 : SUCCESS in 8m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/368134cc192744cfa93de77de5c63236 : SUCCESS in 13m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a3e954806164cddaac812be4d70f44c : SUCCESS in 13m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e4a50c159c62445fa5a368b40ae42a54 : SUCCESS in 13m 20s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/d955ff34ffac48159545cd0b8dcb90d5 : SUCCESS in 11m 31s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/5af052210b2644809eec2de0f1bca61f : SUCCESS in 11m 26s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/40966afcf64647f6b87f988212fa9e01 : SUCCESS in 10m 48s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/61e427735d434c3eaf6b7072a2b453d7 : SUCCESS in 11m 42s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/c92353d40fe94bec9aa1d73efc34a6fa : SUCCESS in 15m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0368b93261134be791db057e86b7c774 : SUCCESS in 11m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2b37f6ebaed047db82d24495726a0ba8 : SUCCESS in 12m 27s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/eb589184a3724bd8a6342304360f8446 : SUCCESS in 11m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/072fc86bf3134f8a8d0a2d1cffd41c9c : SUCCESS in 11m 58s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/19a54286f3674c70be16ed06ca27af97 : SUCCESS in 14m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2476e840618345e3bf132f0af4ede661 : SUCCESS in 16m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1edf3cab13734b1a84dc011667afa7ae : SUCCESS in 34m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a57bbf5d96264c0aadaabc5df8e0ba83 : SUCCESS in 26m 30s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/ceb0f5d8193a4366a25845010498cbae : SUCCESS in 29m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aad214a4fd804853a773914a7acdc531 : SUCCESS in 28m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1326b2347024b36877632ffb50c5ae3 : SUCCESS in 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e02dabdcdd84861ac6e33418cb19b00 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f69bbdfd20840569645e25e251c8a3c : SUCCESS in 21m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6248779b1ae5486b979a6c29c0012f3a : SUCCESS in 1h 05m 46s","accounts_in_message":[],"_revision_number":1},{"id":"49816c6fb4bb2b613e5e7cc8264abd8dad1278e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 03:06:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a17965d8a5e126ee5a74ef64ad48015cf76fc57b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 04:13:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bdd70dbcc2e4ffd92b489a116bbdb84 : SUCCESS in 7m 44s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/90efa6b10542419eadf9994965fe9fba : SUCCESS in 54m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8558d436359649be8e852756871be467 : SUCCESS in 54m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bdcb97b88a72469b94d599a7bed8714d : SUCCESS in 9m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fcdb38f12be7456fbd508fc51d011c30 : SUCCESS in 13m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/67afb514137f427487f01b8fc2eaea9c : SUCCESS in 11m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc9d33c5282f44fc8d842e6297757370 : SUCCESS in 11m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a60f7aff9e7a4b009a4b84c3f2cd5122 : FAILURE in 12m 34s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/73cc3c96436443e59a21b70f456ea72f : SUCCESS in 8m 57s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/8661052f9f49449fab08172ce6d775a4 : SUCCESS in 9m 09s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/0a1d949cd6e74fef9e424ddee6c8d464 : SUCCESS in 9m 02s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/979a0dbb52754b9c94515388e5bec016 : SUCCESS in 10m 26s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/211c0876d12c44e19a00b6752d67b7bc : SUCCESS in 12m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3a40ef5188fc4466a1c2a6019a7c3c18 : SUCCESS in 9m 32s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6af42c66568542a4a5c016c957974fd5 : SUCCESS in 9m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c99904a940ae4c1590e68e29c0619478 : SUCCESS in 10m 44s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/db3512611b7c4593babbba4591235581 : SUCCESS in 13m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8e12e01b3a8c45f7a000a8b6f29c0a6e : FAILURE in 29m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d90d73fa51ab4862b0c7152da43d3637 : SUCCESS in 35m 33s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/a54338a32cbd426dbda8478c9659f8d8 : SUCCESS in 34m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20f9a05ded714a78b03c56d815037e3c : SUCCESS in 33m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0c5815110104ad684903bea7e54cd21 : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e97257a4977a47b1bd422dc5508dcafb : SUCCESS in 5m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7045772b4f8e4d36ba777e650b2e457f : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a60f1ce62ca1e88ae64a6737cb37e7f741a9f00c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 16:39:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3495709f7400ef1bdcedc27c15ab1b2df28bdfc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:51:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/228a9c863ea6495da3e07a73c16a3b8f : SUCCESS in 7m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1d7822e3487a48998fd5d49babe20795 : FAILURE in 53m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cdc83594d5f94ebdbd0223a0cebdad48 : SUCCESS in 52m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3742bd0113c948a2bf4477f5f7fa37c8 : RETRY_LIMIT in 2m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/05664ab7125c42cca5bda4ad0e3fc77d : SUCCESS in 11m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3efc24daa91e42839c3324092f43b38f : SUCCESS in 13m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a91c7ffffff47389d0c8945d74d03f1 : SUCCESS in 12m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4f7d125da7c54880a83e15b85416ca09 : SUCCESS in 12m 23s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/06834ce410654319bf47dbc6b4b19342 : SUCCESS in 9m 33s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/57d31fb5d13f48bda68ccf54dbb062d6 : SUCCESS in 13m 30s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/bed94b95e1b94df0ba96a0d0a826d9c7 : SUCCESS in 12m 50s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/70b3f7524c0a4a50a17f810d25952c68 : SUCCESS in 13m 21s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/eb3bdcf992104739a58c3290f270f41a : SUCCESS in 13m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/55c5a6f41084478facd91688bbc6de1e : SUCCESS in 10m 22s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b9c53df6fb8c4901b9fdeeb6e1169e2b : SUCCESS in 12m 41s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/10f3d677bdd14a3ab06b708f586288e9 : SUCCESS in 12m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a6a6aa66673d472cb86127b2188d6bcb : SUCCESS in 11m 03s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/bde275cb8abd4cfe9a32fc5ad4c37c0a : SUCCESS in 12m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/775db678870b4624875e4042bcd9794e : SUCCESS in 13m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0071be4bd1c2413f98fd345a57f68d13 : SUCCESS in 36m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c30e6b31bb0940718a0fda7cf11bbf27 : SUCCESS in 37m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c5cd2531775e4a869078ed03e1f0343e : SUCCESS in 34m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/81769225ad174aa2953fc882a71fa71d : SUCCESS in 28m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f333040d286041d1ac30f2da41f3d019 : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d683f8f25548c7b938fe0aa64741bf : SUCCESS in 8m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4fbf2eb618114532b1cac50191298a48 : SUCCESS in 23m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/636d90125a1c47dd9e6c57b9bf26a422 : SUCCESS in 1h 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"cb5a7e751ee8b5db28e1f6f501cac349b23293b2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 18:54:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b3c4dffe5434865c9d65cacaa8aff5c67ed5851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 20:08:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0668c3575c4a12b6de3be3a8cbfcba : SUCCESS in 7m 46s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/985ef7e950e34da1b22fad55da93ed6e : SUCCESS in 59m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c88fc392f32440bb84249eda28090fb3 : SUCCESS in 52m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/614fbd1aaafc4618b2bb6b9d8b505825 : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/97b86019a1cb467388f0219ca275b085 : SUCCESS in 9m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0de817f702c34070b65e6eec19463ae9 : SUCCESS in 13m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/35523834ba5645eda0f65992413a8d4c : SUCCESS in 13m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ed0883e345974de1ae1c670b72767cec : SUCCESS in 11m 28s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/455116d1776d4b3ab56fb0307b60498c : SUCCESS in 9m 50s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/90644db7bc9e4f8aa8332023c84f8de5 : SUCCESS in 11m 38s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/8b3cbd49859b43db8a56746b466e09a4 : SUCCESS in 9m 32s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/b305be5c592d42dcb6d7afbe6db4fa39 : SUCCESS in 10m 03s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/afbe6d0aff7d4b88a41a349be49099a3 : SUCCESS in 15m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1253029adc1c4012aaacf52bfd0581ea : SUCCESS in 9m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d3d2b16d5d9b463992f9998434822a30 : SUCCESS in 10m 19s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f3eed9394458489eb44bc29881254f30 : SUCCESS in 13m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/05a1498b00b1418392f98b81bb34ee67 : SUCCESS in 10m 45s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5255355265da4f778ceec12b4b387511 : SUCCESS in 14m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e12f98f6207e42469f6749ec8cf0f54d : SUCCESS in 14m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/29d14eb49854412d963ada80ac2f1a5a : SUCCESS in 32m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74f0b1b5d8d241faa1bb9e81381497fb : SUCCESS in 30m 28s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/acb65d2ffb1047edad6b8ae8a478e8a9 : SUCCESS in 31m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/257dcd61b15e47deae9f347ccec34f51 : SUCCESS in 35m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f40145a019a4267b2589ce97a88abf2 : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af7523ad0d814226b312db91a10847bf : SUCCESS in 7m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4b3cb1130e245968bd29b2f5828d3bb : SUCCESS in 17m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a012f7d80f642e7a25f905e3bc56ec5 : SUCCESS in 1h 12m 18s","accounts_in_message":[],"_revision_number":1},{"id":"9f8cd871bbdf0f0e91e5d541f50023a4cb676b8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 20:08:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e22e3b2a579c28e8f907b93b8fa12b46f8de61ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:56:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4805aa5ea991414ab6e175da871cf19c : SUCCESS in 8m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/727f599cecb940e8a8269b750c7ebacb : SUCCESS in 1h 00m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdc0e4809ebc4915a7e1bda909a510a7 : SUCCESS in 54m 39s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/043e190f330148b48e2a77551b325f8c : SUCCESS in 10m 54s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7caee823f82049508b8c27ffabe59679 : SUCCESS in 12m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a72915b20c34034a68d822daf2c8e72 : SUCCESS in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/97cc207f0a3f4478bf209e01c81f9a5c : SUCCESS in 20m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/64d2821ce3dd41b785da20dbf8b0b90f : SUCCESS in 21m 03s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/57e4b5a7d13b45c9a6a32975d312d4c8 : SUCCESS in 10m 22s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/2ed4729bac4a4c11b3a364a31ed8c631 : SUCCESS in 18m 24s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/efd10b617ab14afabfc79fd5d0af9250 : SUCCESS in 9m 22s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/9aab10f05b3d44b08dfb85badcc0b58e : SUCCESS in 15m 17s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/b9935e9868734bb4ab657f2e83a5ae8b : SUCCESS in 17m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2fb2c1db71df4098ac8220bffdff30db : SUCCESS in 18m 46s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/6b0a7c23953e476fbfc6f2bd8effd6fe : SUCCESS in 9m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bd66a9124e414a0b8068a46b5702f888 : SUCCESS in 13m 56s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/b14d3a001f624101b5d383ed03e4877c : SUCCESS in 14m 45s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/de83bfae1b4147c5a1bd613779094786 : SUCCESS in 39m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b8e14aab2d004cee9d026e9653bfe6c2 : SUCCESS in 27m 28s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/62609c36b4fb4a0fbc19516528ad962f : SUCCESS in 30m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/acf3bdecc9f442d99b313c08d5f65a34 : SUCCESS in 28m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33672801227d4c3daa73553c0c17eaa1 : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4693eedb2a8442ba42784cabfd03e73 : SUCCESS in 7m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad545f045f534a27afc1a3bcfa661753 : SUCCESS in 1h 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"723f0912d1b35a2f55c159816b4727c50ba150fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:56:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"48d0674ad7142e41c9bda23daa58373f1423705a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:58:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a83a7951644a50a412e12517cc378e : SUCCESS in 1m 18s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fdc8a96f296f462198089e32c2c546d6 : SUCCESS in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/bae6b95d3d0e4e599f732f2d783affcc : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3bd5653873bbb0ddc154481c3804ce6d801664b3","revisions":{"3bd5653873bbb0ddc154481c3804ce6d801664b3":{"kind":"REWORK","_number":1,"created":"2019-10-15 00:49:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/89/688589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/688589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/688589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/688589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/688589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/688589/1"}}},"commit":{"parents":[{"commit":"405a2b2a55866b74f196ad81eaafedeee4615fa8","subject":"py3: Fix swift-drive-audit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/405a2b2a55866b74f196ad81eaafedeee4615fa8"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 00:40:42.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 00:40:42.000000000","tz":-420},"subject":"Use `\u003d\u003d` to compare against the empty string, not `is`","message":"Use `\u003d\u003d` to compare against the empty string, not `is`\n\npy38 will start complaining like\n\n   SyntaxWarning: \"is\" with a literal. Did you mean \"\u003d\u003d\"?\n\nChange-Id: I799e288f66230a7847a13b7f303e4cd8e63eea59\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3bd5653873bbb0ddc154481c3804ce6d801664b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3bd5653873bbb0ddc154481c3804ce6d801664b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
