)]}'
{"id":"openstack%2Fswift~688594","triplet_id":"openstack%2Fswift~master~Id91fdd19f226e9ec0d9c702d40d041c385c52b88","project":"openstack/swift","branch":"master","topic":"688594","hashtags":[],"change_id":"Id91fdd19f226e9ec0d9c702d40d041c385c52b88","subject":"py38: cgi lost some names","status":"MERGED","created":"2019-10-15 01:35:15.000000000","updated":"2020-01-30 20:36:30.000000000","submitted":"2020-01-30 20:32:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688594-1580416369743-46a14a57","meta_rev_id":"48444ef03e7d96eebd81b96a30238e4d6668a23b","_number":688594,"virtual_id_number":688594,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-30 20:32:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-30 19:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-30 19:06:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 18:07:23.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2020-01-30 20:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ce88d988c3da0db10cfe57de95498d06cb1ce21d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 01:35:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3bdadb2b2ff40c685daea37a706cd79bf5b2cc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 02:46:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcc71874fa124511baa716a76039c851 : SUCCESS in 8m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7024508af1b9494f9ecf464fa0085da7 : SUCCESS in 1h 05m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/718eb9d676cf449c94e5694aed365542 : SUCCESS in 43m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5900424cc9b243059b2157c9009dd868 : SUCCESS in 9m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/09a82910b21640459708aae8f3c7bc84 : SUCCESS in 8m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/99725de20d2e496d8f9bb9500b5cfbb4 : SUCCESS in 28m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/56c30eb92ba345578a8b9fcba08ec02f : SUCCESS in 16m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/00477fa5ef63414387d14fd91dc41090 : SUCCESS in 29m 29s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/9c6f118bd0ca4aba98d2ed02afc30a7d : SUCCESS in 14m 36s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/17c61f4c748e4d708a047b63004d449d : SUCCESS in 25m 24s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/330394aca1a74eb2b1f49db06b41b6c5 : SUCCESS in 15m 00s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/63fb2b2865b448a890e031e821debade : SUCCESS in 10m 35s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/bbc090e4498d4a21a6c668af05e4f312 : SUCCESS in 21m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8e4074a9cda64eeaa7bdea266c83f8f4 : SUCCESS in 26m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/00e85dac3aed4450a899ad3429929a0e : SUCCESS in 9m 20s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5b5313131e034034895e7a8a90024224 : SUCCESS in 27m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d981ea5afa824886b1aedf17b68435af : SUCCESS in 10m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ed95e86bb8754c608510ce924b3d587e : SUCCESS in 31m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a3cbf71587a345fbae01ce884d48ab0b : SUCCESS in 15m 40s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/afc9b17384304acd85c8f5e72590fa53 : FAILURE in 30m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bcd083adc17641e0af6cfc2a8582c018 : SUCCESS in 42m 42s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/983f2f1cc48748c7a7e43602a19838d5 : SUCCESS in 28m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d639aebcc9f14b76bab3c3fab87a5a69 : SUCCESS in 27m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4385e3b212704227a79abbc3df64522e : SUCCESS in 29m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6df8c2fc797143ebb3f3e24c0f7789f2 : SUCCESS in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0dc7a9cbc75c4f738e58a9426a6a5b00 : SUCCESS in 17m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/590d387c20904cfda34d91e491f6c547 : SUCCESS in 1h 07m 15s","accounts_in_message":[],"_revision_number":1},{"id":"eb6365c6ce70045a3dd72b21dd957efae2f1f3d5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 16:41:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3e75ea7ea096eecc717bcc5c20d57edda4def1be","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2019-10-15 18:07:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nwere we doing something to suppress the deprecation warnings in our py3 environments?","accounts_in_message":[],"_revision_number":1},{"id":"b448845cc8474eb5eb25c885c052a9449548cc87","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 18:36:00.000000000","message":"Patch Set 1:\n\nHmm... https://github.com/benjaminp/six/pull/290 has a comment about how\n\n\u003e cgi.escape and html.escape have different defaults for the quote parameter, so this isn’t a straight move.\n\nI don\u0027t *think* it should matter for us, though?\n\nI\u0027d also be content to wait a bit and see what six does over the next few weeks -- surely with the *removal* of cgi.escape, this will get addressed sooner rather than later, right?\n\nMight still just mimic what they do, though, rather than drag forward the dep to a version no distro has packaged yet.","accounts_in_message":[],"_revision_number":1},{"id":"f56ff960727759a8784f00a6de40f1d54901b6c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 18:53:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ccb38a85ae476d9db43c5de434e672 : SUCCESS in 6m 55s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/65de0e4f81e24bb09ada96b860359cfa : SUCCESS in 57m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ef526ed92ca4b52894a6c259e750b7b : SUCCESS in 58m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/None : FAILURE (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/None : FAILURE (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/05b55b57dd26454294d0c4e369b53ac8 : SUCCESS in 11m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4e376ac9ff7a4040bc016d1cc218cc81 : SUCCESS in 12m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/04a97f3f15aa416f9b1479230cf6b67d : SUCCESS in 13m 49s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/dda6bc3236a745bea2311cabaac7e23a : SUCCESS in 9m 23s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/cd5b2cd7b7d14caf9ae27f3c3befb95e : SUCCESS in 12m 43s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/ca6eea2506f64fd683eab537b194f466 : SUCCESS in 11m 57s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/c69586c665f44a6596470f3cb72ee7e3 : SUCCESS in 13m 14s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/2f82364cbff14caca36cd565f9db9820 : SUCCESS in 12m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cf6da8e6d81a4cb1bd1d10ca295a54f7 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1be9dc7307ef4afcb36bea5c4ef07323 : SUCCESS in 21m 22s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5e3a101ac6f1470e918fadd73167fa1f : SUCCESS in 9m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f19c13d61cdd4dac9033f8267e3ca3be : SUCCESS in 15m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/ef5bf6717dc54427bdbd0e523f2d9cfc : SUCCESS in 17m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8571d0700c09421c8bdd9df711cc7a5a : SUCCESS in 15m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/15c536a4454741c98c73fe260c33e194 : SUCCESS in 34m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08b5df0bc78c49f599fc8261c7eef7ec : SUCCESS in 33m 11s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/4773861189be48e28bb51df00a8b9bfb : SUCCESS in 28m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/46b26f4abac14cbc8d0e27a773703052 : SUCCESS in 31m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a11c647820d344bc9a634da467e11009 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c54df0e7c61a452688ba271f65278736 : SUCCESS in 6m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1961ff9db37441fb876a0a949a75a1b8 : SUCCESS in 27m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/73be7eb926e245ac9b2e8f40daef28e2 : SUCCESS in 1h 06m 51s\n\nWarning:\n  Requirements [\u0027docker-image\u0027] not met by build 09f9b87b89734f328383ca51692680ce\n  Requirements [\u0027docker-image\u0027] not met by build 09f9b87b89734f328383ca51692680ce","accounts_in_message":[],"_revision_number":1},{"id":"8b278caa5fa2c9b9ee83255473040fe02a0881f1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-22 19:39:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98f0fec282560c6b1ab2a1aad1c975c91b8c0635","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-22 20:19:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"81b74506d35596e6fb7647d255a7b995c5ddf4b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 21:54:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e4340f6771471c8c1d88f7a62dc5c0 : SUCCESS in 9m 24s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0f2ed861c0f64f048180c64aa36303b1 : FAILURE in 1h 05m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/71999f086b664f529fd3d01d180bbd35 : SUCCESS in 1h 13m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8854c689487c4e95a906b766a533ed47 : SUCCESS in 13m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aa70c529b6894575be04d596599c59bd : SUCCESS in 9m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4a987e8c896b4d4a9102547473637079 : FAILURE in 19m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9775850d76ed40388f389c8a145ec45f : SUCCESS in 23m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/82f5d4f278e24ff9b96d657c4dfdd5c4 : SUCCESS in 15m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bf02735f445a419298d6df1dfeda3edd : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/925cea2f6ed4490ea0be774977376042 : SUCCESS in 16m 25s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/d00695c78c894f58a4631f21872e79e1 : SUCCESS in 16m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4c1c2a8a29de402d82ead790d67dc9eb : SUCCESS in 13m 00s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/54dac54272b34fbfb1e6bc3e5c8975ce : SUCCESS in 22m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ca31065f6cc844c5ace11c99b52f4d5e : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a262f78e60eb4dbb84597652d8d09e39 : SUCCESS in 15m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/1b921169a7e04dc79abab12cf8ca5f75 : SUCCESS in 12m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8074a9d662814eb9ac14b85492b96963 : SUCCESS in 18m 52s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/91839e00e6b74a4da7d1eaada3bb5213 : FAILURE in 26m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e1c2636533b40af94c755294ae2e363 : TIMED_OUT in 31m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7e2011a03b1244bda4e8ee18521c3dc7 : SUCCESS in 49m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5dd4ae6ef40548f7ae56bae0d7ac563e : SUCCESS in 39m 55s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/60cb7d0c70554330bf002e95cb5f650b : SUCCESS in 41m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d5c2e1ea73bd4827a415f652d456304a : SUCCESS in 44m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be9bd1e39413412a97a66c40a22bd19b : FAILURE in 18m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73b8c749ace4202af7338ba863d4035 : FAILURE in 9m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b8d332933154df58f4ef7c637271f2e : FAILURE in 28m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/716dc30ba318430eaf78d82c0ae9e772 : SUCCESS in 1h 26m 45s","accounts_in_message":[],"_revision_number":3},{"id":"c4dd82b5645e355d6a9f45ea664b9724546f9ff2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-23 00:51:33.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6dfb8ab23dacd76b7385c619c7b38f886b0b6296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 01:56:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1ac7823bba345d7b5e4819920c14131 : SUCCESS in 7m 22s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/01ff52cde4514ba5896f5b40d104b8d9 : SUCCESS in 55m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d563323543ed4cafa6a9f65018234303 : SUCCESS in 50m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/039e7d370fc54202b0991544317681b8 : SUCCESS in 8m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7d510a80f62548cebaa602834324bc2e : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bf2119f5428f4ceebe4614a22536c04e : SUCCESS in 11m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/02d74f4da55340ea96ef4a3ebc50de39 : SUCCESS in 11m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f48fe3fbda4f4833b80e5a42e2fe3f01 : SUCCESS in 14m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/017cda2a716c437ca1cb5747507d3b80 : SUCCESS in 9m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/91782a1e9fb14381a2d408eefe56bf8f : SUCCESS in 12m 03s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/6e6b8c48ec2d484d8eda3b5160870cf0 : SUCCESS in 13m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf1f4b50c0ae4d9cbd723017c3a4fec8 : SUCCESS in 13m 36s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/717ee296da2a4cd9b43a30f883d2910a : SUCCESS in 16m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3304a8614f71400da39b09093a8d0271 : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/22e5e67e044248cebf11d1fc43adbc8b : SUCCESS in 13m 50s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/58b25bb1821e4d368eefb05c39c733a8 : SUCCESS in 13m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ce6e1ccfb1d6410ca6dc72ef0b7e6aac : SUCCESS in 14m 16s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c89c2cc8c6f54adcb35da0d72aeba95b : SUCCESS in 16m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef30743238764e7086399c02d35eea09 : SUCCESS in 19m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aacf3521355748b19850f9b2cefd3c66 : SUCCESS in 43m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/79f2338a069841168785f61bc886122d : SUCCESS in 27m 39s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bd95b3c1d78a43419950f963617ef44a : SUCCESS in 28m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72d3fdb2a72144f9954695f3ec610f64 : SUCCESS in 28m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ff15c888fdc1426ab7ac7cb7d4cdb666 : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c12741802dac444fa7cb2551c5493d35 : SUCCESS in 7m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf3dbcfdc54d4678a55387a115116eb8 : SUCCESS in 24m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f40316f5d94409988bec3fcc428a0fe : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":4},{"id":"e668b084bf4c881f7e2904f835ef5b88f1180fce","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-29 20:12:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c600bed36b4893824ce8ed57d000c82a7d1ca18","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-29 20:57:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f45af3bcac22a5dc0654a541c3f9c1119f8da252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 22:20:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425ca37d1eb14e7dbf346077caac8650 : SUCCESS in 9m 50s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3fff2cc991be4655a3f9a13fc8430891 : SUCCESS in 1h 04m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/028ec733c8ed496dbaf83ed0735f752f : SUCCESS in 56m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/55d8164a21964044ad496f77e2efa2e9 : SUCCESS in 10m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/469b69587f374fa095362a43a6b59382 : SUCCESS in 10m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9d97dc13de7a47f68605153f9eafc3ca : SUCCESS in 11m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4b1edc53a84b4b378fe5959caf33ff57 : SUCCESS in 12m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5b59d4ff31314997b7aa7092f9220c68 : SUCCESS in 14m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e8e838807503464f8eb2fa5460f5b051 : SUCCESS in 11m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5786137d8c614f8a9050bfb7142e683a : SUCCESS in 15m 42s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/0c5ce04ad3e34f8d8429aef6abe44110 : SUCCESS in 11m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e66422f7d15743e6be1ebd0067262e65 : SUCCESS in 13m 43s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/d237537620524c97af3c218771b67e20 : SUCCESS in 17m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c2c0ca68f001488b85eeee2d22304299 : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4892fb5d8f404f93be5982320bd8bda6 : SUCCESS in 16m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/182e78552b8c482f8d1ea3e495da4371 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b2a1665070554b4e9d156a256ffe7a8c : SUCCESS in 17m 58s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/abbdf66fc8a2470e8e9aa48af54c68ce : SUCCESS in 20m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f487a4766247456b9bc03768d4a235f9 : SUCCESS in 26m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c5040425671142a09d4ae997cc8e9d92 : SUCCESS in 49m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/35ac07cf74c54da3b4e31d1549c1497e : SUCCESS in 32m 09s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/275342ce5fe94530b1ca1870337f4ccd : SUCCESS in 27m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/948d4039205041e7b57a1c9fa7f7fd9e : SUCCESS in 43m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2384c777bddf4433a84474eb1ee48bb6 : SUCCESS in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b877dba1e8584349a7860b662b8214a4 : SUCCESS in 7m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d78c030b4b6470791c7553ef79678d0 : SUCCESS in 22m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4970f069b3f3443ca0fd3003f0ce3d6c : SUCCESS in 1h 17m 49s","accounts_in_message":[],"_revision_number":6},{"id":"57fa17ad1144128184f01ca940f3457612b00e69","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-30 19:06:26.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f516c6b36cfd51604a802ad42be44266a1cbbf37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 19:06:58.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b8c4b45910b9337b16ff86dfae5fef0b5c3e27d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 20:32:49.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139f009e12814271936c0e52f2736969 : SUCCESS in 18m 41s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/782e25bd8dc64b73a2efc78845524976 : SUCCESS in 1h 07m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f64bdeee616432bb826f870ac335ff5 : SUCCESS in 1h 01m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/969fae97e4ad48269f1cb8cee5390cba : SUCCESS in 12m 01s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/71a98992ba594016a6710468cd4f32fd : SUCCESS in 16m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b742fc0db0214f7c8cddab555725ce06 : SUCCESS in 15m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fa4efdf34de8425fa49c0242d87e3169 : SUCCESS in 12m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/393a3b7fae1f49258b5be7f5032dda44 : SUCCESS in 12m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f772160098fe40cba523a97c3eaa84c3 : SUCCESS in 10m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2bf8f6a82c7547d7ad4ee4b76ae9250c : SUCCESS in 15m 00s\n- swift-tox-func-domain-remap-staticweb-py27 https://zuul.opendev.org/t/openstack/build/9a6ea566bae04b6c90e9dd740a5b6502 : SUCCESS in 12m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8fb6a2e6851947c990ff503f5b314f71 : SUCCESS in 12m 24s\n- swift-tox-func-s3api-py27 https://zuul.opendev.org/t/openstack/build/fd940b8f708d417fb64a995d31324f91 : SUCCESS in 19m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7ef8dc62fb124307afe208f796033bfd : SUCCESS in 14m 48s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/37228739affc44cbb12b7e8665c0cc17 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/205f857484a44ce4aa957aa0001ba63e : SUCCESS in 16m 59s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/683c1899692d477f9b69605c0456a3c7 : SUCCESS in 24m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f13f8d2f70964624b4ede2e553007727 : SUCCESS in 48m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/401330ffc2234eeeab5eb1d380cd5698 : SUCCESS in 38m 01s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/bf7112473b6d4d779e5ca66b98af9a8b : SUCCESS in 28m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a73f4452963f460882e7e12b3fad83a1 : SUCCESS in 31m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62d5e59840b548b8851026422eca601d : SUCCESS in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53de3114df7947c29e5737c10665dbc4 : SUCCESS in 7m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21687a65a353475f923315d873fa71d9 : SUCCESS in 1h 11m 59s","accounts_in_message":[],"_revision_number":6},{"id":"18297eea2729fafa24cd986d355dd1a599e2b929","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 20:32:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"4cc7711d27bd7b7b54bb4991b55dcec2b7488e46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 20:36:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9894fd1b004d7f8c53f09f9ee63b11 : SUCCESS in 1m 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cdf2a6cb703f4ce4ac8bc374ec4b503a : SUCCESS in 1m 11s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/086693f6abe3413a96084f9918897cae : SUCCESS in 1m 10s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"e3ee6ce31aed8d24fefdd20900d20688a12bb55d","revisions":{"9d06c4d44f33bfcceb4ea7732e6bd32794666219":{"kind":"REWORK","_number":1,"created":"2019-10-15 01:35:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/688594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/688594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/688594/1"}}},"commit":{"parents":[{"commit":"8efaa3556fdb21f741ff897dac4730c3c3654421","subject":"py38: Provide readinto() interface for RingReader","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8efaa3556fdb21f741ff897dac4730c3c3654421"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 01:35:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 01:35:00.000000000","tz":-420},"subject":"py38: cgi lost some names","message":"py38: cgi lost some names\n\nChange-Id: Id91fdd19f226e9ec0d9c702d40d041c385c52b88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d06c4d44f33bfcceb4ea7732e6bd32794666219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d06c4d44f33bfcceb4ea7732e6bd32794666219"}]},"branch":"refs/heads/master"},"102146f394dfe861553cd376d08841ae6eda1281":{"kind":"REWORK","_number":2,"created":"2020-01-22 19:39:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/688594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/688594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/688594/2"}}},"commit":{"parents":[{"commit":"8efaa3556fdb21f741ff897dac4730c3c3654421","subject":"py38: Provide readinto() interface for RingReader","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8efaa3556fdb21f741ff897dac4730c3c3654421"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 01:35:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-22 19:38:43.000000000","tz":-480},"subject":"py38: cgi lost some names","message":"py38: cgi lost some names\n\nChange-Id: Id91fdd19f226e9ec0d9c702d40d041c385c52b88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/102146f394dfe861553cd376d08841ae6eda1281"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/102146f394dfe861553cd376d08841ae6eda1281"}]},"branch":"refs/heads/master"},"0efeb2b9a53106877edfe1e28a00575a9626e73e":{"kind":"REWORK","_number":3,"created":"2020-01-22 20:19:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/688594/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/688594/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/688594/3"}}},"commit":{"parents":[{"commit":"8efaa3556fdb21f741ff897dac4730c3c3654421","subject":"py38: Provide readinto() interface for RingReader","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8efaa3556fdb21f741ff897dac4730c3c3654421"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 01:35:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-22 20:19:13.000000000","tz":-480},"subject":"py38: cgi lost some names","message":"py38: cgi lost some names\n\nChange-Id: Id91fdd19f226e9ec0d9c702d40d041c385c52b88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0efeb2b9a53106877edfe1e28a00575a9626e73e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0efeb2b9a53106877edfe1e28a00575a9626e73e"}]},"branch":"refs/heads/master"},"8756717cf0fdffc006e218ddeb651e3a57169fbd":{"kind":"REWORK","_number":4,"created":"2020-01-23 00:51:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/688594/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/688594/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/688594/4"}}},"commit":{"parents":[{"commit":"8efaa3556fdb21f741ff897dac4730c3c3654421","subject":"py38: Provide readinto() interface for RingReader","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8efaa3556fdb21f741ff897dac4730c3c3654421"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 01:35:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-23 00:50:27.000000000","tz":-480},"subject":"py38: cgi lost some names","message":"py38: cgi lost some names\n\nDrive-by: always use quote\u003dTrue; it basically never hurts.\n\nChange-Id: Id91fdd19f226e9ec0d9c702d40d041c385c52b88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8756717cf0fdffc006e218ddeb651e3a57169fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8756717cf0fdffc006e218ddeb651e3a57169fbd"}]},"branch":"refs/heads/master"},"d12c381f9b0b4184e1144a7866db13664d9f0055":{"kind":"REWORK","_number":5,"created":"2020-01-29 20:12:18.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/688594/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/688594/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/688594/5"}}},"commit":{"parents":[{"commit":"8efaa3556fdb21f741ff897dac4730c3c3654421","subject":"py38: Provide readinto() interface for RingReader","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8efaa3556fdb21f741ff897dac4730c3c3654421"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 01:35:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-29 20:11:54.000000000","tz":-480},"subject":"py38: cgi lost some names","message":"py38: cgi lost some names\n\nDrive-by: always use quote\u003dTrue; it basically never hurts.\n\nChange-Id: Id91fdd19f226e9ec0d9c702d40d041c385c52b88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d12c381f9b0b4184e1144a7866db13664d9f0055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d12c381f9b0b4184e1144a7866db13664d9f0055"}]},"branch":"refs/heads/master"},"e3ee6ce31aed8d24fefdd20900d20688a12bb55d":{"kind":"REWORK","_number":6,"created":"2020-01-29 20:57:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/688594/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/688594/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/688594/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/688594/6"}}},"commit":{"parents":[{"commit":"8efaa3556fdb21f741ff897dac4730c3c3654421","subject":"py38: Provide readinto() interface for RingReader","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8efaa3556fdb21f741ff897dac4730c3c3654421"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 01:35:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-29 20:56:58.000000000","tz":-480},"subject":"py38: cgi lost some names","message":"py38: cgi lost some names\n\nDrive-by: always use quote\u003dTrue; it basically never hurts.\n\nChange-Id: Id91fdd19f226e9ec0d9c702d40d041c385c52b88\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3ee6ce31aed8d24fefdd20900d20688a12bb55d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3ee6ce31aed8d24fefdd20900d20688a12bb55d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
