)]}'
{"id":"openstack%2Fswift~688768","triplet_id":"openstack%2Fswift~stable%2Ftrain~I1f140ae00cbd25b23c9a40ee91dccee8c7c15d81","project":"openstack/swift","branch":"stable/train","hashtags":[],"change_id":"I1f140ae00cbd25b23c9a40ee91dccee8c7c15d81","subject":"py3: fix swift-dispersion-populate","status":"MERGED","created":"2019-10-15 17:45:41.000000000","updated":"2019-10-15 22:56:42.000000000","submitted":"2019-10-15 22:56:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688768-1571180202418-f62c7b6a","meta_rev_id":"8bdc2a17045b436f5975faf3445b55e0c24d0441","_number":688768,"virtual_id_number":688768,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-15 22:56:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-15 20:06:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-15 20:06:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 22:56:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e1c39bd94c52f076c671b5c38942da8b6effc3a4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 17:45:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1ae1353c1199b21b2346da2d0e74d08dd26db3e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 19:49:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eef1bd9e34946b488f9f405772d2692 : SUCCESS in 5m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3afd7cf2a33742fba41bc70637d5e675 : SUCCESS in 1h 01m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e97502227244f2682af4e91e311305e : SUCCESS in 47m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb7c314eaa6944f3a93e18afc4ed5f24 : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/11c2fb18a6804d988b549c6350411eaf : SUCCESS in 12m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b71d4a190c3148018607693900b0b769 : SUCCESS in 12m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/04bbf4c72bff4b15b2feb33170de647b : SUCCESS in 20m 33s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/10b4b4dc0e234de7bdb78d2cb6cdad7a : SUCCESS in 12m 39s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/8ad4aebae8d143bc9b743e9bcd78c081 : SUCCESS in 10m 02s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/09542f7989764595a90430abcf7d8ca7 : SUCCESS in 10m 24s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/a0cd66fb6b7d42c183c1a955b093a74c : SUCCESS in 10m 22s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/38a81c7f04b44e38b567c768737aa0ae : SUCCESS in 11m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c7d7cfdb3fa54e46822d22351ed5ca0f : SUCCESS in 10m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/805bac2868f5487ca03bfe599f702894 : SUCCESS in 9m 51s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3ab6071a5ebc48efb0561e59e5c33867 : SUCCESS in 11m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4de101905d3f48999d57d7734363eff0 : SUCCESS in 10m 00s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d731a473b28f47598a5b24a5698ac809 : SUCCESS in 13m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c5b511b93cb74f9cba0a4c7d8c02723e : SUCCESS in 20m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a715031d0d794d79a3deaa97256fa610 : SUCCESS in 33m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b313d64fadb1407c8132fa126e3c8c54 : SUCCESS in 31m 35s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/14b063554be34a449cc06f101cf824b0 : SUCCESS in 30m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b5f50ffee34411184c5de73f8b1ffc2 : SUCCESS in 29m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e6be51a804a421facd7da84516e6ded : SUCCESS in 13m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/788b3821ee5344eaa74884d21e82e669 : SUCCESS in 7m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/245372e3b140468eb21aa56f785e35d1 : SUCCESS in 38m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10e36a263d2b4fa7a99c45289226dc61 : SUCCESS in 1h 08m 29s","accounts_in_message":[],"_revision_number":1},{"id":"63a4c5129a1539fa86ededcc5251aaa681581be3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 20:06:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\npy3 bug, clean backport","accounts_in_message":[],"_revision_number":1},{"id":"1b83ee6499338363d7b0e655f686099bdf7524a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 20:06:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"172c59dc4912f85d569da77fad7d841a00102ebb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:56:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d7682bf2104db38008c5f23aeaad04 : SUCCESS in 5m 42s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7c9b4c49c9424b878d94c23669752270 : SUCCESS in 59m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4f4abc52734a426dbe57aca530648840 : SUCCESS in 48m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/78dea3a8dc5944bfbb6ca8464936c514 : SUCCESS in 16m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e5be24553f774b3f9da79fee9b2a83a8 : SUCCESS in 14m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bd66b4c64b8344649a4a69c3fdb4182d : SUCCESS in 12m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/68c3599728904d46ba7940a4f658283c : SUCCESS in 19m 56s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/f327bee6a3f5459cb15cae90078d21a8 : SUCCESS in 15m 42s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/db4bf3079a8949b0980b0935a426b407 : SUCCESS in 11m 58s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/fe1c7d0ae3db4d3382487f21054bb11e : SUCCESS in 10m 28s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/4fbc5520bd8f49b0accbab6524e86d43 : SUCCESS in 10m 04s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/51a0aa22e8cd42eabf7a9b60cd7053e5 : SUCCESS in 14m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/375513bd3e5e4199bd87521f0f09a0cb : SUCCESS in 10m 02s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/d30d86fa8e474d7f90ec4187dac3b2fb : SUCCESS in 9m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bff0544e65614c44b02be13349ddac79 : SUCCESS in 14m 48s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/5f169eca1a524b61918845aa7b032e10 : SUCCESS in 19m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/79ad504e45b2409589b7ffb0ff40d0eb : SUCCESS in 29m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5329a2313bc74dd491a835ca366384d1 : SUCCESS in 32m 55s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/e4df10edc89044a0b20dba0fcc9d518b : SUCCESS in 35m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1bbe0a6d2e5b4b44ad85c0880227a988 : SUCCESS in 32m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87ab34d30d94484db8a44ffc95e3a77a : SUCCESS in 13m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d833f3345a2426988e5c1ba3c978c17 : SUCCESS in 6m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa44a93edbd441f2866fe89f8f3365fa : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8bdc2a17045b436f5975faf3445b55e0c24d0441","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 22:56:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"203f13163a0d92b5a67decb7334face3353867a5","revisions":{"203f13163a0d92b5a67decb7334face3353867a5":{"kind":"REWORK","_number":1,"created":"2019-10-15 17:45:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/688768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/688768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/688768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/688768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/688768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/688768/1"}}},"commit":{"parents":[{"commit":"c3eed8ab1998b7a61f0472342572fc990b15268e","subject":"Merge \"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3eed8ab1998b7a61f0472342572fc990b15268e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-14 18:34:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim@swiftstack.com","date":"2019-10-15 17:45:41.000000000","tz":0},"subject":"py3: fix swift-dispersion-populate","message":"py3: fix swift-dispersion-populate\n\nChange-Id: I1f140ae00cbd25b23c9a40ee91dccee8c7c15d81\n(cherry picked from commit 4c66596e630ee63cf2cd54d216c8e443203e59fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/203f13163a0d92b5a67decb7334face3353867a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/203f13163a0d92b5a67decb7334face3353867a5"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
