)]}'
{"id":"openstack%2Fswift~688826","triplet_id":"openstack%2Fswift~master~I97beace3094f11286730b99653dc1e8f699c5aec","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I97beace3094f11286730b99653dc1e8f699c5aec","subject":"tests: Stop invoking python just to get the real source directory","status":"MERGED","created":"2019-10-15 22:09:26.000000000","updated":"2019-10-16 07:16:11.000000000","submitted":"2019-10-16 07:14:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688826-1571210065734-f62c7b6a","meta_rev_id":"e9e325128e24b20062fe59f5feb99a6c1c387cdf","_number":688826,"virtual_id_number":688826,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-16 07:14:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-16 04:35:43.000000000","_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":2,"date":"2019-10-15 22:52:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":1,"date":"2019-10-15 22:52:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 22:52:35.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2019-10-16 07:14:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d289d1f7d9449f9f9e5e5db9200ce219d140ab3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 22:09:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d692fb9d253898d3316927bbcc1a886e52e9e4d9","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2019-10-15 22:52:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nlooks good, and this should help with my py3-only saio","accounts_in_message":[],"_revision_number":1},{"id":"9d05a9e50b2df60635f1f15bf3fbd6d547efbe0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 23:15:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff13b8008bbf4cdc8ee2de320c9c0dff : SUCCESS in 7m 02s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d58525cb99d546638dfe59d984149002 : SUCCESS in 57m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5762a4f59dd34d9c97450fca2c90f11d : SUCCESS in 43m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9bbb473c056540faac82524a937b07ba : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91685a48dcc240ab9dbbd72954692a6c : SUCCESS in 9m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0aeffe0af89e4a8482594b85b9f611c8 : SUCCESS in 11m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c20e9db2482a4503acf570562234529d : SUCCESS in 12m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/44b61073a6d94ce5943a24c90d9d1642 : SUCCESS in 13m 11s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/758c7e54816949ea91d0f4b553e22522 : SUCCESS in 9m 24s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/da7b08c104f948238e1702ceb3c40931 : SUCCESS in 8m 57s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/7f799ee0354849c2bc62c89582004411 : SUCCESS in 8m 52s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/9018421a6cbf435b993fbf2a86789abe : SUCCESS in 9m 46s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/8f57d180f9a1447f83d1eb8e4319d4e8 : SUCCESS in 12m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/01bbf4a3a87549b18517107679203268 : SUCCESS in 12m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/acd4fc1122b74294b2f5656d87a76454 : SUCCESS in 9m 21s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/5b6ea71e6c0641f6b9d015f40b5aaed9 : SUCCESS in 12m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4fc8d3802f8f4043a47f8c49e7750038 : SUCCESS in 11m 45s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/d229c30bdbd24f73b1879132e135329e : SUCCESS in 12m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/37527527cf6745a8ae1a2356fb0c498a : SUCCESS in 19m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1a762c0d80d04f13b498c9ae2f27e704 : SUCCESS in 29m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6efb432a45d247b9b6a4882e6decb8ab : SUCCESS in 34m 14s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/b1af4d707c0d4a2f8cd12c74043bfb9a : SUCCESS in 34m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35eab7b007354b57a0dadbd674f225e1 : SUCCESS in 31m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/19a204ca88df4f65998ce06277d405e1 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5765e7af1e5473f9555616fe47a903a : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88f2b26d7f7d435c84eedb8c23e13d14 : FAILURE in 17m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e643887ca2a4d4ca6538e2bb7fca94e : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":1},{"id":"93990b1480ad7bfc63a1466a5e36a30f87ba291c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-15 23:18:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ab18269f19e59c28c9baa836363da07f397333c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 00:22:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702026026768401cbbea46513a212976 : SUCCESS in 8m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/0cb8ba25655b4e06a18dd209cd54fabb : SUCCESS in 1h 01m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/64567e43ce6a4670886b70b8689b2f6c : SUCCESS in 53m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b8f5fe9915e142c59c0f32c9f5090831 : SUCCESS in 9m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2b7daca08c8a4028ac9c48224fcc0c09 : SUCCESS in 10m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/86d76a5f7bf64cca84815b8459bf2ca7 : SUCCESS in 13m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/18ef66a1d6db4cfb94f8b7560bccca3d : SUCCESS in 13m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ec9e565c0be244208e4cbea174d82c27 : SUCCESS in 11m 41s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/0b4bc188f2c04fd3ade4d76a81a5f1b1 : SUCCESS in 9m 12s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/a603f018a14446d192cf1454ef2460ad : SUCCESS in 10m 12s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/a8e573ae5e9a403bae6d2f461f83db4e : SUCCESS in 9m 48s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/03204256a4894424860dd6448cd17aba : SUCCESS in 10m 40s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/f353abf06e924506ad88b617ef1921a3 : SUCCESS in 11m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4f9dccbe8fb541ebb509896acd96db8a : SUCCESS in 24m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/260a0ea187ce46b1bd07c5672afc2259 : SUCCESS in 13m 06s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/7a255d626ba04dd49f3efed3a888e926 : SUCCESS in 9m 34s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6438ef4133554c2b8f74fdef88d355ff : SUCCESS in 13m 06s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c494a359403f4dcb9b107d6c5cda5d7f : SUCCESS in 12m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0eaecd2fb25247308b4ce1199a81d827 : SUCCESS in 30m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f8f8e138349c4dcc876f3f6cafbe5bc2 : FAILURE in 29m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75b04b4b2ea3481e95309e3ebd9b87e9 : SUCCESS in 29m 20s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/c526f901ea2b47928001bb7adddae350 : SUCCESS in 30m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f2f4e09856974496a5c451985f9a1b63 : SUCCESS in 28m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1f6ee891d124600b6c2ed6cacffa344 : SUCCESS in 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9f01001e284da49952fc599ec2cd39 : SUCCESS in 7m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93a718c45e9140279c08b6273a19640b : SUCCESS in 19m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adaa760189ba47ec9ce9fd9df3099543 : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1fad2b8f3a4794fa12c165c41100ea78df8a7c74","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-16 04:35:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ac031ef38d117f404bcdb76587064d06ddb59504","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 05:42:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc2443fc4fef4cdb9b9e26759f04addc : SUCCESS in 7m 40s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fb3dcc1f50174364aa4b80104a0f597a : SUCCESS in 1h 00m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/60ba1affc8b641dfa09fdd6dba315fbf : SUCCESS in 53m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8aa38b1252444f939f41d2283239dad0 : SUCCESS in 8m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/afb7e2b17aa0415ab1e581c7eeb8b1e1 : SUCCESS in 8m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c3714e969134916ad38bd164b358ac3 : SUCCESS in 11m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62c188748ff943c3a43ee5a42726f2b8 : SUCCESS in 14m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/aefb358612ce4427837ad49fd6c7e89a : SUCCESS in 10m 43s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/4f1ed130fc974664a9593ff4f6e9da15 : SUCCESS in 11m 02s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/951d1bde34a341d9b57be30ac6c41e5b : SUCCESS in 9m 30s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/93ee9de6388a4251abbae68c79ed25dc : SUCCESS in 11m 12s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/c7c7f867f74345cd84183c9fdbc42744 : SUCCESS in 10m 12s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/cd0d0cbc6c64483f819c72ee6d5a000e : SUCCESS in 15m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/54edbadcee75472f9b5af6c7ea1cec19 : SUCCESS in 9m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5540ef8890e84d158778e88c655ceeb3 : SUCCESS in 11m 51s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3c365f30b6a64719acd45412c1070cc7 : SUCCESS in 9m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/461afe663e144954aadfe8a4fec54fcd : SUCCESS in 12m 46s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/c41606903b7544519884498b145e5377 : SUCCESS in 12m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/72a3822014f34284a510cba7b66e9de5 : SUCCESS in 16m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f9812d02ac144adcb3b20940f61c01dc : SUCCESS in 30m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00fe1a1bf04c4a56957670bcd6679cfe : SUCCESS in 28m 32s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/154e13207d6c4a29935f26288e9b5a89 : SUCCESS in 28m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a36765e2006f41c39781c287c3dbfe4c : SUCCESS in 26m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/890a6eb2650c4a3787a03a32feacc74f : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e38f32a9be426991c5a5a3ef201462 : SUCCESS in 5m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/befb0c618b0345f98592e534b012ee08 : SUCCESS in 16m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea6f634be5d241e3beee26901af32c2f : SUCCESS in 1h 00m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c5d294a2a9da6a4fda39117ab578b17a11dfdc6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 05:42:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d36b7e91019e75dcc421d7d792eac354b6a0fa7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 07:14:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a093dec1e3c20b2669aa33db75b7cbe4c2d465a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 07:14:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f23dc7a4b054e8db9de94e9f4c4a8a6 : SUCCESS in 7m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/74ce04c6231442f484cddf59040cd2dd : SUCCESS in 1h 07m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f26f63b41cf94e4c839e0e768fcaf748 : SUCCESS in 1h 00m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d9c462ee98fa40f8918018d3efef8c08 : SUCCESS in 9m 23s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d4e2bd7781774d229d484cb11c61b887 : SUCCESS in 17m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27b2c38dd9f247548ff6869efd08e6a9 : SUCCESS in 15m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a5920e10b884efb98259c70da1fc433 : SUCCESS in 23m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f33e55ff09b4127ab608444fbc74ed5 : SUCCESS in 15m 00s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/3207426c0a664c14925c23c77cb60132 : SUCCESS in 10m 28s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/cf81e049543b481783d602a1f4ceea16 : SUCCESS in 16m 58s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/5eeb2a6317dc4e09a1dd50da1b08cbaf : SUCCESS in 15m 05s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/4c219fc6a0a44012bf66ea6cfd444ea7 : SUCCESS in 17m 59s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e66e3311e5be4b26a7a0f8da947c0825 : SUCCESS in 15m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b98bcaf0fb8145809c4a7595edc0b68d : SUCCESS in 17m 26s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/c31c2348f57c4f6a83273782c0430670 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/01d1c0b944e34683abd5d36c8428e7db : SUCCESS in 18m 32s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f7abc639c79341eeb6b6abed91eb132c : SUCCESS in 14m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cfec471f61c74d6b9d21b2ec8de39beb : SUCCESS in 31m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c48a0d776c84c9ea08baf2527e3b94a : SUCCESS in 34m 22s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/48b9707480c743548c59043ad6b8bb42 : SUCCESS in 32m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e3c0e2b9ab7d43b0886320a2ec33c76c : SUCCESS in 30m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca200735e5bc4c70830391dc1427fbfd : SUCCESS in 18m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e17f421f8c3745fb96e1830b778915ef : SUCCESS in 5m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64193d795e6e4a3ba4c859fb1bb912b1 : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e9e325128e24b20062fe59f5feb99a6c1c387cdf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 07:16:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a246a884be7e410daa42be9b4db74991 : SUCCESS in 1m 20s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/aa01b81aef3446268e50c7a0173c03af : SUCCESS in 53s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/13aa895125f1455dac3afbbfb51ef33c : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86952dd3edf4ca45f8df7440cb8b6d31d9e415d7","revisions":{"86952dd3edf4ca45f8df7440cb8b6d31d9e415d7":{"kind":"REWORK","_number":1,"created":"2019-10-15 22:09:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/26/688826/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/688826/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/688826/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/688826/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/688826/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/688826/1"}}},"commit":{"parents":[{"commit":"294472464a8dfeab1f680eac78d4dd8c45552395","subject":"Merge \"py3: fix swift-account-audit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/294472464a8dfeab1f680eac78d4dd8c45552395"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 22:06:35.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-15 22:08:42.000000000","tz":-420},"subject":"tests: Stop invoking python just to get the real source directory","message":"tests: Stop invoking python just to get the real source directory\n\nWhile we\u0027re at it, stop checking for a seven-year-old version of nose --\nwe already list a newer version in test-requirements.txt anyway.\n\nChange-Id: I97beace3094f11286730b99653dc1e8f699c5aec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86952dd3edf4ca45f8df7440cb8b6d31d9e415d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86952dd3edf4ca45f8df7440cb8b6d31d9e415d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
