)]}'
{"id":"openstack%2Fswift~688848","triplet_id":"openstack%2Fswift~stable%2Ftrain~If3ead999bff21dff6d0a46af2e92b15fbe7b4958","project":"openstack/swift","branch":"stable/train","topic":"release-2_23_1","hashtags":[],"change_id":"If3ead999bff21dff6d0a46af2e92b15fbe7b4958","subject":"CHANGELOG for 2.23.1","status":"MERGED","created":"2019-10-16 04:52:29.000000000","updated":"2019-10-18 00:19:32.000000000","submitted":"2019-10-18 00:18:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688848-1571357894744-9904e72e","meta_rev_id":"7b03f929b6e74787b1e62196e1e5150487933725","_number":688848,"virtual_id_number":688848,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 00:18:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":2,"date":"2019-10-17 21:01:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"all":[{"value":1,"date":"2019-10-17 21:01:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 21:01:29.000000000","updated_by":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"reviewer":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"state":"REVIEWER"},{"updated":"2019-10-18 00:18:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da65ab65736ecd479d5f1b7ade5f752f994a4ca4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-16 04:52:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d884193dbffebd7f28cf61012b49e43fabcc83c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 05:09:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/103ddf6cc47a4c15b622843e4662aae7 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3b14ed7dfb9447699a86f0270344d9d : SUCCESS in 7m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/029ba0f4f5dd4febbc1f9129065a76fd : SUCCESS in 8m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e712aec1555c44a5a50fa7b44b85d344 : FAILURE in 11m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ae64cbb73ce41a69225e6dfa790ea5f : SUCCESS in 11m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/94fc50a292444913a929a5df71309013 : SUCCESS in 10m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31a4d21f689d48f9b7d5f3a33b47a6c3 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2bcd4112dd14b7682eee539b9219369 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1c44c68b8a4def71a9e322e12282b5763d266543","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-16 05:18:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"180667c818d1b9f42cfe4a556e8ebb7a5a4dc66b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 05:38:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aae1583d2388425bba83262e044c1f62 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7f76e2d60e24fd7880ba797bfd3b1b5 : SUCCESS in 7m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d50eb45eb4149bf86074785273e2ccc : SUCCESS in 8m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af91129bf20045e0b78daafe2454ffdd : SUCCESS in 14m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3bc589c44bf74a928ea5640e52335246 : SUCCESS in 16m 27s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a677fcab772e4d7bbb50960fc45dab80 : SUCCESS in 15m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0f0f0ca258642009fe19a8626fc3e7e : SUCCESS in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750875a957c94993a1b4ca9f3bc1a024 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8e9f9f2c91c73dc9a7628529ca22f852ca89fc6a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-17 17:16:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9abea4d82ee26e215885f7a30c84c02017875a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 17:32:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/527ede0075f749ff9556264ebf50bc4e : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e5bd76dc05b491fa26cbf7406295c57 : SUCCESS in 8m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dd5d7fa66c92463799f6202b242f2bc0 : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d0822859925849e3874f55b047b1d9a4 : SUCCESS in 12m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2464e07e9b39476c86dcbfcf8047a028 : SUCCESS in 12m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f7e15fb1e7a4c00ae1e6b3778f6da70 : SUCCESS in 13m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/292b2d043a4d4e3bb2f1504b7f96fde0 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1a912b515a4292b40560e697dc7b2d : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":2},{"id":"84372436ed0e3ffdf4edd649ec3ae17adac083f7","author":{"_account_id":330,"name":"John Dickinson","email":"me@not.mn","username":"notmyname"},"date":"2019-10-17 21:01:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nlgtm","accounts_in_message":[],"_revision_number":2},{"id":"045507ae33968690d0b695669baa7da5c6f2f418","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 21:02:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"529a09d5cccbb744aa07ef5776ee97d8acc2230d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 00:18:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3250d938989c895133ef28ed2f658b15c179e7b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 00:18:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199cc2788f2f4d8fbab5f55c654f181c : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7aeb314bb1f4ff6bf04038fbe29eea3 : SUCCESS in 7m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4fbc2767ef164468b8d2e4b9504fb475 : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b820a8fdb473436f8874a47905f86b21 : SUCCESS in 12m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ebfa9758d6d340788ddc4964b2dadac7 : SUCCESS in 12m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f5692f03d6d841c2b9cbe0324b9603f4 : SUCCESS in 11m 56s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/8969181ecf6445bf86c4ff9f26a4d7a5 : SUCCESS in 11m 15s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/71bfca6408814a54a30329e805f868e4 : SUCCESS in 10m 23s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/2034667b5447448687e1eac102680fe3 : SUCCESS in 9m 46s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/a0fece04c1214249bc0a325763250171 : SUCCESS in 10m 18s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/aefa50e15e824515b6087132e5d35aeb : SUCCESS in 13m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b1f2d5ae189f4886909d5ff521a6632b : SUCCESS in 9m 29s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/3e489088cb0c48ae869345b48d805a85 : SUCCESS in 9m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3afbc753a1df40f78312f0e075c8f353 : SUCCESS in 10m 58s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/55b6983222ac459fafd761139ee62271 : SUCCESS in 12m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7bd8e25efff43bab7cb6267016269b8 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/903f998c1de14307866a16084ae680b1 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6522c352f9dda00562f5e57c8c885905a6016f89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 00:19:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522bbbf4b66e458d96dcccce3dc02a5d : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cd174e49d884efda6ce001d34c57149 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cacda8951649408e90b2526f9c2c7a4a : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4e05fa832a48db113386b40be7ccacfe73bfbcc6","revisions":{"f031f8f1d8f331f4434007d93029dd30c6487de3":{"kind":"REWORK","_number":1,"created":"2019-10-16 04:52:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/688848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/688848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/688848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/688848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/688848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/688848/1"}}},"commit":{"parents":[{"commit":"fe5433da3c152c9463e66efe389b0393b43061c6","subject":"Merge \"py3: fix swift-account-audit\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe5433da3c152c9463e66efe389b0393b43061c6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-16 04:51:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-16 04:51:58.000000000","tz":-420},"subject":"CHANGELOG for 2.23.1","message":"CHANGELOG for 2.23.1\n\nChange-Id: If3ead999bff21dff6d0a46af2e92b15fbe7b4958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f031f8f1d8f331f4434007d93029dd30c6487de3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f031f8f1d8f331f4434007d93029dd30c6487de3"}]},"branch":"refs/heads/stable/train"},"4e05fa832a48db113386b40be7ccacfe73bfbcc6":{"kind":"REWORK","_number":2,"created":"2019-10-17 17:16:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/48/688848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/48/688848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/688848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/688848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/48/688848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/48/688848/2"}}},"commit":{"parents":[{"commit":"fe5433da3c152c9463e66efe389b0393b43061c6","subject":"Merge \"py3: fix swift-account-audit\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe5433da3c152c9463e66efe389b0393b43061c6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-16 04:51:58.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-17 17:16:13.000000000","tz":-420},"subject":"CHANGELOG for 2.23.1","message":"CHANGELOG for 2.23.1\n\nChange-Id: If3ead999bff21dff6d0a46af2e92b15fbe7b4958\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e05fa832a48db113386b40be7ccacfe73bfbcc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e05fa832a48db113386b40be7ccacfe73bfbcc6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
