)]}'
{"id":"openstack%2Fswift~688883","triplet_id":"openstack%2Fswift~master~I7a72be3568c2e8afd7211978cd917810c9af22ad","project":"openstack/swift","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I7a72be3568c2e8afd7211978cd917810c9af22ad","subject":"Imported Translations from Zanata","status":"MERGED","created":"2019-10-16 09:02:38.000000000","updated":"2019-10-18 15:54:14.000000000","submitted":"2019-10-18 15:52:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688883-1571413950883-8b4ac2a1","meta_rev_id":"1d713ff3e6b04c777b64dfb54377dadfb30d73cb","_number":688883,"virtual_id_number":688883,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 15:52:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-18 14:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-18 14:39:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 14:39:22.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2019-10-18 15:52:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9e4b028b94457cdc925929d2eb7abc646f14a52","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-16 09:02:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"563077254bb0c7447115a5be705617fd51836670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 11:08:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/494664e37a2845d2ab8667d4abe4ef59 : SUCCESS in 7m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/530a3f68f4744262adcf83044cf11263 : SUCCESS in 59m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4a46b2cc4654cd49263064547d52978 : SUCCESS in 52m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a33a187ae924b6cbaadb507e9d0fea6 : SUCCESS in 8m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0bd088db676145319b398117f5d22f8e : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a2a9c120fb9249ed8ba51d0a37ba738e : SUCCESS in 17m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/82f4dcab29c54f228619a353afac8095 : SUCCESS in 13m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/666ddc410f8840afa4709abbefac27f1 : SUCCESS in 16m 19s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/e4b6298ba3f34bb6999965d588c28c28 : SUCCESS in 10m 32s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/d81da076d9294003b42f1c2b25cd3700 : SUCCESS in 11m 17s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/a58d43bfd1704b9bb3cc4c68a6dad51f : SUCCESS in 20m 30s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/4d36fbd768514aa988e2afa7e0e55b7a : SUCCESS in 12m 10s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/9b1555b437bd484898e2b3e0845c144d : SUCCESS in 16m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/45a714a1afeb4d64a60e466e1fed0831 : SUCCESS in 11m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5f422f4fef80484abff622e10aec85f7 : SUCCESS in 10m 17s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/f621f528960f4e7c845007ba61dc262e : SUCCESS in 10m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/22cae129c690481e96ffd5a3846ffc6e : SUCCESS in 9m 50s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/456711c763bc440fa3f86bf924efc776 : SUCCESS in 14m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/367334aabc844456801f303f7b328911 : SUCCESS in 17m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/02d763cde03e42039b88d8ee6c64dde7 : SUCCESS in 35m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6694b54dd04547efa27cae063734e55b : SUCCESS in 32m 45s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/fe2669df765843779f18f22c8bcfcf89 : SUCCESS in 28m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cb3ae86dfb744868a909422ab66e93c5 : SUCCESS in 33m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d919121462344853a39b0cbd14778718 : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52281c5ea0c3418e93af6f7265fb2374 : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b03fa3015a5945a0827dac63b5754506 : SUCCESS in 21m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb22f522e6fe44a5817c83f4be04bcb2 : SUCCESS in 1h 07m 18s","accounts_in_message":[],"_revision_number":1},{"id":"724fb8f413ba3ee8816cca964cb320e77dbc746d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-10-18 07:10:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfd92dab83acebeb802c96eda416c568675b8f22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 08:26:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8701ce0af5de402dac33b300ba0672af : SUCCESS in 11m 25s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/230f7b07c9014ad5911abb29d78c848b : SUCCESS in 1h 01m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ddb67604c07c42118adc985739f70a1b : SUCCESS in 54m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21772a7c3b644c79b0274c9c51df8f8e : SUCCESS in 8m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5d5513a8ae8149b3b1a1325594e05f31 : SUCCESS in 11m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fc97ae29cd5449df9c061360c77a86fc : SUCCESS in 14m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eb8a5af5dc6740e9a129739d1a7d7bb6 : SUCCESS in 13m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/07721c47f5894585a4d20b61490fd54b : SUCCESS in 14m 38s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/d79287bc1dde4ebeb81b779111f060dc : SUCCESS in 11m 35s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/454894fe6e344288a60a6e81bc39f0bd : SUCCESS in 12m 34s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/ac68915408094f55904227bb903ee229 : SUCCESS in 11m 07s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/7920c0dae1384f5d8b15df1568da1114 : SUCCESS in 11m 09s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/77841064a0894c4fba7cdbc73e7a1308 : SUCCESS in 15m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a7bc74152a434db8b6d07ee9ae3a6abe : SUCCESS in 11m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cd311ff9f04f4360873e61f3cf16dfb4 : SUCCESS in 13m 47s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/a856198d4a604811a5be7039ab80b857 : SUCCESS in 11m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9e3b137743144497a60b1fe74a160920 : SUCCESS in 13m 55s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/e06fc69743694deca587f9254a685812 : SUCCESS in 16m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff073cfa684a461abcb3c1a775163763 : SUCCESS in 16m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5db3daa0e49d45338a8b9c80446d61f7 : SUCCESS in 37m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06483ebec6224e47a1bc102a7e29dce4 : SUCCESS in 30m 04s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/d1534e8672d94e719f20466c31759581 : SUCCESS in 28m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/38360f53f7ca4cb2a265ec6978c5f136 : SUCCESS in 30m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49521efbde864a4dae39594ad25ccf55 : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f671ac6e613b437f8edb02ff25e42979 : SUCCESS in 6m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12b4a601038f4a26ab58c97c8ec467bf : SUCCESS in 26m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7da974c8558549cd86b37ab314a07e4a : SUCCESS in 1h 09m 30s","accounts_in_message":[],"_revision_number":2},{"id":"29d4ac4e5f79f09a35a0c35f4122acf4906d959a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-18 14:39:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5b78f39ab5f998e2079371db9e00f4f352b98dc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 14:39:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a0e33f7e96f5ba1dcb625e9ba7867a7c32cffc71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 15:52:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38da93655f54461d940a1ecf8648c69c : SUCCESS in 8m 49s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bdb020fa61134e4f9d3162cd06b153c3 : SUCCESS in 56m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/712d29a9fad7448fa3ce9f1ac198ab26 : SUCCESS in 55m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d6e41cb343cf44ea9d2892f9dc658f8f : SUCCESS in 8m 41s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/cd5255bd1c3d41fe9b97e0349a3bac24 : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5a3d9758d11945a7a421ab8df1662b39 : SUCCESS in 13m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dcf21b96dd584d3db046c707a72cee27 : SUCCESS in 12m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8283aec0f56e469185f019592c5f167b : SUCCESS in 11m 28s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/476ccbd7f5c94da99855eb89cc677b65 : SUCCESS in 11m 20s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/b37d49bf85f64c9794db9c5c7f38030b : SUCCESS in 13m 46s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/8bff9de8c0124d30b080b9483d750fe0 : SUCCESS in 10m 24s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/0ec0578b05bc438383ce50e2c58eb340 : SUCCESS in 10m 22s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/abff73f57a2b4d5cb32d93f7b08ba89d : SUCCESS in 12m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d6070245a29546b7a9be5f293e91b6dc : SUCCESS in 11m 16s\n- swift-tox-func-domain-remap-staticweb-py37 https://zuul.opendev.org/t/openstack/build/142f8b02f19644fdbcff61a187b42451 : SUCCESS in 9m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0b093e936f4344aca2482c3b9cb07ea5 : SUCCESS in 12m 46s\n- swift-tox-func-s3api-py37 https://zuul.opendev.org/t/openstack/build/f5502d0a152444778dc432391016338a : SUCCESS in 14m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4d08fec08d09481481888261447703f8 : SUCCESS in 33m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3845f0ad2eb1402fbe4e3301f4f6563f : SUCCESS in 33m 36s\n- swift-dsvm-functional-py3 https://zuul.opendev.org/t/openstack/build/0361419355594529bd9b42fbe4a4dd64 : SUCCESS in 31m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/524de02ad81443459a9b6d515b3727f2 : SUCCESS in 31m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f26afe55b5cb435b8d3d7aa420db86ed : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa533b88755c497db25653033909f747 : SUCCESS in 5m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b5db32578d64dfbbb9f1474bbdb1108 : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":2},{"id":"c69bc6b488a4173889d423cb1a590f78b61ac73e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 15:52:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b2bb6277870694790250733aa9398f15f0c00536","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 15:54:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea953215466487698a0d28cb0688a19 : SUCCESS in 1m 20s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/726e514c2f03417ba5694330d1a91171 : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f3811f696b144e07b8afc8f9170f4f5e : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"739c3704107bd991493472c18bfe7ac5f933077c","revisions":{"5d35833ceaf90a5df9f759e25b75da0bbfd175ae":{"kind":"REWORK","_number":1,"created":"2019-10-16 09:02:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/688883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/688883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/688883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/688883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/688883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/688883/1"}}},"commit":{"parents":[{"commit":"f4fc8118626ab5b780e0b14a4a3fbf3f0b99f288","subject":"Merge \"swift-account-audit: clean up some error formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4fc8118626ab5b780e0b14a4a3fbf3f0b99f288"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-16 09:02:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-16 09:02:33.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7a72be3568c2e8afd7211978cd917810c9af22ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d35833ceaf90a5df9f759e25b75da0bbfd175ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d35833ceaf90a5df9f759e25b75da0bbfd175ae"}]},"branch":"refs/heads/master"},"739c3704107bd991493472c18bfe7ac5f933077c":{"kind":"REWORK","_number":2,"created":"2019-10-18 07:10:35.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/83/688883/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/688883/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/688883/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/688883/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/688883/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/688883/2"}}},"commit":{"parents":[{"commit":"d6967a1fb6ef71a36f7bc98303ccf5910539c683","subject":"Merge \"docs: Fix then/than grammar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6967a1fb6ef71a36f7bc98303ccf5910539c683"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-18 07:10:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-10-18 07:10:33.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I7a72be3568c2e8afd7211978cd917810c9af22ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/739c3704107bd991493472c18bfe7ac5f933077c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/739c3704107bd991493472c18bfe7ac5f933077c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
