)]}'
{"id":"openstack%2Fswift~689217","triplet_id":"openstack%2Fswift~master~I948d65d0d3b58fcb13f8141fe095ccad4b8f1425","project":"openstack/swift","branch":"master","topic":"bug/1848485","hashtags":[],"change_id":"I948d65d0d3b58fcb13f8141fe095ccad4b8f1425","subject":"docs: Fix then/than grammar","status":"MERGED","created":"2019-10-17 16:53:31.000000000","updated":"2019-10-17 19:07:51.000000000","submitted":"2019-10-17 19:06:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689217-1571339185863-cd6aae0a","meta_rev_id":"5d4af03648b354ee3bbfac0c36c995a7fc4b2226","_number":689217,"virtual_id_number":689217,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-17 19:06:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-17 18:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-17 18:07:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-17 19:06:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9e6c59251b6aedd8515a766a951cb0de259c5e3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-17 16:53:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"389c2aa769f9083e560e7941659c0c468bfd13ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 17:07:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5658d04572741c991e0cf11a8de0688 : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b1d19299c0e4a59082d09f343c17f40a41b60c62","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-17 18:07:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dc01e34f50bed8ba8d5ef55b1fcf91bf5b49c2c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 18:08:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10f10c9580e77d9f3494499bf0138f020a7ed02e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 19:06:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f167a164aea740d9b9d5551ee9ee8ba3 : SUCCESS in 7m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c8d27d27b9d640eda07219f693a583e1 : SUCCESS in 8m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/464d6430b4c740fb9587f4ef97d5568d : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74397120e785de124ac09dac0e6641ab89b143b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 19:06:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"93cfc96398863f58cf8f02287f9d987e56f4513b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 19:07:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e0248840314a989052c3bbff941079 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2f6e558981514244bfb56c0199834802 : SUCCESS in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cc1b4f6cfae94135a9800e576691e778 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ac50079c1a9f2dfe2db4f9671d823086915e9d2","revisions":{"2ac50079c1a9f2dfe2db4f9671d823086915e9d2":{"kind":"REWORK","_number":1,"created":"2019-10-17 16:53:31.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/17/689217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/689217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/689217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/689217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/689217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/689217/1"}}},"commit":{"parents":[{"commit":"f4fc8118626ab5b780e0b14a4a3fbf3f0b99f288","subject":"Merge \"swift-account-audit: clean up some error formatting\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f4fc8118626ab5b780e0b14a4a3fbf3f0b99f288"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-17 16:52:50.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-17 16:52:50.000000000","tz":-420},"subject":"docs: Fix then/than grammar","message":"docs: Fix then/than grammar\n\nChange-Id: I948d65d0d3b58fcb13f8141fe095ccad4b8f1425\nCloses-Bug: #1848485\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ac50079c1a9f2dfe2db4f9671d823086915e9d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ac50079c1a9f2dfe2db4f9671d823086915e9d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
