)]}'
{"id":"openstack%2Fswift~689580","triplet_id":"openstack%2Fswift~stable%2Fstein~I106b94c77da220c762869aa800c31b87c3dffeeb","project":"openstack/swift","branch":"stable/stein","topic":"bug/1837270","hashtags":[],"change_id":"I106b94c77da220c762869aa800c31b87c3dffeeb","subject":"slo: Better handle non-manifest responses when refetching manifest","status":"MERGED","created":"2019-10-18 20:41:06.000000000","updated":"2019-10-19 02:07:22.000000000","submitted":"2019-10-19 02:05:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689580-1571450751562-70f2cce9","meta_rev_id":"ce97384c0ac9206e9f8f64fccd9608d0e1f5f96a","_number":689580,"virtual_id_number":689580,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-19 02:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-18 23:43:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-18 23:43:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-19 02:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4b29db5f07a6317f85098d49d0043305de82b02","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-18 20:41:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dec982b2e5b91d2ad6b610dd349c7311f1561f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 21:24:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52cca16de3b14cf788651cebbea199d3 : SUCCESS in 4m 17s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a5ae1aff09a74025b5a4906171f71bd8 : SUCCESS in 10m 26s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/07e682629f3f4930aef3ee45510991d2 : SUCCESS in 9m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/546104fa7e0644c1acb889fb70a3c677 : SUCCESS in 9m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/39ec094efe35479daec54e9e6c8ef65c : SUCCESS in 9m 52s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/c929bd27de2848cfb349906846a1c2fe : SUCCESS in 7m 31s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/55642d1c0b1f4610adb89c289ac61039 : SUCCESS in 8m 46s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/e6adb92c62c04768b34fc8722a46a42c : SUCCESS in 12m 54s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/4f965f76b5aa4a3ea1998788d1a348c3 : SUCCESS in 14m 53s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/171659a531ad43aeb9316047b2057c4b : FAILURE in 17m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/106fbceaf0cf4f05873081a0ef3bd630 : SUCCESS in 15m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e18d1d2a39db426bbaa5b923b2eb0424 : SUCCESS in 42m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8045ab4650b245e0ad28cd16de8d7f19 : SUCCESS in 40m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6bc59da02af438e945b9d0cb721bf37 : SUCCESS in 27m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbb30535dc03416a856f069e952f8bb6 : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f951411806c4a248af9a019791afb9c : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8310fd4d4a2a4889a784621c64d87930 : FAILURE in 19m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dfbd502854674fd71e5f74db3d3b8a3bdeb5382c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-18 21:30:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ec4a68df3c34c3e9ac73d6bb41df5bb0c2c9dddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 22:11:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e126be413d684b17bbf38815806bbe1d : SUCCESS in 5m 16s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/252e9791438f45aaa4edf2a40e510eb8 : SUCCESS in 11m 27s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/318a109c7fd0439a88593f8ac091496a : SUCCESS in 20m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14844c3be0be435081b6b438ff005d9b : FAILURE in 20m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/93579e1a38054133af11ba7cffb4ac51 : SUCCESS in 10m 55s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/5f7cf336fed24632b5119647c8dbbbba : SUCCESS in 20m 19s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/47a821021546456fbf6261aaf2be8dae : SUCCESS in 20m 24s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/75b31b6394f14d4e9083aac8be692129 : SUCCESS in 21m 40s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/d0ac1cd1fa1f40c59e84b647b3bdc237 : SUCCESS in 22m 54s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/07307f5ac2864ef0b87b1e80567dfa96 : SUCCESS in 24m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e72cbcf129ae49eba3e9aa2874eb4a70 : SUCCESS in 13m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/494d3701173c4cdb93d0c791cf5c3aa7 : SUCCESS in 30m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4437dfc52ee14048b1acd28e56275749 : SUCCESS in 36m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51046c3d6a564ec1acc6626079b4e61f : SUCCESS in 36m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20c46f3452704c138b2fe3751bc82b3d : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d022444589084d7bbac0429382a8dcbe : SUCCESS in 5m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1f11f7e5b13b481392d37f4527078ae5 : FAILURE in 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa15dac189164e06db3e72ef58adfea5eac5dd82","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-18 23:43:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport, tests pop without the changes under swift/, closes bug.","accounts_in_message":[],"_revision_number":1},{"id":"3c60d2880d348752ec474c11834be61aecc00ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 23:44:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf3971849578b095564bcd90f4d936f4719c5824","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 02:05:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6baadd28925340beb90b8fcbde58cecb : SUCCESS in 4m 43s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/823a39c04f4d45088b52a4634a16c4ea : SUCCESS in 11m 00s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/3fdb273ced404148909e09e785b83d05 : SUCCESS in 10m 03s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/41607cad83e148fe9119df29d526da94 : SUCCESS in 7m 47s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/e3e7911621a343efa101d86332be0391 : SUCCESS in 8m 45s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/9d6b498bef1c48d28bf85414735061b1 : SUCCESS in 9m 16s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/5921d601f36c4e6aa9f07c0abf07c6e7 : SUCCESS in 9m 50s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/1d6f87cb5f66451d9bca1cc928c0b105 : SUCCESS in 10m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/10b2088989784956935aae15d86cbd1f : SUCCESS in 29m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67b2c5208b414fa5a8bdd7baeca874e1 : SUCCESS in 26m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a996054e8091478b91f5e918f9f26c49 : SUCCESS in 27m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fad14ddcfa845fcb7d0aafa3e197100 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55168030493e421497b556025031fd42 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b7f0a93d31eb19d88f6a8ece7158c6aabeb31b5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 02:05:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"474cadcbe9118a6019a62ebec078add89763187a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 02:07:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf36332407e4c739f1040b66dd44cb8 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f353b78b92ad7a3f15aa6b21d1f8be592e8ca212","revisions":{"f353b78b92ad7a3f15aa6b21d1f8be592e8ca212":{"kind":"REWORK","_number":1,"created":"2019-10-18 20:41:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/80/689580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/80/689580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/689580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/689580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/80/689580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/80/689580/1"}}},"commit":{"parents":[{"commit":"3d5f7aa41db7dd06e0965bd7494f2b23e7caf3f3","subject":"Merge \"doc: Fix the swift middleware doc needs more info  to set s3 api\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d5f7aa41db7dd06e0965bd7494f2b23e7caf3f3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-07-20 04:24:47.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-18 20:40:36.000000000","tz":-420},"subject":"slo: Better handle non-manifest responses when refetching manifest","message":"slo: Better handle non-manifest responses when refetching manifest\n\nPreviously, we never checked whether the response we get when refetching\nis even successful, much less whether it\u0027s still coming from an SLO.\n\nNow, if the refetched data is newer, act on it. If it\u0027s older, 503.\n\nCloses-Bug: #1837270\nChange-Id: I106b94c77da220c762869aa800c31b87c3dffeeb\n(cherry picked from commit ef5a37c2bf5a32b0fa6404974aa76332a88c0841)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f353b78b92ad7a3f15aa6b21d1f8be592e8ca212"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f353b78b92ad7a3f15aa6b21d1f8be592e8ca212"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
