)]}'
{"id":"openstack%2Fswift~689582","triplet_id":"openstack%2Fswift~stable%2Fstein~I023b8a46c06e9a2755b5aa890a7992ef9633cba9","project":"openstack/swift","branch":"stable/stein","hashtags":[],"change_id":"I023b8a46c06e9a2755b5aa890a7992ef9633cba9","subject":"sharding: better handle get_shard_ranges failures","status":"MERGED","created":"2019-10-18 20:45:12.000000000","updated":"2019-10-18 23:05:39.000000000","submitted":"2019-10-18 23:04:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689582-1571439854496-fe29816b","meta_rev_id":"e69f9560a515252cd353dfb6b95750e0ab71307d","_number":689582,"virtual_id_number":689582,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-18 23:04:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-18 21:29:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-18 21:29:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-18 23:04:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d211c86b5f867e807b44a05197ad857634a64a22","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-18 20:45:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad3a4afbc39459a97531aaed9d4c2af2bf3c3cc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 21:23:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed49f32016274dc496c68a2f2d1d311b : SUCCESS in 5m 14s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e14babdf72e04d64bf4540295a875fe8 : SUCCESS in 9m 22s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/78d111231d7c4e2f8b3d271d971503c2 : SUCCESS in 8m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/74cc85063b87437eb731ef981dbafaa8 : SUCCESS in 8m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2ffdb29c48d3407780d34ac99899917b : SUCCESS in 8m 05s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/bd1aac078f0348a298cc632ad007acb0 : SUCCESS in 8m 37s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/8e9fcaec04cc4036ae89afa42c8907c5 : SUCCESS in 8m 06s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/7ac59f5b758542b581a14ebf1719157e : SUCCESS in 9m 20s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/eecc5076f7004d028b3906d782cab946 : SUCCESS in 9m 21s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/311650b073f64191accf1317b201642f : SUCCESS in 10m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/24b250c7f7bc4b7db193fa1fa6227752 : SUCCESS in 15m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6c10ce11fb894b9a99250e2ed5e978da : SUCCESS in 33m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/39aad67eed6d4d88869c2b40c322b902 : SUCCESS in 28m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b124ef0018d248b981b4f59a7f9aa07f : SUCCESS in 27m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ac03b115f1b4f48a8710c7d517e0430 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21af3e49dd634b25a38385f0ff5f30dc : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/513ed63602fa44a09ce4aad8b7e79330 : FAILURE in 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f414e9fc53b97f258f64a8fb0ca4e3632fa0eead","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-18 21:29:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport, and such failures are much more likely since https://review.opendev.org/#/c/675432/ was backported. (Of course, that\u0027s still better than the alternative of having the rocess hang ;-)","accounts_in_message":[],"_revision_number":1},{"id":"7f4f66911cd3cd27e5bf75b1dcf3beda770e9628","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 21:29:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff0472cebba837aed75745c2a747747b13780118","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 23:04:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d6228e1bb6b4a228bcad32f27b248b2 : SUCCESS in 6m 10s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6b42c8907499458fa04749e08915edf7 : SUCCESS in 11m 43s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/11cc9623926d4fe0bb92066c359dcad0 : SUCCESS in 7m 51s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/69a9fdd6e4114749872b05124b34c97e : SUCCESS in 9m 24s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/436802180eb64456beb5fae63bcb60e7 : SUCCESS in 10m 12s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/cac4774d84e848aaa1de14c90fe48c0a : SUCCESS in 9m 26s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/8fdca654b2bb4f1aa033908eaa61983d : SUCCESS in 8m 57s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/a26b237072834bb7846a6a138ef59a2e : SUCCESS in 11m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5c74e7a66c074a89960bd46fe5147d07 : SUCCESS in 30m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf5029aa8aad49d8ad40371e41060ff1 : SUCCESS in 25m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cfcc4e46e2b5411fb2170d05e8b72c2a : SUCCESS in 27m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb3f7727318d45b4b91ee756ec68d115 : SUCCESS in 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4391dc53e03e4e629562ae37b0e140e0 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f30fdd8d0a1662a963ca3de2754699c7dba08a0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 23:04:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e69f9560a515252cd353dfb6b95750e0ab71307d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 23:05:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9304413b0169491bba4bfcd13d8d598a : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"657ee1092d8caeb0b9be8a011741cfcdfba6dc99","revisions":{"657ee1092d8caeb0b9be8a011741cfcdfba6dc99":{"kind":"REWORK","_number":1,"created":"2019-10-18 20:45:12.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/82/689582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/82/689582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/689582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/689582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/82/689582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/82/689582/1"}}},"commit":{"parents":[{"commit":"3d5f7aa41db7dd06e0965bd7494f2b23e7caf3f3","subject":"Merge \"doc: Fix the swift middleware doc needs more info  to set s3 api\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d5f7aa41db7dd06e0965bd7494f2b23e7caf3f3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-04-25 19:57:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-18 20:43:58.000000000","tz":-420},"subject":"sharding: better handle get_shard_ranges failures","message":"sharding: better handle get_shard_ranges failures\n\nThe contract for ReplConnection.replicate() is that if we can get a\nresponse, we return it, and if we can\u0027t (because of a timeout, or a\nsocket error, or some other http_client error like BadStatusLine), we\nreturn None to indicate the error.\n\nPreviously, _fetch_and_merge_shard_ranges() always assumed the\nresponse would have a `status` attribute and raise an AttributeError\nwhen response was None. Now it will treat that case like other\nget_shard_range failures.\n\nChange-Id: I023b8a46c06e9a2755b5aa890a7992ef9633cba9\n(cherry picked from commit bc5f4c061122aee5965cfb055beb3cb079073536)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/657ee1092d8caeb0b9be8a011741cfcdfba6dc99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/657ee1092d8caeb0b9be8a011741cfcdfba6dc99"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
