)]}'
{"id":"openstack%2Fswift~689589","triplet_id":"openstack%2Fswift~stable%2Fstein~Id33a7ab9d0b8f33fede73eae540d6137708e1218","project":"openstack/swift","branch":"stable/stein","topic":"bug/1829959","hashtags":[],"change_id":"Id33a7ab9d0b8f33fede73eae540d6137708e1218","subject":"s3api: Fix ETag when copying a MU part from another MU","status":"MERGED","created":"2019-10-18 20:57:28.000000000","updated":"2019-10-19 04:29:03.000000000","submitted":"2019-10-19 04:27:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689589-1571459241634-d762e54f","meta_rev_id":"f62efe178428ba8164ddb8c7cf09425692ada283","_number":689589,"virtual_id_number":689589,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-19 04:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-19 00:03:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-19 00:03:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-19 04:27:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd449ef17afdc57af5ff57b7132cc05937116026","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-18 20:57:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"721c67807f786ed2d7cedf0b4aa65bb5fd21a3fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-18 21:44:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cefd87fdac6049668239350397b86689 : SUCCESS in 5m 07s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/86fc53deb66241e78edfd88b712e1d17 : SUCCESS in 13m 11s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/17064da5579d47f3be17d7d1436ffba6 : SUCCESS in 11m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f769e7933bbc48eba9d291a8b82301b2 : SUCCESS in 9m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3fdd489aaeaf4a48848d4a66eaf7e585 : SUCCESS in 10m 16s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/732c58927b62453dabbad68e1edad717 : SUCCESS in 10m 19s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/8dffa41d6f4c4e60b399b2f9fc63cb3b : SUCCESS in 11m 32s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/0f42d299b67441e6a1c62ab9b6d5efe8 : SUCCESS in 10m 36s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/58c0773670364eb0ae1c67e99b41b650 : SUCCESS in 11m 12s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/8b6ba6a9808a4ce3b56dac8d59ffea0b : SUCCESS in 11m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/734d55ddd13f449f9c772f707c5c5759 : SUCCESS in 12m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9cf6ccc0368246dc92c8d975dec4a4ff : SUCCESS in 39m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ebfbbae843194e7fa633aa796fdf9417 : SUCCESS in 41m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/78f50b4ece8e48dabac6af5eff32b908 : SUCCESS in 26m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3626d520a9bc4b85a1e60c5eb94d128d : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6252b804a1425385b8349fba7c4e97 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3b919b0ffcc4b50b47958c8c62b8c02 : FAILURE in 25m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22cc396a5f36a69be4295b46b8c8461b9c128142","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-19 00:03:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport, closes bug, tests pop without changes under swift/","accounts_in_message":[],"_revision_number":1},{"id":"52811d0714288df1507a4f868efa38062ecd92f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 00:03:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9c746f9da649aefca43ae2b2ebc140318e61058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 02:05:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a7d35db1174c95a87f65a4041419e1 : SUCCESS in 6m 27s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1be77880f61c4f81b5884e32ce695fed : SUCCESS in 18m 48s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/39a5bb77c9cf468284019a2adce684b4 : SUCCESS in 13m 03s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/d1eb78ccc16548c2807234416a7a0e72 : SUCCESS in 15m 51s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/85e5b09bc9834631ab904c778438640a : SUCCESS in 13m 35s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/745e5668c322490588a09a26e4309f0f : SUCCESS in 16m 00s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/7df71b11803a4a05a853e5ded97073d4 : SUCCESS in 14m 05s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/6be4cf86fcdb40909918acb8ec6cea1d : SUCCESS in 18m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f842756669c640998059f22af49ad2bf : SUCCESS in 31m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/59c14aeee29844ea8a70027b6c7d031f : SUCCESS in 27m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8f6bd4139f5d4e268c4784f20f10bb10 : SUCCESS in 28m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f3a1b0bb7fe48be8b212de7f3fc80f7 : FAILURE in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3335a3d75bae44c38de7c9e153f425b4 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fe751b574a258eca4d41c554d071b80ee6fd8c84","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-19 02:20:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1b41527ef7b1348cdea7745f2925829c60c61618","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 02:53:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46010be6c46d4b02854ac95cefc81709 : SUCCESS in 5m 03s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0b9440bbcb0640e2a2b872a427a8e04d : SUCCESS in 10m 20s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/ab3df08c04d540a4899c2807c6ab78c9 : SUCCESS in 7m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5546befc412f48669ba84655cd1e0323 : SUCCESS in 7m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b6f1db84a79a45d5a31382444d00af2d : SUCCESS in 8m 25s (non-voting)\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/73e62c71fd72415eab8624370d71f6e1 : SUCCESS in 7m 44s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/3cf5dd7ced084daeb918d74008e6ed04 : SUCCESS in 8m 10s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/19059313cd644101b7b8ac9a413e8ad9 : SUCCESS in 7m 31s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/07edc415bb714866abae7b2f6dbbb0c9 : SUCCESS in 9m 05s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/bc92ee1cce894a08954c5129541f7ec0 : SUCCESS in 9m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2fcedc23c1247e6b7d79feb156afa64 : SUCCESS in 13m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/55305d676af64b2297d2b80da1e103b8 : SUCCESS in 27m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57f28796393d48c5a13991493082a790 : SUCCESS in 25m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c1a7100fd4ba4eaba3888940d7eb65b0 : SUCCESS in 24m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b346bd542a824c7c996e705b3e633f45 : SUCCESS in 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f238710d54f4254b6e66cd690783713 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/48c94461cfe949c691b3f2f6019a053e : FAILURE in 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0a9907a4ee618bdd55d4f7a04a4df12089ef3b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 02:53:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0c436adcf144eb1db5b6f2f59e5f2451904043b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 04:27:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2ae9f14601facdd567647293062941e4e5fd692b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 04:27:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2ec8cc2b3ac451597e499d553148429 : SUCCESS in 5m 33s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/274126164d8f4922a9da3f52d1a414fa : SUCCESS in 12m 48s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/095146edd1e44c188a4d395524c3a0dd : SUCCESS in 9m 39s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/83b4d29eb47143dda8cc92f9a5f7a2f7 : SUCCESS in 8m 24s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/f18d368db6ba4b889a88b0ea7d2df1ca : SUCCESS in 10m 04s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/9fb62b0947e446d3b12733a5ae3600ba : SUCCESS in 9m 27s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/c2327312a9824317b9f71fd005400622 : SUCCESS in 10m 16s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/d8c6976fdddf471ebb19c05b242762c9 : SUCCESS in 12m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2ad92a99dbd44f8b9ea9a130a09616a6 : SUCCESS in 32m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a0ea95cc2ba34e97b796de6489841e33 : SUCCESS in 26m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff06b80f80c5423a8239ce80dc9d55fb : SUCCESS in 22m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d46d54369f794653aefd028571b845bd : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3b407b0c934b5cbd4deaf4e69e25c5 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"36a3bebf1e7d1d684a955912e974d10278f96ab6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 04:29:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/192de58103ac44dfaf41d0dcfe8d9fcb : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e838c4f7089b07cebe17c19f190d28ec81f0ca9","revisions":{"5e838c4f7089b07cebe17c19f190d28ec81f0ca9":{"kind":"REWORK","_number":1,"created":"2019-10-18 20:57:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/89/689589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/689589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/689589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/689589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/689589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/689589/1"}}},"commit":{"parents":[{"commit":"3d5f7aa41db7dd06e0965bd7494f2b23e7caf3f3","subject":"Merge \"doc: Fix the swift middleware doc needs more info  to set s3 api\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3d5f7aa41db7dd06e0965bd7494f2b23e7caf3f3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-05-21 04:54:40.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-18 20:57:20.000000000","tz":-420},"subject":"s3api: Fix ETag when copying a MU part from another MU","message":"s3api: Fix ETag when copying a MU part from another MU\n\nPreviously, we\u0027d preserve the sysmeta that we wrote down with the\noriginal multipart-upload to track its S3-style etag on the new part,\ncausing it to have an ETag like `\u003cMD5\u003e-\u003cN\u003e`. Later, when the client\ntried to complete the new multipart-upload, it would send that etag back\nto the server, which would reject the request because the ETag didn\u0027t\nlook like a normal MD5.\n\nNow, have s3api include blank values in the copy request to overwrite\nthe source sysmeta, and treat a blank etag override the same as a\nmissing one.\n\nChange-Id: Id33a7ab9d0b8f33fede73eae540d6137708e1218\nCloses-Bug: #1829959\n(cherry picked from commit e22960fd718c4594bc8a504a12aba3f6a52175bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e838c4f7089b07cebe17c19f190d28ec81f0ca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e838c4f7089b07cebe17c19f190d28ec81f0ca9"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
