)]}'
{"id":"openstack%2Fswift~689885","triplet_id":"openstack%2Fswift~stable%2Frocky~I7df13e670f702a2956a9c8057fcb971f4bfb7319","project":"openstack/swift","branch":"stable/rocky","topic":"bug/1810397","hashtags":[],"change_id":"I7df13e670f702a2956a9c8057fcb971f4bfb7319","subject":"s3api: Block ETag header from CompleteMultipartUpload requests","status":"MERGED","created":"2019-10-21 22:06:51.000000000","updated":"2019-10-22 18:14:20.000000000","submitted":"2019-10-22 18:14:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689885-1571768060078-cd6aae0a","meta_rev_id":"9ab7cde3e957e1fc1429a2137a11f00e6353b367","_number":689885,"virtual_id_number":689885,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-22 18:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-22 15:29:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-22 15:29:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 18:14:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"259aaa191418f73b3d952aa8302d1e989efb4b38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-21 22:06:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81020fab0dcd8a2d4f69905d607a5aabddd8499d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 00:06:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ec28e74d036b4f6182cd681be457992b : SUCCESS in 4m 42s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1c6ba2dcfcac4e43a19f58531d6e907c : SUCCESS in 11m 26s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/90a94ffeacac471da881b9a44a25d7d0 : SUCCESS in 4m 06s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/5581b18e2fc44d01a07cbb1407effd2a : SUCCESS in 8m 43s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/3270e9b315bf4bf7838071ec57bb87f0 : SUCCESS in 8m 09s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/3fc15b993b1e49df96d6e42f03ea77b5 : SUCCESS in 19m 53s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/db3723514c754355b0558cf4fa5c8cda : SUCCESS in 8m 36s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/7df3b9ba62504100979545552dad38c2 : SUCCESS in 9m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8fb0b61f84404d5ba2435ed090b44d37 : SUCCESS in 29m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9a9e97e9d9a4ff5be59f3e98dcedc54 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869f72ab38f44f44b27b336fd040ed70 : SUCCESS in 5m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d87910a058d743529cda613beb515cc7 : SUCCESS in 28m 13s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d331bafbec9f46ad9fb9faa56ba9d102 : SUCCESS in 1h 58m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9435a6fe563f4a039e03df373fac84bb : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":1},{"id":"59ec1055ce845467bd52eacaf28d568bac90331a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-22 15:29:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nMostly-clean backport (there were a few differences between about importing hashlib and using hashlib.md5() vs importing md5 directly), test guards against regression, closes a user-facing bug.","accounts_in_message":[],"_revision_number":1},{"id":"cb75459d4c8f179b91ad056816c1d4b44b9720b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 15:30:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9dc9748edb61f5b62de646e9f8f8d4108682361e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 18:14:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/44759e20a09b437bb5f299afa3928c04 : SUCCESS in 5m 07s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9706f7dea8a84a5088f5d92998b6b954 : SUCCESS in 11m 39s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/102b468695834a9abd64381c958bd6de : SUCCESS in 5m 08s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/8e3f0a2da2884d40a7b46f83ca5569db : SUCCESS in 9m 46s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/4260910ae1a748f599c4115077267e7a : SUCCESS in 10m 49s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/e661c9fdb0484579a6619ef11acddc81 : SUCCESS in 8m 13s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/c785d9fb1f5348aeb2b01440f0cf5705 : SUCCESS in 10m 23s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/4230764a319f44de9f71eeebba1c1a6c : SUCCESS in 9m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1989fd6dd5cf40d6b13d341ab5d7097d : SUCCESS in 30m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdb9834b0a5c470ab8deeb025aa2fb4a : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0479d811aa04ccd8605198a3d91a1a3 : SUCCESS in 4m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff3197e0a8364cb7b9134b43eb9101f8 : SUCCESS in 28m 06s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/a57955de6bcb49469db28ff564aeb708 : SUCCESS in 1h 26m 30s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2651b2bf9ccf473997dbe2290b1f6f86 : SUCCESS in 53m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5c8f333babd52accf5ed99b7a020c12c2df66ecd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 18:14:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c33fdf51eb0e5d8d1b6775fcb7ef12067efdbe23","revisions":{"c33fdf51eb0e5d8d1b6775fcb7ef12067efdbe23":{"kind":"REWORK","_number":1,"created":"2019-10-21 22:06:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/689885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/689885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/689885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/689885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/689885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/689885/1"}}},"commit":{"parents":[{"commit":"c6e2fbb41775e065be7cdab2366d846785b57804","subject":"Merge \"doc: Fix the swift middleware doc needs more info  to set s3 api\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6e2fbb41775e065be7cdab2366d846785b57804"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-01-03 20:12:15.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-21 22:04:14.000000000","tz":-420},"subject":"s3api: Block ETag header from CompleteMultipartUpload requests","message":"s3api: Block ETag header from CompleteMultipartUpload requests\n\nChange-Id: I7df13e670f702a2956a9c8057fcb971f4bfb7319\nCloses-Bug: 1810397\n(cherry picked from commit c51db382cb4f637fd91d1ff3d0163907906e3cc5)\n(cherry picked from commit 3996032a7e122514e34e085220345f3bfee0e0e0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c33fdf51eb0e5d8d1b6775fcb7ef12067efdbe23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c33fdf51eb0e5d8d1b6775fcb7ef12067efdbe23"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
