)]}'
{"id":"openstack%2Fswift~689887","triplet_id":"openstack%2Fswift~stable%2Frocky~Ifead1f9ca6509d50dbbef294b7cb3d7f11a09229","project":"openstack/swift","branch":"stable/rocky","hashtags":[],"change_id":"Ifead1f9ca6509d50dbbef294b7cb3d7f11a09229","subject":"s3api: Increase max body size for Delete Multiple Objects requests","status":"MERGED","created":"2019-10-21 22:09:29.000000000","updated":"2019-10-25 10:56:55.000000000","submitted":"2019-10-25 10:56:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689887-1572001015955-f546210b","meta_rev_id":"870644cfac5a07086b40a3ea0236486371af61fe","_number":689887,"virtual_id_number":689887,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 10:56:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-22 15:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-22 15:22:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 10:56:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37c945fbaad5e4fa34fe980fa68a6cd9096d0757","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-21 22:09:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05fb6463510dd7c4eb9438c196f35fc8c2d38d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 23:50:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d3f87ea4b43a41e0ac3e562c1a9007d9 : SUCCESS in 6m 02s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8da1e598faf44152a444250085c6c78f : SUCCESS in 16m 19s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/4c59bcfb96854fedb0709571d8f1e976 : SUCCESS in 5m 05s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/d1b6145bd03f4ec4b6b3615363c81e72 : SUCCESS in 15m 03s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/594a5aacef394eb4aab61e3ea48d9eb6 : SUCCESS in 8m 14s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/147ae39cf1674a00abd3826388b26b68 : SUCCESS in 15m 33s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/d62dcb26eb9d41a897e20d8f507f9396 : SUCCESS in 9m 39s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/029961f5a7964dc2b4c605356e17cc89 : SUCCESS in 9m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/908280e1a1314333a7f53a66b82e3758 : SUCCESS in 28m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b10ad6fd538f40fd91da704e6e3cf866 : SUCCESS in 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55abfcfca1bd402fb7012d41feb1bece : SUCCESS in 5m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc435363f25940dcb919db9210cc5c44 : SUCCESS in 24m 37s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/7f8b6c1b4d7e487fa13c32c181a55184 : SUCCESS in 1h 34m 51s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a5da60a12d0647cf89da2540bb24068c : SUCCESS in 55m 48s","accounts_in_message":[],"_revision_number":1},{"id":"747ca5e6eb8f42e21d5c473eabeca7b27c9d054c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-22 15:22:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport, test guards against regression, fixes a client-facing behavioral issue. I should have written up a bug for it :-(","accounts_in_message":[],"_revision_number":1},{"id":"591e717be3ec0e98836128dd1e0751abb2d901ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 15:22:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7175a31cd284fc78dd8ad63913340ef0128016d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 17:54:22.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c099dddf195e4e239da899c4c31edfe8 : SUCCESS in 7m 02s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1af8b2ebc8964a4fb879b171f9bd96d7 : SUCCESS in 12m 59s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/77057f9d2d664334b8af0a7655a2fe1f : SUCCESS in 4m 42s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/8ec713f8c7f64a5793f36ef6250c0ffe : SUCCESS in 9m 28s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/1c90e8712ed8447ba6101c1eccb54529 : SUCCESS in 9m 23s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/6d85dc3bf45b497aa73239990d669ca9 : SUCCESS in 8m 49s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/0a72932a80384e9da9d84c9c9caa5592 : SUCCESS in 9m 19s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/ebd375c7a1874902800822ccd5faa36a : SUCCESS in 10m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bcd6e1c0817d4d33902ed0b38bd6278c : SUCCESS in 31m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0bdc54d040943899478ff409abc1c8a : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7905fd9ea80e4e6bb091231bde578ced : SUCCESS in 5m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a96b1b1166a640c0b77f27509bf49fda : SUCCESS in 26m 32s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/a67c810c486a4f9abcf5aa6aaec7e4d7 : FAILURE in 15m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/077f59b2d3ab4b998072ee71a54d643d : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7bb1d6c8a0155cbdc11489d0c6a1c0a9eca634ca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-22 18:14:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"78a445d38024617d38b19591ea470ffa76589daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:45:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1563a338c18e4bc8884bcb6f63928393 : SUCCESS in 5m 39s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9ad51373892e4755a1efba81471364d7 : SUCCESS in 14m 14s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/f6534f15d4f94dc584f61cfcf811a046 : SUCCESS in 4m 53s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/ee018679fe334de2a9e106e9f3b6d2c3 : SUCCESS in 13m 04s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/e406c7532876494fa948492527d099e5 : SUCCESS in 14m 25s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/e8a70cef85944abf88d0d342dfbbc061 : SUCCESS in 13m 37s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/c50477741c0f4c75a94715cdd6c1e4ad : SUCCESS in 11m 15s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/d48da88b11804297a7a40a6d015a79b5 : SUCCESS in 12m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5fdb6ead1dd8437c8a2a8a1542655aae : SUCCESS in 31m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcf30ac9b60e4e77a2ade29b3027b7bd : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6ba221cc07f43c291d0712018c636d8 : SUCCESS in 5m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1ef33725d7d34f5a817846f587d6334d : SUCCESS in 26m 32s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/99c55cf868bd4db8bd05672f1e3341da : SUCCESS in 1h 25m 01s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0aa832d86da04be0bf0a3fc0c17f0c29 : SUCCESS in 1h 01m 26s","accounts_in_message":[],"_revision_number":1},{"id":"38b578c8968ecc0d24bcbebe8c4fd40563af820d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 19:46:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08f1f0551ebc4b3d591a67388019bfb69812a72a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 21:13:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d220f405db964bf4aa14d4f2c6016ed6 : SUCCESS in 5m 11s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a1a1f1ac69c74a0292caf2980539da77 : SUCCESS in 12m 01s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/5a39398771aa44f0b4abd32bd0b73683 : SUCCESS in 5m 03s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/5110bc92ecff49798ef99bff53045a0a : SUCCESS in 8m 57s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/448f3c88a8684c2091f0fbbbeab55de9 : SUCCESS in 8m 32s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/b0cc2f418c9944c996fab38de4e4d556 : SUCCESS in 9m 36s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/08d2550df3b346cbb5c3bb0a6314495e : SUCCESS in 9m 54s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/9a3f40a947344d71875cc64ce3a06dee : SUCCESS in 11m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a42a7ad77ad64011b613bfa17bc17ef0 : FAILURE in 30m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc4f855c61c54971a520a3bcb2cc1e0c : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b4a98aaa38448cca3fb0e7f6f4d2b9e : SUCCESS in 5m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1f7742a67d734c17abfc8ea4ffccb607 : SUCCESS in 28m 22s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/7068ea6275544ad9a957bd52565a732a : SUCCESS in 1h 26m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/074f3983da904380a1586ce985d36346 : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":1},{"id":"30bf4b047b44c83c4c8adf39933102712770a7c0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-22 21:35:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bf5033c3982b5775590273059f437d7c006f18e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 22:58:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d57177053c7a4c98b5d39724d6eccd29 : SUCCESS in 5m 21s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e77125fd62e64a0db96cef47b683bcd0 : SUCCESS in 12m 50s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/44ee6a43c74e45b8ac9078e3d077d8bd : SUCCESS in 4m 52s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/9ed1bd232aa046ef84ab22048fcb8e6a : SUCCESS in 9m 30s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/a7398b3363924d639ed584a440b24555 : SUCCESS in 10m 38s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/dc739e98d3df457881b46b2a2b7e443c : SUCCESS in 10m 04s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/23c7c63fde72404b84d2249ef5c98d4b : SUCCESS in 11m 10s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/0d93197892e44ab7ab7444d4bdafffe5 : SUCCESS in 12m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2f9c23c8925448fda5275d80b9d1401f : SUCCESS in 35m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d48b49cbaee432e8f2a84fbd6e671ef : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb43e4313734fb08ecac19cf2609c7e : SUCCESS in 4m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91cf9ccd3b3e4e94acfcba4f41706702 : SUCCESS in 24m 21s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/fc8622f2121a4123b8b86167b51dbb01 : SUCCESS in 1h 22m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3853f5d376424879b585f288c45f2d44 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a5fd591d12eada76994b275b19c8e37e96fb8080","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 22:59:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"731f344e2e05431c6a5f47c74b60cb6cd03075c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 02:16:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b063a123b8fc4ac5b6b7ad8f4a438547 : SUCCESS in 5m 52s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eeb92a56a93843f986d6094a642a65d1 : SUCCESS in 11m 02s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/2111c26798c74a0db6347eed7d7ebb03 : SUCCESS in 5m 45s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/3751eb282676429db7cf0438d21c4833 : SUCCESS in 9m 07s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/46a218eae7b24df6bc540bc5b4d5b71e : SUCCESS in 9m 58s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/5bceaccb779940b09ae540bc3d1a99b3 : SUCCESS in 10m 16s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/328da5cc3b3c4f04ad784d8dd253e11e : SUCCESS in 9m 05s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/196723721fb64dd68dc5c7db79669a21 : SUCCESS in 10m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6c4e2d55b31f4affb8606759754bbeaf : FAILURE in 41m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21193e0c28dc4ac49dfb9c3f1b5da915 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c5a3bdf6d04521928bf8a64e3d368d : SUCCESS in 5m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/376e4e5ffcee4bb6b9612847b47edb4e : SUCCESS in 28m 57s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/75cec45bd2fd4ce7a7f46ea6d923fc19 : SUCCESS in 1h 20m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/27d72a620cdc4ff98e405fcd5917eb3f : SUCCESS in 58m 35s","accounts_in_message":[],"_revision_number":1},{"id":"28a95ea29f0f8b81a03442f3df1aa5801c3f8ee9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-23 14:50:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f47fea5c5c6aa19be1d5df0e79caf98d78bddc78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 18:22:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/afe991a7b51d4c51b5e198fab3dcb0f2 : SUCCESS in 5m 21s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d1cae1bdc0e34a21b2d70cb37986617b : SUCCESS in 10m 32s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/6cdf4e177dd84898b221e9d5ab5a1f25 : SUCCESS in 4m 28s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/52df35b1c3ce4c35af28a9ce59f4b04c : FAILURE in 11m 10s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/31ad847fb4e64cdfb3fcf138cd150e2b : SUCCESS in 12m 25s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/d577d9bd80b94befb9b173a52600e954 : SUCCESS in 11m 42s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/4d4284b06a8045c7bc4dd5299b4f7bb8 : SUCCESS in 9m 13s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/2036cd6c278f426c8571721778419537 : SUCCESS in 10m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/98c3a253794a4c41aabb3ea7f2ecc0d6 : SUCCESS in 35m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93c8420ad0454e988eb5f32d967b9e97 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b65512407a0d49eb8bfb1338cfdcfa05 : SUCCESS in 6m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/20d35b5114aa49b5a282a87d3ab316f9 : SUCCESS in 32m 54s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/c682dcaa4c7848b69a9069f40a2ef03b : SUCCESS in 1h 25m 17s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/eb556a4ecfc34742bcaa343ed8781274 : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":1},{"id":"183435d967e6ae6df05c9499f219c9cc6389976a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-23 18:23:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f014b859598689a06d146d86d5843b98d772f193","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 21:05:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5aec144356e44c9a8f7773e566c36b83 : SUCCESS in 6m 27s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e2438c0303e74b2d84713a7550b4122f : SUCCESS in 11m 02s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/e6321c1d58b143798c67628f1943192c : SUCCESS in 5m 08s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/ac9c489f2ecd4f4bafd5f3b003228fed : SUCCESS in 8m 55s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/690a491194744a1e8b16136b311dbd69 : SUCCESS in 8m 56s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/eac75257e54b417f90feb5530e56b8d9 : SUCCESS in 8m 44s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/0b9857344efb4136a340b0252a2ce9fe : SUCCESS in 11m 49s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/e9c943f1d9794723bb8d7ce6eae40877 : SUCCESS in 10m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/032c59edfc1a43bea92f8b66107d5ff3 : SUCCESS in 35m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fd73c28dc804dd6b61008fca5433994 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01502d4b7ee5496c920b4fd38864d2c1 : SUCCESS in 5m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/071f6672ad564a4eb4b77222c4ab01d0 : SUCCESS in 26m 49s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/b9c3af4b14824f649b2b14093456d820 : SUCCESS in 1h 33m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/40285f02556843498dc3a1ac0d932bee : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f6f3f112af428e9f00cde9273668b90b4a764a9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 21:07:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17c4e77ebc6dee64c12b2f5f439f6ba26ca4ab61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 01:32:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/453bc0899e3d4cf58dc6990319a95339 : SUCCESS in 5m 13s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7d39606391c74a82bef4cec4dbd597b6 : SUCCESS in 11m 20s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/077fa455e9f347a69fe8f199ba81d5eb : SUCCESS in 6m 01s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/6390aabb354a4bb0b6f615728e7e3c17 : SUCCESS in 9m 59s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/c39ee754a11947f59be6d5bee990b331 : SUCCESS in 9m 38s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/85f0983bb0de4982973298819bd777ce : SUCCESS in 8m 52s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/9913512257a24306ade079acd8a37f55 : SUCCESS in 10m 36s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/1ef7891a4f0c46ce90e0d86e58af3859 : SUCCESS in 10m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/85c9085663ac46d4be801066547f8e89 : FAILURE in 30m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/786b86921b0643639e1a691e2ff8be08 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0766d98c9ee3410fb9ba45752081d7c2 : SUCCESS in 5m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f430d4474cd340af88ef3f6c429d9361 : SUCCESS in 31m 23s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/708268ba6ad54cdca9478f97184abe5f : SUCCESS in 1h 30m 58s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/685e1b43014d428c9c25ac5056b0a449 : SUCCESS in 55m 35s","accounts_in_message":[],"_revision_number":1},{"id":"6bd5515dbbdcffb3624d6c71a2946b438820618a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-24 17:54:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dcfd369815c7f9e4e289b5495defe37ccb92d88a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 20:20:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c893042210f4458694aaee57aa6aaf91 : SUCCESS in 5m 18s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/862813ad19994868a665614295fe6d8e : SUCCESS in 13m 16s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/1cf8f79f187e4067b64253a3ff2bca26 : SUCCESS in 7m 19s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/4f29aa8a89814e458865c121e9165fc9 : SUCCESS in 8m 32s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/e1cc7af1e663481f92bc07b1605daad0 : SUCCESS in 10m 47s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/e1379f0d99d442149f4f5b439fefe693 : SUCCESS in 12m 43s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/cab18d97e34f46079958f254237cfc6d : SUCCESS in 10m 32s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/97387eeab7ec49c49636f0b1e9b8b5f7 : FAILURE in 13m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dd799e67ae0147e6bc581792afd07187 : SUCCESS in 43m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1cebf8b0b5e4f499d0e3a71da9263e0 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da57d5188f034db5a7309c70851d0f69 : SUCCESS in 5m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4dbf8a056eee4b5eb9fa6e7d2309aa23 : SUCCESS in 26m 58s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/ccfa71021853406f864c6619c592b640 : FAILURE in 1h 24m 45s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f299bccb45214f34808d57b4f1123f22 : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a223356ffbe1564972be5d5b9c1d78016b1ffffe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-24 20:21:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2c9a961e7dba866f949dfeb9e4af2eb1261a5b35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 21:44:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e44cfb76614b4e12a7581617af70ec97 : SUCCESS in 5m 14s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ffe181479ea743d78c1feac26c015bc2 : SUCCESS in 12m 19s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/dbff2cc2ab3a49829ee26b51ad3e3e45 : SUCCESS in 6m 03s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/69effec64fea4fafa55ad06448f88aa6 : SUCCESS in 10m 13s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/4fbc5fb16ed7465b974a6cbc24c263cb : SUCCESS in 10m 42s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/8a688f1a92ab41e0bc01896109e9ef17 : SUCCESS in 10m 26s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/7ad9fa8f2c704756a01c3ba7feca4511 : SUCCESS in 9m 51s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/39cadcef692343cdbcf6bb4e81d9c3d8 : SUCCESS in 10m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/da595f1598d649e6a8c1578aa41a282f : FAILURE in 42m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fed110498a54e79bd9824d7052c5e19 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17596c1498654897aeff8112ef4cc5a3 : SUCCESS in 5m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1c4fba3328fc4d4684c6075ae34b1763 : SUCCESS in 30m 43s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/cda2fb912be24a71b508022b2d3862d4 : FAILURE in 1h 20m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/65ddb6773bc442459b5d8d2239be1425 : SUCCESS in 58m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d2c4d8cb7a9ffd6f5aa006e594ccb7ecd1069a09","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2019-10-24 21:45:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6e39e4609c479ea976610397270ae9ad84cf3fdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 23:24:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4d2b82703ecd4a40a9b8e6a7285d99ab : SUCCESS in 4m 59s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b0bd3b2164c64b96859fb17f22241b10 : SUCCESS in 11m 03s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/690238144e9f4d52a3cdec70ad8d89e7 : SUCCESS in 4m 37s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/d3a2e79276bb4fb8a24967c6afff5495 : SUCCESS in 11m 01s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/a08382281c0a439893c8286cdae9f320 : SUCCESS in 13m 29s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/2f15ca6b5d9a46afa19bf24ad07d787d : SUCCESS in 8m 40s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/c71d6312d0e348208b4c92d230b123d9 : SUCCESS in 9m 13s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/0175a877b3d84c2a96ce326bd3e2c08b : SUCCESS in 10m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/062b6151641a463cb424b736dace8467 : SUCCESS in 35m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d180a942dbbf4502a1f53075c0c26b26 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/596c255910a642b187f0d07cc6df1d95 : SUCCESS in 4m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24f2afaa827b4b7685ef7210e04421de : SUCCESS in 33m 55s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/dfe2055647e645dea3af605eb50a70b0 : SUCCESS in 1h 32m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fe2f5bf6ec784a1fb9a930cde0871b09 : SUCCESS in 1h 03m 07s","accounts_in_message":[],"_revision_number":1},{"id":"bdd88e1c308e21f93d33f0920d000ec73e9bbe0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 23:25:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dde639d1e3754dca76d8600c4824535ce810fbee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 10:56:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"870644cfac5a07086b40a3ea0236486371af61fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 10:56:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0c6b0e05533c4363892cef68296f7557 : SUCCESS in 6m 36s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a9ad10a6e5e42fd932ca4af5d7df9b9 : SUCCESS in 12m 26s\n- swift-tox-py35 https://zuul.opendev.org/t/openstack/build/77d406b26cfa43899453bb65f4d12717 : SUCCESS in 6m 03s\n- swift-tox-func https://zuul.opendev.org/t/openstack/build/ca57f031edac4da4b1313f107d55bbc4 : SUCCESS in 10m 25s\n- swift-tox-func-encryption https://zuul.opendev.org/t/openstack/build/910bff53ad37464ba029c25a2069801e : SUCCESS in 11m 35s\n- swift-tox-func-domain-remap-staticweb https://zuul.opendev.org/t/openstack/build/796bdd14ccbb4ba8abb066912208436f : SUCCESS in 9m 53s\n- swift-tox-func-ec https://zuul.opendev.org/t/openstack/build/65f89ff9cb4547dcaa7e1bbba9fa5832 : SUCCESS in 11m 38s\n- swift-tox-func-s3api https://zuul.opendev.org/t/openstack/build/94b5ddc4aa2548d1a1dcc5cad270af4c : SUCCESS in 11m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c50419fe557a4cf2937b8e084fe406f4 : SUCCESS in 38m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb0b65fa62714564b4e906d137ec520e : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b11a86cadd2e49ff9be14264f6e704b0 : SUCCESS in 7m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7d9cb637df1a4c81a821db90807665ff : SUCCESS in 27m 00s\n- legacy-tempest-dsvm-neutron-full https://zuul.opendev.org/t/openstack/build/d08829c43e064275921a58020339b326 : SUCCESS in 1h 41m 06s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/51ef1b6a699348119eb79559d30b6f84 : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b8c333ef31d2fe9983603c411a3648c612b3928","revisions":{"6b8c333ef31d2fe9983603c411a3648c612b3928":{"kind":"REWORK","_number":1,"created":"2019-10-21 22:09:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/87/689887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/87/689887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/689887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/689887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/87/689887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/87/689887/1"}}},"commit":{"parents":[{"commit":"c6e2fbb41775e065be7cdab2366d846785b57804","subject":"Merge \"doc: Fix the swift middleware doc needs more info  to set s3 api\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6e2fbb41775e065be7cdab2366d846785b57804"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2018-09-20 20:39:09.000000000","tz":0},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2019-10-21 22:08:55.000000000","tz":-420},"subject":"s3api: Increase max body size for Delete Multiple Objects requests","message":"s3api: Increase max body size for Delete Multiple Objects requests\n\nSub-64k was *way* too low, particularly when clients expect to be able to\ndelete as many as 1000 objects at a time.\n\nI have *no idea* where the previous limit came from.\n\nChange-Id: Ifead1f9ca6509d50dbbef294b7cb3d7f11a09229\n(cherry picked from commit ecebea18203d1ca53da8709026acf3e83c3f5357)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b8c333ef31d2fe9983603c411a3648c612b3928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6b8c333ef31d2fe9983603c411a3648c612b3928"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
